)]}'
{"id":"openstack%2Foctavia~927196","triplet_id":"openstack%2Foctavia~master~I6251e5df6c4f9a5417eb0a6b46693e44b21ed8cc","project":"openstack/octavia","branch":"master","topic":"remove-blockdiag","hashtags":[],"change_id":"I6251e5df6c4f9a5417eb0a6b46693e44b21ed8cc","subject":"pre-commit: Bump versions","status":"NEW","created":"2024-08-26 17:46:32.000000000","updated":"2026-05-07 13:05:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe5579160e44f3c679d53e40bdf3b6c222910e3f","_number":927196,"virtual_id_number":927196,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-07 13:05:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 21:21:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8575da3a31de540a3d7b2deb7208a29d7e3b3813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-26 17:46:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d79b0eefaa0cbe235318b69e74184a38d3f529ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:21:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e050dd97b3ae40aa976d30644e4cbbb2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f2d942a95194da1825e9d8ae7df0912 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb2a71b679f42c7a06bf643ceabc955 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f0a14de8f6a4f00b833579ae23595d0 : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7192f11c3bdf441da5d4959bf34d2638 : SUCCESS in 6m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c8653241ece435481a4451aadd325ce : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e12c5b6b15fe4ed498673670b0c1db1e : SUCCESS in 12m 28s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a9ec13faac114efe8fd2ece5b32f61ea : SUCCESS in 4m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b7cdfd33305240428f4025bd75af73b9 : SUCCESS in 6m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/091b8e54d81741e78f579b3faabe0080 : SUCCESS in 5m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a887e3e83764e4aba903d467e622c1c : SUCCESS in 7m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b5802da87cb499e90594d9f07b80893 : SUCCESS in 6m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5d2ecd5ae8ef48d08fb585705ce5094a : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8ce9836334584b33b4b38c3c9a04e9e4 : SUCCESS in 1h 23m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9aba9b2ad12a4436b98ac99c6a62a962 : SUCCESS in 1h 42m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a91715df7e304e729302c8a0b7e365fb : SUCCESS in 1h 39m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6c1f35161f5446d5b2f3b3ad5d831607 : SUCCESS in 2h 06m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a043c5d5173c426cbba469108c1e7d2b : SUCCESS in 1h 06m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1582895a96e64a22a127c2bfe15141a0 : SUCCESS in 48m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60e4ee5425bd48ec952de9db3acd4631 : SUCCESS in 58m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ddbe822cc3674ff29f5e3790b12db796 : SUCCESS in 1h 08m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5154485bc4f945fa80ae8929cf4d8319 : SUCCESS in 31m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/16d864e447834c3faa6d486ce47775f6 : SUCCESS in 2h 21m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/93dc4ec1e0e6401e9cd15f8f470c576d : TIMED_OUT in 3h 04m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f79a8727d6c14f88b51df28ded1027e3 : SUCCESS in 2h 29m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/42be8d0f978d457682d5455e2d0e0ee3 : TIMED_OUT in 2h 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9f17e9f8cb82ac056465213fbe20aedd4489591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-07 12:54:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fe5579160e44f3c679d53e40bdf3b6c222910e3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 13:05:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ea6a53474246e89aff953172232580\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e9e44f322242b88e45871f7eeab8e1 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd1951c918a4fcf8f8d78ca1e43678e : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1ebe91b6a62ba18ad20c0a8828a4bc33c5d0711","revisions":{"68d6044a334b3a84cecd58f24ad23977b5e3380d":{"kind":"REWORK","_number":1,"created":"2024-08-26 17:46:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/927196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/927196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/927196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/927196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/927196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/927196/1"}}},"commit":{"parents":[{"commit":"e2aa78ddb6a3b9ff7571cf26b499b887651fb973","subject":"pre-commit: Fix issues that have crept in","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e2aa78ddb6a3b9ff7571cf26b499b887651fb973"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-26 17:26:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-26 17:33:36.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nNo changes necessary. We migrate hacking to its native plugin rather\nthan using flake8\u0027s and adding hacking after the fact.\n\nChange-Id: I6251e5df6c4f9a5417eb0a6b46693e44b21ed8cc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68d6044a334b3a84cecd58f24ad23977b5e3380d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68d6044a334b3a84cecd58f24ad23977b5e3380d"}]},"branch":"refs/heads/master"},"a1ebe91b6a62ba18ad20c0a8828a4bc33c5d0711":{"kind":"REWORK","_number":2,"created":"2026-05-07 12:54:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/927196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/927196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/927196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/927196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/927196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/927196/2"}}},"commit":{"parents":[{"commit":"a4e4555b9fe64cd583ee0cf9964343a8d87e0ee2","subject":"tox: Trivial fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4e4555b9fe64cd583ee0cf9964343a8d87e0ee2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-26 17:26:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-07 12:50:45.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nNo changes necessary. We migrate hacking to its native plugin rather\nthan using flake8\u0027s and adding hacking after the fact.\n\nChange-Id: I6251e5df6c4f9a5417eb0a6b46693e44b21ed8cc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1ebe91b6a62ba18ad20c0a8828a4bc33c5d0711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1ebe91b6a62ba18ad20c0a8828a4bc33c5d0711"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
