)]}'
{"id":"openstack%2Foctavia~930219","triplet_id":"openstack%2Foctavia~master~Ia92fa6140b0fc608281e846d0635dd28217f2630","project":"openstack/octavia","branch":"master","topic":"fix-internal-self-signed-cert","attention_set":{},"removed_from_attention_set":{"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-01-10 12:36:49.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-01-10 12:36:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia92fa6140b0fc608281e846d0635dd28217f2630","subject":"Fix verification of certificates signed by a private CA","status":"MERGED","created":"2024-09-23 15:45:57.000000000","updated":"2025-01-10 12:37:57.000000000","submitted":"2025-01-10 12:36:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930219-fix-internal-self-signed-cert","meta_rev_id":"34c4b5369f7ce9163def8b6c51f2d44acfe88b9c","_number":930219,"virtual_id_number":930219,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 12:36:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-10 09:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-01-10 08:08:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":1,"date":"2025-01-10 08:08:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-01-10 08:08:22.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2025-01-10 08:08:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-10 09:23:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-10 09:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-01-10 08:08:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-01-10 08:08:22.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35638,"deleted":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-23 15:54:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-09-23 18:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-24 19:28:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-10-10 08:11:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-10-10 08:11:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2024-10-10 08:11:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-10-10 08:11:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-10-10 08:11:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2024-10-15 08:42:51.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-10-17 13:11:52.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-01-03 14:54:35.000000000","updated_by":{"_account_id":35638,"deleted":true},"reviewer":{"_account_id":35638,"deleted":true},"state":"CC"}],"messages":[{"id":"ef1b181733e381febf6fce3c5dd93ad2dbab1347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-23 15:45:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbdfdcbd34b687317d971b12d698d2f02beebcf6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-23 15:54:09.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b49b049e477596a44fa4490c30c00d5ee8278b23","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-23 15:56:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94e60c17c6b603d791bc666825853ba9aca2fc64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 18:25:39.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/199fd17f19a045c58afdfff5f8c29120\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b87a351395a40bcb9922488ff4a6bc8 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e678043855474b00aa4ddadd7ffabd29 : FAILURE in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/845e9f365f1a47d6bd5545cb01565583 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4cfa861f7fc44ac90320d589cabedd2 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98014462e77d493daa2d0dabffede8ae : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3e84b21e6848b6a1494f4d3e4e975c : SUCCESS in 10m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e576d2df35e64c958a1fd173f494aec0 : SUCCESS in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5dbf06a68d244fbf9328b08aafe775fb : SUCCESS in 5m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3e6d6eb9973e4764a912a73475b21057 : SUCCESS in 4m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c7322b44a864f13ad103b0bdda36df2 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5ad358a10b904ba4a50c1bad3d555d91 : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c7efd218fe934889b34dfa6ad00ea81a : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55828350160c40cdb306b17c5882a62b : SUCCESS in 1h 32m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/001f88e128554ed489723c2952ec0a84 : SUCCESS in 2h 00m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6ad8cb87c5ba4371a814a68df87d2492 : SUCCESS in 1h 07m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/96e53db9ed994649a1b93a2d86145522 : SUCCESS in 2h 20m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cf072366caac4106a43ebdd61d50f75d : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1fbff01b3ab249399983d78175b1573d : SUCCESS in 48m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/38c3d73367f44678a0d431eb17b1d5e6 : SUCCESS in 37m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cc19d43a9c5942fea720599546e9bc47 : SUCCESS in 1h 00m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2d5de1682364f08b892b93eab8a84e6 : SUCCESS in 29m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2bf7449142b046dc993b21499347a861 : SUCCESS in 2h 03m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0da65fec4bb44a5183d7fb3d9a8959f1 : FAILURE in 1h 56m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/39c1c7dca7214ef8b3c8ce77a2513216 : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/06b1c2cf578d4366be5ec63c06790945 : TIMED_OUT in 2h 34m 30s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/drivers/driver_base.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/data_models.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/utils.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/castellan_mgr.py\n  Comments left for invalid file octavia/certificates/manager/cert_mgr.py\n  Comments left for invalid file octavia/certificates/manager/noop.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/compute/compute_base.py\n  Comments left for invalid file octavia/compute/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/retry_tasks.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/network/base.py\n  Comments left for invalid file octavia/network/data_models.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":1},{"id":"4357375bdc76afeab41e27f84af119ffc77feae9","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-24 08:13:04.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5c6f0be5e1006eb7b53176b68c1eed3243afb78f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-24 08:39:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9e5d0d491978de412cfdeadd24df6cd076e0eb94","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-24 08:39:49.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"26db52a988394512b8b55dc53aaba7859b8433a5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-24 10:45:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4e009d550f06938b30b64839594f94f33171182c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 12:21:24.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b6714253ee4cfe86a2655b4dbce4bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c8c74351ec740debee153215b40ff8d : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52121558328d4a6dad89a66e0bb3f4c1 : FAILURE in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37b1c4d23bd743ec846b5a2dcf9d201d : SUCCESS in 6m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce9a2507f79f42078a1643db184c00fb : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59af0dd2eef64510933db1004354ef3c : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879a34d6788b40efbd5efb1a2bec8485 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2e46322bea84ca1a282e9a8e22a54de : SUCCESS in 4m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/82ee5e16c79d428da45d3431e4e75fbd : SUCCESS in 5m 28s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f2b75142c0094e188f5ec9e98550b48a : SUCCESS in 6m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/39b2613d3bcd458ebabc9af13438af74 : SUCCESS in 4m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aaa6cba30f6c412b933bbd5d355d04b6 : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58cdde0f30134865a23322ae0c275398 : SUCCESS in 7m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/91c59a0d410f474792061d60411c43e9 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/216149ffd2c14c02b6fec2730ca801fe : SUCCESS in 1h 32m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cd12701635d542f5a3a37022aa13f7d1 : SUCCESS in 1h 45m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/466cc7f3343e4aa2aa11ed18a738eec8 : SUCCESS in 59m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e0ea40a9c86b4b76bc03b7c99ee4406e : SUCCESS in 1h 48m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/49aef5ca4479403bab50863df1c3e7bd : SUCCESS in 55m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c259429aaa7e4a6fa96147d4c8b0634b : SUCCESS in 34m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a55f886a1df4a0cb44faa8f86052a48 : SUCCESS in 43m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1cfbe72ccc93475abc5519e45b2fce6f : SUCCESS in 54m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2d4d70b11af4ffaa685f431fe83179c : SUCCESS in 33m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a38b2ee8f01a4733a34154eaf9c3673c : SUCCESS in 2h 33m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a9fce59183d41e499c149ca223d1217 : SUCCESS in 2h 44m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0e0abeddfdd341ce9082e594045b4cd4 : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0371c65cd49c434b98314d1ce089e117 : TIMED_OUT in 2h 33m 25s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/drivers/driver_base.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/data_models.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/utils.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/castellan_mgr.py\n  Comments left for invalid file octavia/certificates/manager/cert_mgr.py\n  Comments left for invalid file octavia/certificates/manager/noop.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/compute/compute_base.py\n  Comments left for invalid file octavia/compute/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/retry_tasks.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/network/base.py\n  Comments left for invalid file octavia/network/data_models.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":2},{"id":"1d5ab9d18e2c4bf0534bf670a1d0f19db5d1a8da","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-24 14:17:46.000000000","message":"Patch Set 2: -Backport-Candidate\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a51500efffd9f401351c2d32c04325c741a8a4d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-24 14:18:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b9b5ca665e22f9b75c8e1d9f324941558bf8d904","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-24 14:18:58.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"24c192dd1286d0ce425d6e4df1c4634b0c8a5b20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-24 15:33:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ec0f5b6607d9bea1aa033e392d6fa1991d6695f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 17:25:45.000000000","message":"Patch Set 4: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cf4fdc2e9554cdcb4cb5aa3bc1038e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1363f1b176da49ecb4d68e2ab3d962fe : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17527a45d2cd4e5193f9a3b48dfb6a11 : FAILURE in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dba1b261a50f4461bd134d64a5ea7241 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0836c1aed3cf44b3a89ef57996d1c160 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0fa30148cf343b298f2b7333dcc8c62 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb144113cfaf4c848719d93a5c2d5392 : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1c408ca1e134b8f8635f646f75d915f : SUCCESS in 5m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/52b76e876bf84c138aee264377dafaa7 : SUCCESS in 4m 52s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8150065c81fc4472aadb98ed02fdbb96 : SUCCESS in 6m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d49d16671a8f4d4b9f926e28821aceeb : SUCCESS in 4m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/514d8e0063e1454da0fe3e9417787b12 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97b0510d75ad4ceb8394f2c8e879696e : SUCCESS in 6m 49s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ddd14fefe1cf4d37bad0d8dbf0cf0ecb : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6a2cb29a3794d589409cbd4416bca05 : SUCCESS in 1h 45m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9bfe19d1b8bc47b5bae0241f43cb776e : FAILURE in 32m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d51153241d5745ad9a15e6b4872fc28d : FAILURE in 32m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e9597e174028448888a9d810a57fce63 : FAILURE in 23m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/df1ff48ea63b4258b1869e43fa1cb9d9 : FAILURE in 38m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b77f753f8afa4a808c44e3fcfd08bd63 : FAILURE in 25m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14b1a570540b43b7acbfcfdd7df2b64b : FAILURE in 37m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a766f0584d664593abdb1e4daddd15d2 : FAILURE in 51m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b45198cb6324a64af8ae07cebe9d07e : FAILURE in 22m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73ae379525e949c2b03240312930c5b3 : FAILURE in 24m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/369de914bafc4c1b9f559c854f0735c1 : FAILURE in 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5ee61c5a64564b6d992a9afb894c47e7 : FAILURE in 23m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d104c77caa04d3e9c6b9cbc3f1ba1e2 : FAILURE in 41m 21s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/drivers/driver_base.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/data_models.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/utils.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/castellan_mgr.py\n  Comments left for invalid file octavia/certificates/manager/cert_mgr.py\n  Comments left for invalid file octavia/certificates/manager/noop.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/compute/compute_base.py\n  Comments left for invalid file octavia/compute/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/retry_tasks.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/network/base.py\n  Comments left for invalid file octavia/network/data_models.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":4},{"id":"1fd1e5a4d18fad6e369e068b92475e0bf59f982e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-24 19:28:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"72c77a6b3a133e5558712ebfb99b8788515356f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-24 19:29:12.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_36765\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"_revision_number":5},{"id":"59f767e0cb123d139ca17b5c6970b821ee60d761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 21:16:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c6d00622d694d3cb1873504ca9872b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fcaf52306054e45aba102a3cba9d80b : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800f4ffc0a0a4ef1a38b84c2871e5873 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78417cfc9e234bb99095b41095be2beb : SUCCESS in 6m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ac188a4c005410eb0cedebd6f8cb8e4 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f687e786b97407ebb2a930927d3f045 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aead5b9ff0634751aafb00c1e45e86a9 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb7c701a35114ec6807754ab37b38820 : SUCCESS in 4m 14s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6c8185ccaf8345418dfd902e68ba4189 : SUCCESS in 4m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0b65cf5349374e8ab2775cb4fbb186f2 : SUCCESS in 6m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5bac7695cc4a485bb66a46e6ddc05278 : SUCCESS in 3m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6574becbb39a4aee824995e2ec623ea3 : SUCCESS in 10m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f22e97f99c2470f967aa1adf080658a : SUCCESS in 6m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a4eb42bf78d34c808820cf76ea33da3e : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/650377d1c256422dbc6b7da533569356 : SUCCESS in 1h 40m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b89e7719ef6447f188d5145437dfbfd2 : FAILURE in 39m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ef2565636e094f6d9cbcc82a60072efc : FAILURE in 21m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d0ed15ba30d847c3b01f4f997737a9d0 : FAILURE in 21m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/10f0eab1166b4a11809a37720f6ea3a6 : FAILURE in 21m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4698480ced2445578498bc5535684f33 : FAILURE in 21m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b80ba28f28ef4a948565af9859c4dd20 : FAILURE in 42m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/35a91304a5744cedb6f302b06c5a1c63 : FAILURE in 34m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c00d4a5b4ef4dd2ad4ae002f6230d04 : FAILURE in 29m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8d93caac5ef490ba19de6772cc51bd6 : FAILURE in 22m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/011e7fac74094d818cf5679d6f88b800 : FAILURE in 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2074eb588b9a4636a5823cc8224ad0f1 : FAILURE in 25m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/52c024fe64f646f0bca93ec0ecb0bc0b : FAILURE in 43m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e4d6e9291b5ffe897bb4f9f26e0e5ed54dc47f74","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-25 08:14:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b80b32b6144270ed4d77bfaf0105345afeec26fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-25 08:20:30.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6497ddfc0bb2291d1fd3936a82e485ff819b4f08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-25 09:46:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a6d46bd9b65b7be172b6fa6be96797a9bcf1d2a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 12:25:29.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf3a30c3599b4432828eeeb3d2437e6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62f5cbdbf3754f308df64f6ad9fba157 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db4c7c29b1141dc9e485fa8843f1c75 : FAILURE in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f06d7a5042cc47be9718a0c433bc3f34 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a81cf800c6fb4ba9afbeabfb90763a84 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f038f5706ba744b091ee71271b800856 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15dbc5984dbe4b52bc1703daab112aaf : SUCCESS in 13m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c85ff0e09f5a4183b0ba85f80d134cf8 : SUCCESS in 6m 17s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/65c03b577d434698a6721e433749bc63 : SUCCESS in 5m 57s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/28bdecbcd6ae487b9a0f8b5e6f93be05 : SUCCESS in 8m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/399a2eca034c4bdca1a6011ddababfe1 : SUCCESS in 4m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b862e7a053e494b8cf7b42984e842a5 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cacfb32dad374dd3b7ad8123dc085631 : SUCCESS in 7m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a2f22750cc1242e08b1d78ae19b698c6 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dbc83d1186e145dea5846bc4bd41cbf7 : SUCCESS in 1h 23m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/029fad1b08444942a9f94a0b2ac9db66 : SUCCESS in 2h 00m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/aa6fcb2d8b8248d6a5cfd46d5c0790b3 : SUCCESS in 1h 19m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a31fb891adcf4dc8b6a1b10b6aff72ee : SUCCESS in 2h 15m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/418cd6d2a8874a0898eccef8fa73b1d2 : SUCCESS in 1h 22m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b48c4f5f1423402f91878194ec441c0f : SUCCESS in 46m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b4554318b354003a8a2062dc32b4c9e : SUCCESS in 38m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ceb4292d20bc40598b880db22029a0d9 : SUCCESS in 46m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8df200778743499bb810470c7aaa2e23 : SUCCESS in 31m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/26caa376d3d14b148863481ee502ae7a : TIMED_OUT in 2h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8747b02af92c4bc58544db8e8f7db576 : FAILURE in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e2892db4b2834f308687e469c6d30618 : SUCCESS in 2h 04m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ac317ae6aa140e6920690a4634fd98e : SUCCESS in 1h 54m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface_file.py\n  Comments left for invalid file octavia/amphorae/drivers/driver_base.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/data_models.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/amphorae/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/api/drivers/utils.py\n  Comments left for invalid file octavia/api/root_controller.py\n  Comments left for invalid file octavia/certificates/generator/local.py\n  Comments left for invalid file octavia/certificates/manager/barbican.py\n  Comments left for invalid file octavia/certificates/manager/barbican_legacy.py\n  Comments left for invalid file octavia/certificates/manager/castellan_mgr.py\n  Comments left for invalid file octavia/certificates/manager/cert_mgr.py\n  Comments left for invalid file octavia/certificates/manager/noop.py\n  Comments left for invalid file octavia/common/data_models.py\n  Comments left for invalid file octavia/common/jinja/haproxy/combined_listeners/jinja_cfg.py\n  Comments left for invalid file octavia/common/policy.py\n  Comments left for invalid file octavia/compute/compute_base.py\n  Comments left for invalid file octavia/compute/drivers/noop_driver/driver.py\n  Comments left for invalid file octavia/compute/drivers/nova_driver.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/amphora_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/network_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/retry_tasks.py\n  Comments left for invalid file octavia/db/repositories.py\n  Comments left for invalid file octavia/network/base.py\n  Comments left for invalid file octavia/network/data_models.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/network/drivers/neutron/base.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":7},{"id":"c673bd70a8fc6dc676069fc9a5f2be24445915a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-25 15:56:40.000000000","message":"Patch Set 8: Patch Set 7 was rebased on behalf of \u003cGERRIT_ACCOUNT_36765\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"_revision_number":8},{"id":"d4ded4aaa57d2460a1951942cc496a1a9d5b8787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 18:53:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87dfb64874c544bdac5132137437b637\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a4a9a898d0b49c987b9548fe6190a2f : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9390508437e8407e8d1d0c08e89c34b8 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45ffd238c72c469e857eed2074ef3349 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a3c4be7563a451dac7096eb14c1a7f5 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5db4e697c9ea450993fc96a4bcbce83a : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85c2bd5027245269235155eb15f1d8e : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f5be1a94cc04c44b062c2f137a31c27 : SUCCESS in 5m 55s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/04ffbbabbcc149039c1f0c2581483169 : SUCCESS in 6m 00s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b5d44b031d1647f585f8b0421fdf0ecd : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d8e1165760504907a3e470652c977c86 : SUCCESS in 4m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/01beaaec6c8c4298b312aa9c091949a9 : SUCCESS in 7m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16a4c9ca6df34733aa7b5530de444fe6 : SUCCESS in 8m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a219d39aa197406ca6bb117c36bd799b : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e758623a6e0451ba43bbd626b3f8637 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/847084814b4f49f2b111cfcaa56b82c6 : SUCCESS in 2h 20m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ee92912b4f3f42c1b72e7a165dfce9e3 : SUCCESS in 52m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a0b4b77663164c7496b5de0e3ce0c0f1 : SUCCESS in 1h 41m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b99cac63d74b4d91a23c8e44b5bd5c40 : SUCCESS in 1h 07m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b25392950944470971a18e8c1003755 : SUCCESS in 45m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4616bdcf8a2e43aeae46963b1b763dec : SUCCESS in 54m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/96aab77e871448e180d28526e2abeb78 : SUCCESS in 1h 08m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6feb08164c84433f96d213471943b1ae : SUCCESS in 43m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/55b70653f200481a9f04e294d7851e6c : TIMED_OUT in 2h 34m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9a075d8db9e74017beba5ccdb2a88c12 : SUCCESS in 2h 52m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d6b19ce4d16545baaa9e3dcb65a05e3f : TIMED_OUT in 2h 32m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/022340d509b641ad8a3813c83b15b3c0 : TIMED_OUT in 2h 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"305e92a2336eedfb8afc0d53425be41750a6f93d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-25 20:49:40.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"8b553ad438361f987c0f50d37027ca180bddfbcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 23:38:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d39abfc7cc44e995b0d50a7e07b8f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a16a9c83e74bae842894e9f3634710 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a446379a4b42e0bd8f3a099cdacd9c : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf06aa0663884a2993347a57e41f440f : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67436c89594e42a2aa55d64025a2069f : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2626f7e78fbb43298ab9cb2b63663f58 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e5e0b734484b479d66f769f8b86ce7 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de86e0179ddc4e4f9a9825bbc3921b29 : SUCCESS in 4m 20s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/450c4ffcc1fc4d87855485ab992f3b90 : SUCCESS in 4m 36s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/cdb155f3f74a43e899ba751f84ee6e8c : SUCCESS in 7m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4f4b54ff0d924531ab4f4e597abed424 : SUCCESS in 4m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/382d11d73b73481dbfa00f9af406686c : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45c947e88eb74045bdc0d5a0cc9ff20e : SUCCESS in 6m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f3fe5561ba9b411b89d30ad415e89db1 : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e33cd082df6c41b492b2ff43101d4a3b : SUCCESS in 1h 38m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f2b14b907d9c4e45ab172e3be121a2ee : SUCCESS in 1h 50m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e5dae4c228f04de594058463f10dd7d6 : SUCCESS in 1h 02m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8a2edba33ae64ce7b416af38a9963dda : SUCCESS in 1h 54m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dd0fed3c726e4512b97299d840dc7562 : SUCCESS in 1h 17m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c18dffe88cae4bf3ad9417ed86bf3849 : SUCCESS in 40m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9446ca6b44ae4aefa74064fccb0ffe26 : SUCCESS in 45m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dfb2893c28274cbca956d813950895ca : SUCCESS in 43m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16e5368195a4481baa5f9ffbd7973c3f : SUCCESS in 37m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d57104cc1aa49fd81d2e6810a8dffa6 : SUCCESS in 2h 27m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0fb2e52a56f042c0801d7fe82483f70b : SUCCESS in 2h 47m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/94beb05cbb7a4080b355918b09521f21 : SUCCESS in 2h 10m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/85d2ff0c03554ebda1543f5f9fc3a385 : SUCCESS in 2h 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4cb046918442289097488e8b3fefd3c50e229f13","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-26 07:26:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"20bf2fd2a9ce7f6fa54d9fd4e007c43f7c2c9ed3","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-26 09:03:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0ba8285ff386c4f865328c99e5e1da2753d91572","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-26 09:05:31.000000000","message":"Uploaded patch set 10: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"c3d9d9337b12df22e700141ea0b2af10e0c82507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 11:43:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c2e82022b342fcb2937bdf3f7ca0de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b8449a6fa4244a4aa4e83aa6279d5f3 : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8039c187f3234db186a710490b7366e6 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/820acb73e5d74d1c9d2ac87cb3567079 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9835bdfc9a64a48abafa69efa06906a : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa5f30aaf9784bf4b2682e3ec039267a : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8edcfc1c3d481fad19336ceeff376f : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/095c8310ced54a2283ec777a481ff3c1 : SUCCESS in 5m 05s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7a2b471cd8fc473fbbc8e6d74914e2e4 : SUCCESS in 4m 53s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0093a29e269341f0a4e90d419b576a8a : SUCCESS in 6m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f1fa743bfd149ccab3ee26a660e0337 : SUCCESS in 4m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2d5033bd8934fb987def7346c79af96 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81dde7be4ea943439d8f83e32a185f51 : SUCCESS in 7m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/46b09677246e454baf073cb7f645f2ab : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4eb3692ce8a4855bc9ee1dd47f4bab1 : SUCCESS in 1h 40m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f437410b405c4574ac66ea23cb60e8c4 : SUCCESS in 1h 40m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4c4aba55d20545299a51ebf5f183275a : SUCCESS in 56m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7bdec3da32d8488788984b650531458f : SUCCESS in 1h 47m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e40a7142620d497d9d147cee58e6e95b : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2be0106827e4267ad2a90667573965c : SUCCESS in 35m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d6dacb2bee4b4d178f65903e86b5dba3 : SUCCESS in 38m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c1d2e4911bdd402aa5ad02ca0259a623 : SUCCESS in 58m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14165b1a6cdc44a6b40f69d7f7a7ea86 : SUCCESS in 32m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07faa71cbc4d43f9bd10ff93e94c662b : SUCCESS in 2h 10m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7b34f8f84f648ba8b07a67229fb0e2d : SUCCESS in 2h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a9a4c5cbbef24eabbe8ea56b41887b0a : SUCCESS in 2h 25m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/46e1d4e35bf64f22af97b48a921ff43d : TIMED_OUT in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"02fd7eb0e706522f4b69133a37e4c15f8bd3ec56","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-26 14:56:08.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"929218d0f60bda0f080086a4f0bfb58bb7514921","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-09-26 15:15:20.000000000","message":"Patch Set 10: Backport-Candidate+1","accounts_in_message":[],"_revision_number":10},{"id":"8bbcc8a8484993aaaa998113032b83de5b8067e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-28 06:46:09.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"bb91018c792c2fb276f07d7accbb516f8b63f49a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 09:21:01.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad95e59a603c4c4f892a0d0c90483594\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c755ff5b2444d53bf223147a16b573f : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07966921c6e74a19b069c776ad620691 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3518c9b4c484273af2b8cdde8261473 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d16e933a43f6491da238393e23d3abd9 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7836fda39af24258be87bead2100d7c2 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b5e35ccab6d439a862d7983bd0b257e : SUCCESS in 4m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/dc9073ad29664a7699a9802cf91c20bf : SUCCESS in 5m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/4d7912f0eb0b46c8b6ba4ebdf5fb6e8a : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/583e2c971ab145ff9a161764d506ad28 : SUCCESS in 3m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dccfc0722079428eabec93b9a40039dc : SUCCESS in 6m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/17fe4db85fa543289a6359814de298c5 : SUCCESS in 6m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/513eaf8944324579b9701d590823fd5e : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b86204ce066c44798c70a1e4a4774242 : SUCCESS in 1h 11m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7c8e4f9c10454435a400780c689bba42 : SUCCESS in 2h 05m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5e58a9cab69a47c792ec0a47c4aa34a1 : SUCCESS in 1h 20m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bd3fafd990834cd786de09b838a19865 : SUCCESS in 2h 05m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9f51e8191108472a91ed02399508a9f0 : SUCCESS in 1h 03m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/77ae40dad9ef4437bcb852c8c305a998 : SUCCESS in 49m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89da8a8f345f49fdaf0ac556dfffc42f : SUCCESS in 37m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e9710e97cf40431190420454214d559f : SUCCESS in 50m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a465ab820e364d7f84c4a37ef24ba576 : SUCCESS in 29m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b16328f446f34504959cbdb0bb4ed992 : SUCCESS in 2h 09m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a17dc8e1eb274523b799247474adecfd : SUCCESS in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/86e2fc30481d4e1cae8f396380aefa64 : SUCCESS in 1h 58m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcfea7ed3dd24552837fac004c1e4371 : SUCCESS in 2h 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b9d600da698dd5a8a3f6cd4a9779aae69d3337db","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-10-01 09:54:50.000000000","message":"Patch Set 11: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"c07a2fddb40817af5800deeb86deb4205f6c2ac5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-10-10 08:11:46.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"434fd036d8ab7edba8884e5a94de9e07e291750e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-10 08:24:45.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b6dbfa409fedacbddbc8aff07c7cf846dfdc4739","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-10 08:26:59.000000000","message":"Patch Set 11: Backport-Candidate+2","accounts_in_message":[],"_revision_number":11},{"id":"c2c8cd514f165119d0022a542e41e353ccfc66a8","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2024-10-17 13:11:52.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"914bf7c2ef3918352d5bcd0e66d80a934f07f71f","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-01-02 13:21:04.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"46886bb90e4e26e15506c189e5aa903891493a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-01-10 08:08:22.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Backport-Candidate+1, Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"f4c9d6731178efa1f5afe371efe08262d359861c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-10 09:23:18.000000000","message":"Patch Set 12: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"7d8ea37679f959638f589e8211402a25c9bf93ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 10:51:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38c87f2ed3b6458cb726f95c6eee3afd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/933540875afe4b21af2efb291e8386ae : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf527a37396649289f31502b4ea9dca7 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81148d354f5e455baa4049e9242c7c86 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8945a88572f4e9f834ebb7806828ed1 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9866d6ae1383483ea02377e2dd8cfaeb : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7552301469f24bc6b450261337f78b1c : SUCCESS in 4m 21s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f6905080a97d443e96e4c7a60306348a : SUCCESS in 3m 40s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1342654a117944c8930e8e2b6f81a0eb : SUCCESS in 4m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4e6d793675a24e18b346d98949d32977 : SUCCESS in 3m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e0bc6363e3cc454598ae23091f84256e : SUCCESS in 6m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9421ea9405734cfb99a3b2a006d50dd7 : SUCCESS in 4m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0515c66eea774c8991aa35e0efe2aee0 : SUCCESS in 5m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c21dcda5e10a4800a7ae9effbfd1043c : SUCCESS in 1h 16m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/34bf8a1c64e34b91abbe93fa4c23e3a8 : SUCCESS in 1h 55m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/64e9c8c53d204876ab11b1daf06e527d : SUCCESS in 1h 14m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1b90945dab6b4bb388e3b8e198b139e0 : SUCCESS in 1h 59m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9eb477df4f5d4d30824e6bcd8279e251 : SUCCESS in 1h 16m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/244e04033c6344668a9226429bd2d572 : SUCCESS in 44m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59b4999229234a67a561a07600251a5f : SUCCESS in 49m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d827bd2546414a2ab3b3ee702fce2d4a : SUCCESS in 37m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/904593b8f8cd4deabdfffbbb6b7e07c8 : SUCCESS in 43m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2476af63443943d596a298e086a45dbc : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8829f6b6a92f4213b4ca5db34a32debe : SUCCESS in 2h 41m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82220c3bd80349f38a6f84b6731b51b9 : SUCCESS in 2h 06m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc6f80ce8a6147179d5545e2514b49d7 : TIMED_OUT in 2h 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"24334f4903837b73ee5af9ae0ccb804083020641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 10:52:06.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"3076b8c96f447a9a97fc3340f27fdbda5fb0c4e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 12:36:48.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ce7276db4c4da891c0f671982a70b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96950161fca044d9b923ea47870e8d31 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c57c8788036641129b71848d943ece16 : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4df7ec8ed47a46c697845d32ffbf93a0 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e903685d74487eb115be4c50df0b1a : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b8bb3a1a8048108bd12b86d4894b13 : SUCCESS in 3m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/55adb2060783429586a9322cd9484135 : SUCCESS in 3m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/47a0a750a0704ce8a2e81e7052d36f35 : SUCCESS in 11m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ef353e71ab44a90acc7d4c05f6ae0d2 : SUCCESS in 1h 30m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f4f1af95df25437ab2cc40182b95d32d : SUCCESS in 1h 43m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f20765fc4045466a86bda33a4ba05b37 : SUCCESS in 57m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/00bd9c607b694423b4f695988320164f : SUCCESS in 31m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c27d24161ba4c499027a5e1bbc8c334 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":12},{"id":"1ff0314850fc3149cdc1024692d7797792a34a8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 12:36:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"34c4b5369f7ce9163def8b6c51f2d44acfe88b9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 12:37:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02186407737d43dca58ebb18208469ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c99f59520a46a386b53d4b9dc560c4 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e23b5ef041ff48e498cd75cc088537b3 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"b0aa2a36f1c56dac6a54930b934eb13c31fc114e","revisions":{"49ef2810cc820e720173317d4bd302c9aa063fc9":{"kind":"REWORK","_number":1,"created":"2024-09-23 15:45:57.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/1"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:45:56.000000000","tz":60},"subject":"Fix self-signed internal certificate failure","message":"Fix self-signed internal certificate failure\n\nWhen using self-signed certificate for internal neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49ef2810cc820e720173317d4bd302c9aa063fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49ef2810cc820e720173317d4bd302c9aa063fc9"}]},"branch":"refs/heads/master"},"13d7d11b4db15389be86c179be75fa82ff877f9b":{"kind":"REWORK","_number":2,"created":"2024-09-24 08:39:31.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/2"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-24 08:39:08.000000000","tz":60},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13d7d11b4db15389be86c179be75fa82ff877f9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/13d7d11b4db15389be86c179be75fa82ff877f9b"}]},"branch":"refs/heads/master"},"fb0ae373107ad03ae564ee0556cb2cb5da2d22d5":{"kind":"REWORK","_number":3,"created":"2024-09-24 14:18:40.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/3"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-24 14:16:54.000000000","tz":60},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb0ae373107ad03ae564ee0556cb2cb5da2d22d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb0ae373107ad03ae564ee0556cb2cb5da2d22d5"}]},"branch":"refs/heads/master"},"77546c99d8d1351ec163836d76b1a70a511246c2":{"kind":"REWORK","_number":4,"created":"2024-09-24 15:33:41.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/4"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-24 15:33:34.000000000","tz":60},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77546c99d8d1351ec163836d76b1a70a511246c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77546c99d8d1351ec163836d76b1a70a511246c2"}]},"branch":"refs/heads/master"},"7d92a1aa0d4ed07ae257db6a158c621ff951423a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-09-24 19:29:12.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/930219/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/5"}}},"commit":{"parents":[{"commit":"3c6f10f189eecb0020120819c89ec658085f4953","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c6f10f189eecb0020120819c89ec658085f4953"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-24 19:29:12.000000000","tz":0},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d92a1aa0d4ed07ae257db6a158c621ff951423a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d92a1aa0d4ed07ae257db6a158c621ff951423a"}]},"branch":"refs/heads/master","description":"Rebase"},"9ca8c11005d3b6c41d8c76beec361dff3b66fd9a":{"kind":"REWORK","_number":6,"created":"2024-09-25 08:20:30.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/6"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-25 08:20:17.000000000","tz":60},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ca8c11005d3b6c41d8c76beec361dff3b66fd9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ca8c11005d3b6c41d8c76beec361dff3b66fd9a"}]},"branch":"refs/heads/master"},"8edb703affed3ef86c80cf4e161cf8a2859fde7d":{"kind":"REWORK","_number":7,"created":"2024-09-25 09:46:46.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/7"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-25 09:46:37.000000000","tz":60},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8edb703affed3ef86c80cf4e161cf8a2859fde7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8edb703affed3ef86c80cf4e161cf8a2859fde7d"}]},"branch":"refs/heads/master"},"bedc0cc3b3191fa2dcd299fe51ea1d7b207c3f77":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-09-25 15:56:40.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/930219/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/8"}}},"commit":{"parents":[{"commit":"3c6f10f189eecb0020120819c89ec658085f4953","subject":"Fix pep8 with pylint 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c6f10f189eecb0020120819c89ec658085f4953"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-25 15:56:40.000000000","tz":0},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bedc0cc3b3191fa2dcd299fe51ea1d7b207c3f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bedc0cc3b3191fa2dcd299fe51ea1d7b207c3f77"}]},"branch":"refs/heads/master","description":"Rebase"},"3b3c9f1cdf94d681e11650817050a5470fe9561f":{"kind":"REWORK","_number":9,"created":"2024-09-25 20:49:40.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/9"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-25 20:49:12.000000000","tz":60},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b3c9f1cdf94d681e11650817050a5470fe9561f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b3c9f1cdf94d681e11650817050a5470fe9561f"}]},"branch":"refs/heads/master"},"6c34fb68e88060e28e38e6f430fb1011bd5323ef":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2024-09-26 09:05:31.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/19/930219/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/10"}}},"commit":{"parents":[{"commit":"eff69261f45686ec757504fab57a65d46f8881cb","subject":"Merge \"Fix amphora image builds to use DIB bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eff69261f45686ec757504fab57a65d46f8881cb"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-26 09:04:24.000000000","tz":60},"subject":"Fix verification of private CA signed certificates","message":"Fix verification of private CA signed certificates\n\nWhen using certificate that was signed by private CA\n(including self-signed certs) on neutron endpoints,\ncertificate verification fails.\n\nThis was because custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is\nprovided from configuration file of octavia.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCo-authored-by: Ilia Kerbs\n\nCloses-Bug: #2046382\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c34fb68e88060e28e38e6f430fb1011bd5323ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c34fb68e88060e28e38e6f430fb1011bd5323ef"}]},"branch":"refs/heads/master"},"c149a8cd8c53dbfc1c28b379c2a40051df41d683":{"kind":"REWORK","_number":11,"created":"2024-09-28 06:46:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/930219/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/11"}}},"commit":{"parents":[{"commit":"b65d730316341b21c296583e2cc32090e300a16a","subject":"Merge \"Fixed retry functions with tenacity 8.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b65d730316341b21c296583e2cc32090e300a16a"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-28 06:45:04.000000000","tz":120},"subject":"Fix verification of certificates signed by a private CA","message":"Fix verification of certificates signed by a private CA\n\nWhen using a certificate that was signed by a private CA (including\nself-signed certs) on Neutron endpoints, certificate verification fails.\n\nThis was because the custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is provided\nin octavia.conf.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCloses-Bug: #2046382\nCo-authored-by: Ilia Kerbs \u003cikerbs@protonmail.com\u003e\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c149a8cd8c53dbfc1c28b379c2a40051df41d683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c149a8cd8c53dbfc1c28b379c2a40051df41d683"}]},"branch":"refs/heads/master"},"b0aa2a36f1c56dac6a54930b934eb13c31fc114e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-01-10 08:08:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/930219/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/930219/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/930219/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/930219/12"}}},"commit":{"parents":[{"commit":"9f0634c23a5b20352ba27eb971550409c91a7880","subject":"Merge \"Unpin pip-extra-reqs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f0634c23a5b20352ba27eb971550409c91a7880"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-09-23 15:38:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-01-10 08:08:22.000000000","tz":0},"subject":"Fix verification of certificates signed by a private CA","message":"Fix verification of certificates signed by a private CA\n\nWhen using a certificate that was signed by a private CA (including\nself-signed certs) on Neutron endpoints, certificate verification fails.\n\nThis was because the custom certificate was not used when creating\nuser\u0027s session even though the custom certificate file path is provided\nin octavia.conf.\n\nFixing by feeding neutron cafile as a keyword argument.\n\nCloses-Bug: #2046382\nCo-authored-by: Ilia Kerbs \u003cikerbs@protonmail.com\u003e\n\nChange-Id: Ia92fa6140b0fc608281e846d0635dd28217f2630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0aa2a36f1c56dac6a54930b934eb13c31fc114e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0aa2a36f1c56dac6a54930b934eb13c31fc114e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
