)]}'
{"id":"openstack%2Foctavia~931135","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I8e489b251ff0e58809b66c0145023769abef6e68","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-04-12 02:54:42.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-04-11 22:15:51.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I8e489b251ff0e58809b66c0145023769abef6e68","subject":"Fix amphora image builds to use DIB bindep","status":"MERGED","created":"2024-10-02 07:04:47.000000000","updated":"2025-04-12 02:54:42.000000000","submitted":"2025-04-12 02:54:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931135","meta_rev_id":"3e258d02b0d9addc3d34104c5f36ce8b561548ac","_number":931135,"virtual_id_number":931135,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-12 02:54:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-04-11 22:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-11 22:15:48.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-04-11 22:15:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 07:04:47.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-10-02 09:43:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 10:20:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-11-20 19:01:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"df3b3beffc4510f561f3e450cfdbd0e897880058","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-02 07:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab7d249e0995f5b0b701b7e2b2426f96f756ee65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 09:43:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99dcd36df46a4cb887c91cfd478e143b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7651f7ab21804185b6e75013f5d938c1 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c020f1d09a477ea41558b1bbe24e26 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed94cf420afb45fcb0c00b3166d7dfd8 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cd895347a0b43ae84a2168e3cf12907 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b290ceb805aa44238cf0972c508409d3 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34011c764e814740a857fc7dd7c47a9d : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a19366c081a44f88647884d70282332 : SUCCESS in 12m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/387597e99a8e47bf85d9aa3a2c5a2ef3 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/512cc37b52c34583acd242148750ac74 : SUCCESS in 10m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c98dadb5691480e88be584554723028 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85dc02f8f4894c11a8cba1459b95e958 : SUCCESS in 6m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/83887d23b2ad4e0399a0098f45f57111 : SUCCESS in 1h 14m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9ab56c3ee4c44f39a8cc18c18c3c621a : SUCCESS in 1h 27m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b43dabb5441e430a8a631ff9e05e3fab : SUCCESS in 1h 07m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/37b0951355334fc5b286f9692fbeab58 : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a5f8847be56647c9ae515ea7bc33f282 : SUCCESS in 1h 11m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e8549e26fc54f54b3bdae6a8ad35402 : SUCCESS in 41m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/41feec3c139a442b9bb8c9b76c3c4b50 : SUCCESS in 40m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/575d49cd63de46d4a33efda216e11515 : FAILURE in 47m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba95d6da2dd744809ebf1ce4d88fbb85 : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30d44692c3fd4a78a1a9cc92e026e697 : SUCCESS in 1h 53m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f4d8f7cde9e74a5db960b93bdcbec303 : SUCCESS in 2h 15m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ccbeb435aa5040369506a19a40bd2e77 : FAILURE in 20m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/125434e75420444986cde5bd7732d1a6 : TIMED_OUT in 2h 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08c53d7c37d3e19729cf1af6d5fddf71ffc8e53c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-02 10:20:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d0913a85036da2f6ac3302f27dfb50c389417a6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 19:01:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8e7afe9dfc1e76500d9e336b268549109aaa23b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 19:02:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cafc8c360d611817715f19191e2f3c074b59cfd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:50:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38b8904cf9ad46eeacc0eac54eca8f42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cde33c330e94edabfb424d29aeba9eb : CANCELED in 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/244fb1b5d73a4ef0ac73af6d2a64a228 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1b329e855034927b0044eaa92918b1a : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c1526ae464a41b18ec1fd52dbe130d2 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/294cd8ddc5434af7a73c02a5bb80ac6a : CANCELED in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/072a187706d84fdca99c100ab12df792 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0a271f5aa68c4bb0b39ce864c550b39f : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d31c68ac36ce4a93975475bcde920360 : CANCELED in 1m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d72e507ab02f4d029f4e1c6dca9d03e7 : CANCELED in 9m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3b4471a44c7c4bf483af7b390ae20f00 : CANCELED in 7m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eaec09c673864160aa6d5b4c841e6538 : CANCELED in 1m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/360d71f44b174364b61dbd2a90196b79 : FAILURE in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e4eb8af90b064fcd94c77d7a30e881ec3571416f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-11 22:15:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"b0117b945221f725afe70ef4fa316a004cec9e20","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-11 22:15:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aea91da1d1301034cbcbf111267fc34faefaec40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 00:49:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/428db11fd559434f9ca19fd10872f2b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f66baad0a7fc4197a977946b5a3afefb : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8927838003f349b5aa7a707209d6dedc : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaa72a8fd66349ec86252a5a522db048 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/109b7efe4b08402586d018bfba95bcab : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42150215d1a34d9e88c31dcb98328940 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0be4e9b40ad441a1bb6ccb04f71cade5 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439522910e1241eb8b7d91d68f6d9f49 : SUCCESS in 10m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/999f0f1eb7ee43e6837768693d0577ab : SUCCESS in 4m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/df5d14280b164699926449ffc106ab99 : SUCCESS in 8m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f4c9db3ae574fe9bde53382dcab20bf : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75fb92c1f34d47678fbe3047772d1e04 : SUCCESS in 7m 43s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/7bf33858fd944862bf6a06abb847c5f3 : SUCCESS in 56m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bc7dcd646042402ab79a7d73e68b9227 : SUCCESS in 1h 50m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/5deb94e18d0e4703ab14d8c9b32edd46 : SUCCESS in 59m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ff13fa28211042b19a6847dadf3ea7d6 : SUCCESS in 1h 27m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/52c4921e3df44a1c843aca61a7003cb6 : SUCCESS in 1h 30m 08s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/fdfcbe75197c40dbb8ee74db64a820ee : SUCCESS in 50m 46s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e05f8ce017484fa48b67d8361862b539 : SUCCESS in 24m 00s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2023-2 https://zuul.opendev.org/t/openstack/build/0775cd66fcd143d4b7cf52e403bf5540 : FAILURE in 33m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ac992817b18c4a12b3e2e554d794e44d : SUCCESS in 27m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c0d9cef16720414f9517bc1263de491a : SUCCESS in 1h 51m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2023-2 https://zuul.opendev.org/t/openstack/build/020afafda0e24708b160b01e3269f1da : RETRY_LIMIT in 3m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f3fc580bdf544519bda039798a55afed : FAILURE in 14m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9ef93882223342debb4e0d52bcc0ca49 : TIMED_OUT in 2h 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc48b2372d25b1aceb6ac2fe348f4f65356e1d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 00:50:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75e5bea206a59d82c31f87b406607173e3573a8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 02:54:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7db977135a3c41709e4ccfdda54778c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50cd3d865c484b37958823ce4e348a96 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25eda138b21842bbb91272af6ca4dac5 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bed698771b6a4633adafb48ba4416c00 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86c7914945244afdb0aeb2e993cf2b93 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7370cc0a46034706b3ac1f3d179dc974 : SUCCESS in 11m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c54f8d30b8b44bf0b3b1ca3b8cffde4e : SUCCESS in 3m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/654d8a3efa704e82a3d5da264192b954 : SUCCESS in 10m 17s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/7a0884963c64445fa7049e00bb0665b8 : SUCCESS in 59m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/36db76086f0d400da1ba22bb019b363d : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/81e8cda277ca4189adce74b54f911b01 : SUCCESS in 1h 01m 10s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ecf197f2453e4b82ba77147b9f4c12dd : SUCCESS in 35m 48s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/3a81a9176a304b36b21d1abafaf29002 : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3e258d02b0d9addc3d34104c5f36ce8b561548ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 02:54:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37e6e6395eca657205860b542d8525f4c6683300","revisions":{"c04b2b27e77a48798860a9a9c759e72ba149790b":{"kind":"REWORK","_number":1,"created":"2024-10-02 07:04:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/931135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/931135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/931135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/931135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/931135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/931135/1"}}},"commit":{"parents":[{"commit":"09ea1e991a9acddece7cca0c9d1ef6c6a69be37b","subject":"Fix loadbalancer stuck in cascade delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09ea1e991a9acddece7cca0c9d1ef6c6a69be37b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-09-11 16:46:18.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-02 07:04:32.000000000","tz":120},"subject":"Fix amphora image builds to use DIB bindep","message":"Fix amphora image builds to use DIB bindep\n\nThe amphora image builds are currently failing because we are trying to\ninstall the binary packages for diskimage-builder ourselves in the\nansible for the job. This patch changes that to use the bindep.txt\nincluded with diskimage-builder.\n\nConflicts:\n\tplaybooks/image-build/run.yaml\n\tzuul.d/jobs.yaml\n\nChange-Id: I8e489b251ff0e58809b66c0145023769abef6e68\n(cherry picked from commit d682e74d33073dede4476e8de3e20766dc49063a)\n(cherry picked from commit 9cfdc84cb16968eadfc99bbf9ea22f147e355aca)\n(cherry picked from commit 87f5d0a4c8012e0554af29c984f6707e61249de5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c04b2b27e77a48798860a9a9c759e72ba149790b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c04b2b27e77a48798860a9a9c759e72ba149790b"}]},"branch":"refs/heads/stable/2023.2"},"37e6e6395eca657205860b542d8525f4c6683300":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-11 22:15:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/35/931135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/931135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/931135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/931135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/931135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/931135/2"}}},"commit":{"parents":[{"commit":"ce0066f03fb605fca20fea06f1d0967e5957f10e","subject":"Handle undefined protocol field in security group rules correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce0066f03fb605fca20fea06f1d0967e5957f10e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-09-11 16:46:18.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-11 22:15:48.000000000","tz":0},"subject":"Fix amphora image builds to use DIB bindep","message":"Fix amphora image builds to use DIB bindep\n\nThe amphora image builds are currently failing because we are trying to\ninstall the binary packages for diskimage-builder ourselves in the\nansible for the job. This patch changes that to use the bindep.txt\nincluded with diskimage-builder.\n\nConflicts:\n\tplaybooks/image-build/run.yaml\n\tzuul.d/jobs.yaml\n\nChange-Id: I8e489b251ff0e58809b66c0145023769abef6e68\n(cherry picked from commit d682e74d33073dede4476e8de3e20766dc49063a)\n(cherry picked from commit 9cfdc84cb16968eadfc99bbf9ea22f147e355aca)\n(cherry picked from commit 87f5d0a4c8012e0554af29c984f6707e61249de5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37e6e6395eca657205860b542d8525f4c6683300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37e6e6395eca657205860b542d8525f4c6683300"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
