)]}'
{"id":"openstack%2Foctavia~932826","triplet_id":"openstack%2Foctavia~master~I759cde796229bb7886160761233679ec886131d8","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-25 16:37:21.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-06-25 14:07:37.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I759cde796229bb7886160761233679ec886131d8","subject":"Fix inconsistent parameter type and default value type","status":"MERGED","created":"2024-10-20 06:34:29.000000000","updated":"2025-06-25 16:38:30.000000000","submitted":"2025-06-25 16:37:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932826","meta_rev_id":"9c7ff3c09256c741b8f877befd45085805bd457a","_number":932826,"virtual_id_number":932826,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-25 16:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-20 18:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-06-25 14:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-06-20 19:56:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-06-25 14:07:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-20 09:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 18:15:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-20 19:56:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-20 19:56:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"38e013511ee848f3d3e1d4dd0dd4914447d5e6aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-20 06:34:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b12ac1fa5d0fdf39ef0fad43d612de358499c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-20 06:43:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8862259b1cb401dfc17b9fabe2403dc1f5943867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-20 09:41:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e865692ce68d4b3095aec9da2737c339\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2df6f3dc6f54fc0bbabce6320874b1c : FAILURE in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5594470fc69f4062891839c2527660a9 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dda8914e5a164c1982664ad11ce5f7d5 : FAILURE in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f2e1f83707346a4ab30ffd72c6de71b : FAILURE in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c272036244d8432ea03ea756152e014f : SUCCESS in 8m 21s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f63e476b8bed48b8a13731ad29264033 : FAILURE in 3m 45s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f6a55824a0754c989c98b93e25744679 : SUCCESS in 6m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d038e2fc5d1742ab93204e89e9f29593 : SUCCESS in 3m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8cbb6cf714cb40c3b01e9868fcdcfdbc : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5f6bf5c7d0844d2ae85d712622eb277 : SUCCESS in 6m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e99a4137dad34df6b0fa2e3020af2000 : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ab2d3dc81234b05b49209571db5cfc2 : SUCCESS in 51m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/66969b5dc21d44bb9de9d3bed1070167 : SUCCESS in 1h 45m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a201c4a0ec5540969311384ec38854b7 : SUCCESS in 1h 08m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8184566351de46cb9f4e6d1afbcabd59 : SUCCESS in 1h 42m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4540b80f62d14b559ca2377e43b2686b : SUCCESS in 1h 01m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1cf15f69a99e48bd98d9cea04aed4d4b : SUCCESS in 31m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/58476809a65a4d47b179f0a2c39fed70 : SUCCESS in 39m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ee91f93a893946968ffe395a33137e73 : SUCCESS in 47m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f7d197129c748f49fba287d46739917 : SUCCESS in 30m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/57e8cfeeef294182a3fa7462a821dc8a : SUCCESS in 2h 27m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dfdc53ffd60946aaa4a8147c11855f8e : SUCCESS in 2h 49m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/37be39b7c2b24f92b923addfafed1aaf : SUCCESS in 2h 24m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5860c606dea4b138d9c3fd21bee0997 : TIMED_OUT in 2h 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b0f2d94a19736b312948879b727ce17143979ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 01:06:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ae2c900c5dff19cc3c963edd0aabb044018e269d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 03:40:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e788ca1594754ea98e1f9ac39fe125f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/058a538a67304ead97334544f1bf68cf : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdccbf8a655f459386fd6c49c1b67e38 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3054bee1d85442fe9ef8f48934214991 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f52b2888e2794114a48cf5581f89b71c : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dcfcba56e741189fac519df1408eaf : SUCCESS in 9m 23s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6370b5fab99b4ac6bc7c6fb2cc688fa9 : SUCCESS in 4m 15s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1eed58c5a52041cd883efeeacc5446f9 : SUCCESS in 6m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9f7eb0b6ff7448f494fb381ce1990c45 : SUCCESS in 3m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/168877f9a03c42fca8226de53720769d : SUCCESS in 6m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ede8c2c4da8c429184b42e0bf064bcea : SUCCESS in 6m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5c070c54450d4638aec24c54b06f4b7b : SUCCESS in 9m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/61a4c2da8f23453b9b459e88cab727fd : SUCCESS in 1h 21m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ab8b925ec2134b2c93807e9f84550ff4 : SUCCESS in 1h 46m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cbe8cf65835847988570302d1e704ab2 : SUCCESS in 57m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/971d1b4388e64dc6afbd1cb76202aa07 : SUCCESS in 1h 52m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/24c7d542a34a4086b35447741e440d9c : SUCCESS in 56m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d48708c14e7d4c9481433b236c181f03 : SUCCESS in 29m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e5cf6f3c7534073a28320dac1bff86d : SUCCESS in 34m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/caede7379b0e4009b4af1048ff7e6857 : SUCCESS in 49m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cfca8754f5564164bf75232cbdca07f6 : SUCCESS in 30m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c189f7560d24f3faedb0533e56f9ad7 : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/367485fb01954ce8981596c440553f14 : FAILURE in 2h 15m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf9f730e3d0c440bb55eb9ccb1ab3df2 : TIMED_OUT in 2h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0e1961cf3074d71aadf643f301dcca9 : SUCCESS in 2h 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b17907602de68df1f1e07e7a035842570b0b5517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 15:43:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"54c4e27e87edd225b5fc440a674d780b2ea862c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 15:45:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"730627828681f24b566af4c54b74accab9695ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 18:46:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1ad7cf86ae415281edf7d55e3fcb2c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e00ecf35118e422ebfd8da572e608347 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1d73be74b641ebaa8350e4eaf7e534 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62b4753bb4854b7daa90df876ea44dda : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a49588f4ef846caab5f053fe6965bbc : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8693e20c495645678aa3d7d4be324f55 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c85c8443fe4ab38ab71509a011e2d8 : SUCCESS in 13m 15s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2ddd341aa94a43c980d74df7096947b7 : SUCCESS in 5m 07s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/44925b35510c4446b7945c4af98c6c5b : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f48c90e71e8943f0a242fc70daaf41b6 : SUCCESS in 4m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0683c098ece94ebb90be484a6dfc5d39 : SUCCESS in 6m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f4e344801104441aaca3029936c97be1 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2aa3eb3d5835434088210472ba81404c : SUCCESS in 1h 12m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/11dc6bd3f41642189927372770ee8428 : SUCCESS in 1h 50m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b7ba43b39f724061b89e0700f8683e25 : SUCCESS in 1h 10m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acab654f602f41cb9e6a321f014685bd : SUCCESS in 1h 45m 50s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/282c29a7e34f4e70b1851a8b8a5c05c4 : SUCCESS in 1h 19m 25s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0748a223a88a4243a06199378d598853 : SUCCESS in 2h 00m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/82e6e69bbdd64ff1a0ffe3475670cf77 : SUCCESS in 1h 06m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/633bc5f7d6b04898942fb966e5f34860 : SUCCESS in 53m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/150cee3750034fd99abc0695e364fb1d : FAILURE in 38m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c086d2d863f44942943d48f4f4c8698a : SUCCESS in 39m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f3900cacb36e4b0dbfc70899b2984fcf : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/460a9624808b476bab85cce4a8d8b4bd : SUCCESS in 2h 58m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6d0499fad45f45ae9074baa8966eabb6 : TIMED_OUT in 2h 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"55245a53c418743565924ba17a0d7ec25711a360","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-20 18:15:28.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0f4afb48a0c8fdfe0baf473699d4c0d545881ff3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-20 19:56:05.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bd1c55634fb2078c01d37fbd749835adc3c70c70","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-25 14:07:37.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c38d2d5c3a654be9da3724c0802a4dd640a7441c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:08:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5c6df2c7e0ada4c56c53a2c57e5d0f371171fad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:37:21.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e656aacbaae44e8c945e172392172a46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ba1fc395c0418c8add98951390de7b : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abf46bd607b64d738f58be2d455ddb0f : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa9f7cfeced9450cbcd91fd9861f8327 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb06b783fc074e2b9f8e92c654206fb0 : SUCCESS in 11m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/435cc8a3300a4062834eda4c2cb0c092 : SUCCESS in 3m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3a2e616135dd435b87cc731c7785b8c0 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3aea8f1d2c14ab086f377a13b486dc5 : SUCCESS in 1h 24m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/687a8253827d4c45ac131b5edd9e955c : SUCCESS in 2h 27m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/55f64be3ea904e14801d37a4600d68ad : SUCCESS in 1h 08m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6e28e93b78b649a2b5348dbb58a89614 : SUCCESS in 40m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99988c94101c4ffea0943bb6638c1e8f : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":5},{"id":"729503d9a908a65901993fae0bb512cc15c6a3a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:37:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9c7ff3c09256c741b8f877befd45085805bd457a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:38:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38558925905d49fca77fe9ddcc7487e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeae1737c1b24007a7cce6d425352889 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c0891486ab65a0540f8ab2fda0fc26d855a24c9b","revisions":{"5854df2221c0e2d5405723bdc7f127dc4058fd88":{"kind":"REWORK","_number":1,"created":"2024-10-20 06:34:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/932826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/932826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/932826/1"}}},"commit":{"parents":[{"commit":"b65d730316341b21c296583e2cc32090e300a16a","subject":"Merge \"Fixed retry functions with tenacity 8.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b65d730316341b21c296583e2cc32090e300a16a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-20 06:33:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-20 06:33:46.000000000","tz":540},"subject":"Fix inconsistent parameter type and default value type","message":"Fix inconsistent parameter type and default value type\n\nMake sure default value of list options are lists.\n\nChange-Id: I759cde796229bb7886160761233679ec886131d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5854df2221c0e2d5405723bdc7f127dc4058fd88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5854df2221c0e2d5405723bdc7f127dc4058fd88"}]},"branch":"refs/heads/master"},"ac93096b75a97f5dc73bde4e123ef987fd3e419d":{"kind":"REWORK","_number":2,"created":"2024-10-20 06:43:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/932826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/932826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/932826/2"}}},"commit":{"parents":[{"commit":"b65d730316341b21c296583e2cc32090e300a16a","subject":"Merge \"Fixed retry functions with tenacity 8.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b65d730316341b21c296583e2cc32090e300a16a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-20 06:33:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-20 06:43:32.000000000","tz":540},"subject":"Fix inconsistent parameter type and default value type","message":"Fix inconsistent parameter type and default value type\n\nMake sure default value of list options are lists.\n\nChange-Id: I759cde796229bb7886160761233679ec886131d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac93096b75a97f5dc73bde4e123ef987fd3e419d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac93096b75a97f5dc73bde4e123ef987fd3e419d"}]},"branch":"refs/heads/master"},"0264d57466bc39d441a1775b549ec7c9c6e99730":{"kind":"REWORK","_number":3,"created":"2024-10-21 01:06:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/932826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/932826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/932826/3"}}},"commit":{"parents":[{"commit":"b65d730316341b21c296583e2cc32090e300a16a","subject":"Merge \"Fixed retry functions with tenacity 8.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b65d730316341b21c296583e2cc32090e300a16a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-20 06:33:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 01:06:15.000000000","tz":540},"subject":"Fix inconsistent parameter type and default value type","message":"Fix inconsistent parameter type and default value type\n\nMake sure default value of list options are lists.\n\nChange-Id: I759cde796229bb7886160761233679ec886131d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0264d57466bc39d441a1775b549ec7c9c6e99730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0264d57466bc39d441a1775b549ec7c9c6e99730"}]},"branch":"refs/heads/master"},"8cdbd93c786a4e9e7fd70bba27ea6a3ca5f3ae9a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-06-18 15:43:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/932826/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/932826/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/932826/4"}}},"commit":{"parents":[{"commit":"57eb9b49dbc6b87f46e615f07220f048be66e31b","subject":"Update diskimage-create.sh to use noble as default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57eb9b49dbc6b87f46e615f07220f048be66e31b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-20 06:33:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:43:36.000000000","tz":0},"subject":"Fix inconsistent parameter type and default value type","message":"Fix inconsistent parameter type and default value type\n\nMake sure default value of list options are lists.\n\nChange-Id: I759cde796229bb7886160761233679ec886131d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cdbd93c786a4e9e7fd70bba27ea6a3ca5f3ae9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cdbd93c786a4e9e7fd70bba27ea6a3ca5f3ae9a"}]},"branch":"refs/heads/master","description":"Rebase"},"c0891486ab65a0540f8ab2fda0fc26d855a24c9b":{"kind":"REWORK","_number":5,"created":"2025-06-18 15:45:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/932826/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/26/932826/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/26/932826/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/26/932826/5"}}},"commit":{"parents":[{"commit":"57eb9b49dbc6b87f46e615f07220f048be66e31b","subject":"Update diskimage-create.sh to use noble as default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57eb9b49dbc6b87f46e615f07220f048be66e31b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-20 06:33:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:44:49.000000000","tz":540},"subject":"Fix inconsistent parameter type and default value type","message":"Fix inconsistent parameter type and default value type\n\nMake sure default value of list options are lists.\n\nChange-Id: I759cde796229bb7886160761233679ec886131d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0891486ab65a0540f8ab2fda0fc26d855a24c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0891486ab65a0540f8ab2fda0fc26d855a24c9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
