)]}'
{"id":"openstack%2Foctavia~932844","triplet_id":"openstack%2Foctavia~master~Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3","project":"openstack/octavia","branch":"master","topic":"taskflow-hide-dict-keys","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2026-07-30 07:32:49.000000000","reason":"Change was submitted"},"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-07-30 07:32:49.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-24 07:40:21.000000000","reason":"Change was marked work in progress"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-07-30 07:32:49.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-07-01 06:10:12.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3","subject":"Mask sensitive keys in taskflow logging","status":"MERGED","created":"2024-10-21 08:25:43.000000000","updated":"2026-07-30 07:34:18.000000000","submitted":"2026-07-30 07:32:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"932844-taskflow-hide-dict-keys","meta_rev_id":"a398485038ceffa980dd23dfe0b2d8a78f22c76a","_number":932844,"virtual_id_number":932844,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-30 07:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-07-01 06:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2026-07-08 18:11:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-30 05:49:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}],"reviewers":{"CC":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-21 11:30:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-23 11:10:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-03-03 12:20:31.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"CC"},{"updated":"2025-07-10 10:05:31.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-07-08 18:11:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-07-08 18:11:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"}],"messages":[{"id":"ecfb6ad21758d2b3fd66967774c49e73d533c386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-21 08:25:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30e914e4b8361f4c79367603e54f6da5af017bf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-21 08:26:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1918325139e8bac330f74b00e9d3c869a4b8ce5d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-21 08:26:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c90d121adb6f3c4cd3042216ea21bc2cc6686981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 11:30:00.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/468fd95566ea4a1aa51815e6ff006d22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d0dbca55a244a8cbe34da602b042c67 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d2d04fe9f447fa96a2e4595e17bdcf : FAILURE in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fdfb0652c9d4c1f97b642844f587bab : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74ec99782efc45608f7b03bbfd61c08f : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9561ab9de1741e4b38cd4c403740d61 : SUCCESS in 10m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d44b88728b9a4feea94477dd7dd68a21 : SUCCESS in 3m 52s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/acbd7bad72594bf0bf0f7baafa315b06 : SUCCESS in 6m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/18bfc1995bb842d1800cd20b55ef55f4 : SUCCESS in 4m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af596009199642ca9411e4d9e2c100bd : SUCCESS in 5m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ace59672a9142f6a82fd0446bfb644f : SUCCESS in 5m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/173827e106f945e08877d4a55e40a9ff : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b9aa7dec8fe64f4e9139c650a25831de : TIMED_OUT in 3h 02m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b34b8dad14bf4b6f83b81cf943457040 : FAILURE in 50m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ccb861bcc4c14973a2c3348be05b6ab1 : TIMED_OUT in 2h 32m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f6a6d754b22441ed90c1be9b065f6220 : SUCCESS in 1h 36m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/23b9b3db11234268a9263b0fc69a21dc : SUCCESS in 1h 13m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e810d759690c4d05a4bea078f18e1bd0 : FAILURE in 59m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ca2c2da3b6ad4a92b9c4463b9bd450dd : FAILURE in 1h 15m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c996c7011c254dc991ff3a72e097ad05 : FAILURE in 1h 15m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0afda91b403c496db53a0094c6ed5ac3 : FAILURE in 43m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dadafdb7f9234636a637c9cdc4cb475a : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4967c35ebbbc430bbd631483543ed2de : FAILURE in 3h 01m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0778653e7d2046f8a0315b27db8575af : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ee29c449ac345e4b25d6980d31fcdb6 : TIMED_OUT in 2h 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1254b223f7cc27cfebc32e9f302914140c4f2d62","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-23 07:27:01.000000000","message":"Topic set to taskflow-hide-dict-keys","accounts_in_message":[],"_revision_number":2},{"id":"9a6146d573060c840c409f6aa9f9db4584c5bced","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 11:10:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e9d0d6d21222893258c291db41f9cf147832f9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-24 07:39:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ac1fd5b101b2bc63735cfc51ddf3555dae220064","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-24 07:39:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"ed82c33e5e54f25e095d79c926e1779a00d2e303","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-24 07:40:16.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"753eba3d77efb0e6d348fdd2d81c9028cb2ff4af","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-24 07:40:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"dc62cde38ae4a585f46361cd352c88a810d7b2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 10:43:13.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc7bad0a80264e3086393e4ac6911245\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c81248061bac49ab9a3d832efdbbada0 : SUCCESS in 1m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46a3d83d6ea3447eb247c1157d397211 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f35d8cf388e4fed802254a31fda4816 : FAILURE in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a3ede0dfd1249dbaba3b304fda4d3b8 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a36210e46db453aa7e291c9f6e1acd6 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5adf4a6441754d87a5fecd9378eb2725 : SUCCESS in 11m 06s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/30da7c34b60a46799d474752c081edc6 : SUCCESS in 3m 52s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/9c4046a9d010492f81bac211c6272892 : SUCCESS in 6m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b64bef804c4243eda3ebbb8c84c5a56b : SUCCESS in 3m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f1a931dea0e41a69bb078740f053297 : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6bc188e3e1c84b168065eeeb8efa9120 : SUCCESS in 7m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9768553aa95843ccb471a453453c53b9 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a74315c4ede94d9fa2c89c2f6e907005 : TIMED_OUT in 3h 02m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/58770e6156dc4918920652292bc5232c : FAILURE in 42m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f5dab56ba7ae48a2b5a31ee6d80ebb7d : TIMED_OUT in 2h 34m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/555ee887bd88403098b589e8cecdf1db : SUCCESS in 2h 28m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e14b5721eed34f12bf83edafe0a745ca : SUCCESS in 1h 32m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b786349e21b48f99906915c22e229d1 : FAILURE in 57m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/373727902a834fd0af6a061444455e4c : FAILURE in 1h 17m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fefd20301e014787b56b52e61cbc7297 : FAILURE in 1h 16m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08a91318a82e43f4bab3e699e2c04b83 : FAILURE in 1h 02m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/89b546475fd54bf48b4bcdee757ae151 : TIMED_OUT in 2h 32m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a763b5d917ca4e2e8712651eef748c67 : FAILURE in 2h 45m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0afbf80dc40e46cf8d3a6e734f4ff36f : TIMED_OUT in 2h 32m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0670a1f09f4a4f08bab2705d11ba9a7d : TIMED_OUT in 2h 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4197af995b9ff6ffca1dce567269ecb2839f54f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-27 10:06:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f8114c22f45681a7e13c5d759df3b242cfdcb1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 14:31:32.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c8733ffa0ab4780b2e6e884a7ca98d3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f9a814a2f6240a49068c22c309a6b81 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8de2ab770cab4814ac1400349688c6c8 : SUCCESS in 14m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d2826bd07f40ed9b255d1b2471a59b : FAILURE in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4f5a782f21b44cabbf82baef59493cf : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db0445f17c2e4787a56d070728f4a439 : SUCCESS in 5m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c7f334bdef034e7bb853e4e008860231 : SUCCESS in 5m 03s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/04cc2779e0e54db188a0d7df17058793 : SUCCESS in 6m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bf3d747be26d45628ea3398efb3e6527 : SUCCESS in 3m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6c214fc63d75459792fa9869e9345167 : SUCCESS in 15m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a808b49332904b7ba2d764cce623cfc1 : SUCCESS in 11m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/924ef86a13024a76b1c9e7ff94844cff : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6121b0860097468db40c6becee3efddf : TIMED_OUT in 3h 03m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fe9512743f384db886df1f1706a12091 : FAILURE in 58m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d878f5feb82e45f094b4b79ba632b74e : TIMED_OUT in 2h 32m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78895f94eaf1469abca58fd580831162 : FAILURE in 53m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b03f2c0364994ca5bc5af9a78438850a : TIMED_OUT in 2h 32m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b91cd35bbbef4558969a629eca2c22b3 : FAILURE in 2h 11m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/761c9b6871c14a1fbfd0c48f6a863853 : SUCCESS in 1h 23m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ae7a7361f3f4c249f69499e5de587b3 : FAILURE in 59m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aab0bbf43f2f4e4792a074d457db617a : FAILURE in 1h 02m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7fc443f23ae5443ca514e2581c604658 : FAILURE in 1h 25m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0ef7fb081a84b82a35f24102880e85f : FAILURE in 58m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/867c282a758c46c992d130d4b482febc : TIMED_OUT in 2h 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d46dae5a1e624756b5cec8508b26ff34 : FAILURE in 2h 49m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cb718a26892e4171903be902cfd3f644 : TIMED_OUT in 2h 32m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f797ae25d58b4ae9ae6a42bcba9d3b72 : TIMED_OUT in 2h 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33f95e6fbb86b7f6578208742c1e6678d16ed395","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-03 12:20:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9ea0f8b591a29e49fc80e2495d099af66ca9cf12","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-04 09:14:44.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"83a669e79269f432e0bf5c7fefd36b6aaf9f6ab5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-04 09:14:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"f5a9562b29dafb29713671d1d5acfbc94582e6e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-22 07:53:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a47cce86f045559f6bf2e3fd992d4530ca34ec1b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-05-22 07:54:09.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"09ce43ed10e3af99c19981a786bac16a67f1018f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:00:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/722895215bd54e438fbdac907c58c991\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d19277da80404c65962bd3752db2de97 : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02b3eb2c034f44d59df8988494455ded : FAILURE in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2dcaca49c342bf97e1caedf8c1fc4c : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd7400177a624282b9bfa214defb66a2 : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2bd750a855b49a79672cfda7ae030c0 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5071f909f11e43fa925c9f8a99d02816 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78aca0b6f84341848dc5c6c08a8cd602 : SUCCESS in 13m 30s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a832fccd79734cd3863a7bb2e353d5c6 : SUCCESS in 4m 05s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/446d5a284b9948b3beebc979b6a6a357 : FAILURE in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0566b6bfcb294774aca58e1547b6ee80 : SUCCESS in 3m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8264224104e2422dbbb3d60f9241a321 : FAILURE in 7m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/519fb9b9a81843aab346c86361217442 : FAILURE in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/61eb6d0c80de46d7bb0f5a30064c6022 : SUCCESS in 1h 28m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1028166e2789429695268b563b0e8edc : SUCCESS in 1h 44m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ac2a82dbb3b1469bbca271ef54d46da3 : SUCCESS in 1h 02m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2360d02dc44b417180fd3ed34b7d179c : SUCCESS in 2h 11m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b859c018f7549e38784f1bf52b0615a : SUCCESS in 1h 27m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/976f7ff598f743d18fb4230fec255166 : FAILURE in 17m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ddf471ad00cf4acf95779d9d46ea03fd : FAILURE in 15m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/925f2c6db9a84d1d93364c575c2705b7 : SUCCESS in 48m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8d441cff09724d0f976a217ca596dd21 : FAILURE in 53m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab438220428748c09f162c2729ee6eb6 : SUCCESS in 43m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a697d7c2437c40ad9ac83dc158829cb7 : TIMED_OUT in 2h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce087ce5a6c643569a35a41575eaaf8a : SUCCESS in 3h 03m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/21a5d9851c3242e7a6e2836b138af5b4 : TIMED_OUT in 2h 33m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3bc80d95202bf3469b43a94c849b171e09f0761f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-07-10 10:05:31.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7b06fb2804a81d662f5a5360e1bdcb596b5ed9af","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-29 07:52:02.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"23f9ebe94d51f99362c3115610ba9280b1ad8d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-08 21:34:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c7f3eb2e2aece4419fc190901a976ea6e9dab90e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-08 21:35:52.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a82e466b21835357d1771853af270efdba05e3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 00:23:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b28989ce6a424810b2bc3211a06dd47d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c607fdce93084e9e90d668415aee206c : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b0dd46d6dd842bc85c2f9d30faef01f : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62a293db577498b85d2399aadb6fecd : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e618b272b754ed6bf5ef1bce6a0f593 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adbd7653256946e0b914ee080403972f : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dcc23299045c40b5afb3359a6addfdeb : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c333f904be7489682e1a3078508502e : SUCCESS in 10m 39s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/85a8e9a1d3b1409a86439e9f40707338 : SUCCESS in 5m 03s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ca41bef3b7c44d34b846be24433b0ba3 : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9be5220023054a7cb35432e51c0aa799 : SUCCESS in 5m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95e1a9bc1ac0413dade9793ffde619cb : SUCCESS in 9m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3e01cb394d34b3ba62ab0db928bbec7 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43dce1b25dbe49d598d8467c5e3935a8 : SUCCESS in 39m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f261b52ae0294a72878d4ca71950529a : SUCCESS in 1h 53m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/65e0a858a31146c5a14bbcb02df7bbf2 : SUCCESS in 1h 09m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bd639bf4b9704f23b00d4979f2816cee : SUCCESS in 1h 57m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fdfa294f778c4d04a24306d96cdd9349 : SUCCESS in 1h 13m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2deefc2dcfce4d8baf4bb443770e71f1 : SUCCESS in 39m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ab0a3a532bc4b379faf8279ea994a5d : SUCCESS in 47m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a3628e9a63c40d6bcb9f9636a2caa2d : SUCCESS in 37m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d096936cd9e43db9e98d3d916824606 : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f0ce3c8e42674408beee2acdab8163d3 : SUCCESS in 2h 47m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d765fb1e9a045f8a4b041619eb1d60f : SUCCESS in 2h 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"860f5c5f2095fcadfdb21c57068ebd8b20c9df7c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 07:58:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d61abbd9fb40aae140429c95e0cb2e49d0011703","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-18 09:15:58.000000000","message":"Patch Set 6: -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"d08142ed5071ff42a838ea3473fbc9dd5aa95068","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-18 09:16:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9dd91cd7085812d4d8612c1b65855f2bdaa46d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 06:46:06.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"da334295f00a94092bbe27cef8daab8bd463f3fd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-30 07:09:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"204b9ab7e4afbe7baac73d8f33630bd7c466eeb8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-30 07:25:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"62074d079547e1b6623cf02e580da5f545f86665","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 08:31:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d73e27b1d2106da2838b5b28c7deee2499427523","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-30 08:54:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"feba2efc72c26cd14595b2e819bdde9550fc1a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 09:14:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e49d251665148c5955a5c5907803b7e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b5797b3f13143fcb60a7a59be1db329 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a78c08dab7784602bc70db406915529d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f61d89724384d52a3f951069d54d582 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca1509d5280b487ebdfdaed5ad6edfbb : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5c23c80b8e043e3ad06b01e8c735f74 : SUCCESS in 7m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4bb96b94cb99441e9be391c7ab2ca77e : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92be5a10fd384e06990765fcfe4608a1 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83eae737aa834f7ca98b2f42d648b9c0 : SUCCESS in 9m 30s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/08ef1e8413624d5d8e8c7c91fede68d3 : SUCCESS in 4m 37s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/14b9c77012304d259c5db31697e51dd0 : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c8b94ea793a4f7b9fcedf9b3ee87300 : SUCCESS in 3m 06s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2c629affd01148baa1912ba011e5c371 : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86121350a29f445a93a03f750b1df3f2 : SUCCESS in 1h 02m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/557d2a3ab8f24caeb16d950cf8188782 : SUCCESS in 1h 28m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bbc1165ce5c34370ba392307e5e1a50a : SUCCESS in 1h 02m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/198d5db85e7945358c92dd9934b3cf2e : SUCCESS in 1h 37m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2fb3857e504b42a1af724634b847bc97 : SUCCESS in 1h 00m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51bf27202e7a4ae89700ae69043f9b2c : SUCCESS in 35m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44d3fcdcb30847ac9cc5f2ebaa519f77 : SUCCESS in 49m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f2da8658b17345c78c1a08bcf26bb450 : SUCCESS in 42m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/139164f4ff9a4a5ca5f6c2ae9034e615 : SUCCESS in 32m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/259dd92894ea4198b3d4bff8e524c39b : SUCCESS in 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6952f01bbdc04ec4b2d24c6cb3dc805c : SUCCESS in 2h 26m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f514600c6512419b916ddcb4d3f83586 : SUCCESS in 1h 27m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/61316224268d428db3c1da74b3a4304c : SUCCESS in 1h 20m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/08809b52a422449084f57a0905219ffe : SUCCESS in 1h 00m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6fefdbd7b2750eeefc6200c25b45ae934e269e96","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-30 13:48:41.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"24abfe724f199f01e9620ee53e67f8872e867488","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-06-30 13:54:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d1af80d58a1885c61c81bbf599e02294919b99a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 13:56:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"e7d58300cb950df1338dab79aef87562bf1aedc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 14:00:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b25fe9a5cb80a7649c40836d2d38f9d513e4d4c3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-30 14:01:13.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"cbbfb23b9a1bf06a1e24f17ba050f56560be3a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 16:48:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/538afe15eeaf401f8b245af73c520211\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1bc84e2d8e864a498d82f5dc347fde15 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42d87e531cdb492795119546e29b86c9 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44eab77476274cc48f8fa7cedea6c020 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b50f747a4274f55a1ab898950226b39 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12c0ae333f0f4106b99d13a8b727f36c : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b497aa06e4254290beae3bd0419b49ba : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0de596304d3431bb473eba638e6558d : SUCCESS in 14m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e76bd2062cea4d2dbb2916971bde6cc2 : SUCCESS in 5m 08s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c3478294fb8c4badb8bf02870afeed72 : SUCCESS in 4m 00s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/7a95be299c954bb38c3e38ec0e1d1802 : SUCCESS in 5m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e27d65fd43b74c6eadd4173353667518 : SUCCESS in 2m 27s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a7d19c4a2d9f4c8c83ba3f04444140ca : SUCCESS in 3m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a12660db8cb04c89a2d43f8fa4061b8a : SUCCESS in 54m 35s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/419fb2f26d524fc1822969bfd5f2bd0b : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/babe3cf62909415c870347461e222b48 : SUCCESS in 1h 05m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9a932b62152243ab955cee55df4aa8d8 : SUCCESS in 1h 43m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/98e4b8e311154bf2bf4e6541abce087a : SUCCESS in 1h 05m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c8e6f2c44e1435f83810ba147e3dc47 : SUCCESS in 36m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55286c0f58d84267a9c79974bfe112ad : SUCCESS in 47m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6dea847cfd344b7fbe2074b61f3a153b : SUCCESS in 44m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f70ea5048ac846ee920254a33562fd91 : SUCCESS in 34m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/def3c5d9f2e9420ebb0a43a21ea4623c : SUCCESS in 36m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9c95899ff0504d01870a8ddd9a119fb3 : FAILURE in 2h 46m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/35804e1f26664a9da6c4a9874e1d0dee : SUCCESS in 1h 41m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/b10098aea37a4c958f1142eb7f6d0001 : SUCCESS in 1h 30m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e259412bb7d742b9a7e6f2b35486bf4f : SUCCESS in 1h 04m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ad674d9f25ced9af9976d8a57c029c3b9bc05ee8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-01 06:10:12.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cfaf8d7bba36617f38a27dbfb16fe914b0b2f4f2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-08 18:11:21.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"9bb6193647c91d39ca70f64b713c108977cc65bc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-30 05:49:47.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"1a9653a2983b8c364855083893740f4cc6a85158","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 05:50:27.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"6209f2f65b4c2904f091062f22157c35a13b9bbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 07:32:48.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26ca15eac298431d9db10347e5de40ec\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e89390e19b64ae388eaca6dcc090179 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/306b8c8750b849658290639d5c52a908 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf059f724f8c4fb4883fe12ccb78b4e7 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14a89b39fd9a41bca0ad02ef11561c33 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f0f44ee2434c68aa635195ceae728d : SUCCESS in 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d7f3db7ba5c43a396b24398b0bcb3f9 : SUCCESS in 4m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/595be0660a5342b89d8a8b7f155d868c : SUCCESS in 6m 34s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/13f45aadf46f473095888729f320d2d2 : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ff50096c511443a95642cb628b437ff : SUCCESS in 1h 06m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/882b681ad7544ad7b445f2af2c7adf3d : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c246f54889414ee9807bc11d204aaf74 : SUCCESS in 1h 12m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8706d174498d4291b44c9243e23d2ef3 : SUCCESS in 32m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c02ff85e3474292880e4ca39d48140a : SUCCESS in 45m 28s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a570f690b08e49d0b6a10a03603b6809 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":9},{"id":"f169da9f92daccec562bda30a0a787b2c6c02820","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 07:32:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"a398485038ceffa980dd23dfe0b2d8a78f22c76a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 07:34:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/125236eaa8544bf69c86d8c5615bbf1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c93c2294f954c27a15795a58bba52ea : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c661785b7fb34abc805af2ea8a2d86a8 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"e84556244afa374a0dec04ea7d00ee719783155e","revisions":{"43581e9245c6e6c001f95a4fa2092657e0946040":{"kind":"REWORK","_number":1,"created":"2024-10-21 08:25:43.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/1"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:25:31.000000000","tz":120},"subject":"Use hide_outputs_keys for taskflow logging","message":"Use hide_outputs_keys for taskflow logging\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/932843\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43581e9245c6e6c001f95a4fa2092657e0946040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43581e9245c6e6c001f95a4fa2092657e0946040"}]},"branch":"refs/heads/master"},"edb8d14255b8c63d42df0a05fe9dc11f41217530":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-21 08:26:23.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/2"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:25:58.000000000","tz":120},"subject":"Use hide_outputs_keys for taskflow logging","message":"Use hide_outputs_keys for taskflow logging\n\nUse the new hide_outputs_keys param in failure logging\nand dynamic logging in taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/932843\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edb8d14255b8c63d42df0a05fe9dc11f41217530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edb8d14255b8c63d42df0a05fe9dc11f41217530"}]},"branch":"refs/heads/master"},"f97dce0e635c6996d0aa85ac6a61b1ac1c56b36a":{"kind":"REWORK","_number":3,"created":"2024-10-24 07:39:34.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/3"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-24 07:39:24.000000000","tz":120},"subject":"Use hide_outputs_keys for taskflow logging","message":"Use hide_outputs_keys for taskflow logging\n\nUse the new hide_outputs_keys param in failure logging\nand dynamic logging in taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/932843\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f97dce0e635c6996d0aa85ac6a61b1ac1c56b36a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f97dce0e635c6996d0aa85ac6a61b1ac1c56b36a"}]},"branch":"refs/heads/master"},"03243f95b8a9cfe73ba715a253a33475bf08b5dd":{"kind":"REWORK","_number":4,"created":"2025-02-27 10:06:19.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/4"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 10:05:44.000000000","tz":60},"subject":"Mask sensitive keys in taskflow logging","message":"Mask sensitive keys in taskflow logging\n\nUse the new mask_inputs_keys and mask_outputs_keys\nparameters in failure logging and dynamic logging\nin taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/932843\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03243f95b8a9cfe73ba715a253a33475bf08b5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/03243f95b8a9cfe73ba715a253a33475bf08b5dd"}]},"branch":"refs/heads/master"},"e9ff900dcc476410b36e041d77c0e5088bb5b54b":{"kind":"REWORK","_number":5,"created":"2025-05-22 07:53:47.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/5"}}},"commit":{"parents":[{"commit":"599873b01ffc0adadf94ecbdb791351275ef10f8","subject":"Re-enabling voting for octavia-v2-dsvm-tls-barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599873b01ffc0adadf94ecbdb791351275ef10f8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-05-22 07:53:41.000000000","tz":120},"subject":"Mask sensitive keys in taskflow logging","message":"Mask sensitive keys in taskflow logging\n\nUse the new mask_inputs_keys and mask_outputs_keys\nparameters in failure logging and dynamic logging\nin taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/932843\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9ff900dcc476410b36e041d77c0e5088bb5b54b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9ff900dcc476410b36e041d77c0e5088bb5b54b"}]},"branch":"refs/heads/master"},"99d9ce7ef6a4157a38853e249c73ceb5b15d2766":{"kind":"REWORK","_number":6,"created":"2025-09-08 21:34:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/6"}}},"commit":{"parents":[{"commit":"81b890ca3538cb8fceeb1094e5216df40787f659","subject":"Merge \"Remove remaining jobs with Ubuntu Jammy (22.04)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/81b890ca3538cb8fceeb1094e5216df40787f659"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-09-08 21:34:34.000000000","tz":120},"subject":"Mask sensitive keys in taskflow logging","message":"Mask sensitive keys in taskflow logging\n\nUse the new mask_inputs_keys and mask_outputs_keys\nparameters in failure logging and dynamic logging\nin taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/960206\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d9ce7ef6a4157a38853e249c73ceb5b15d2766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/99d9ce7ef6a4157a38853e249c73ceb5b15d2766"}]},"branch":"refs/heads/master"},"26374ee6602ffc8875f994a8aee6c591aef9851c":{"kind":"REWORK","_number":7,"created":"2026-06-30 06:46:06.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/7"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 06:45:54.000000000","tz":120},"subject":"Mask sensitive keys in taskflow logging","message":"Mask sensitive keys in taskflow logging\n\nUse the new mask_inputs_keys and mask_outputs_keys\nparameters in failure logging and dynamic logging\nin taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/960206\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26374ee6602ffc8875f994a8aee6c591aef9851c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26374ee6602ffc8875f994a8aee6c591aef9851c"}]},"branch":"refs/heads/master"},"066b28ecf132f5895d7ca062cbc4b940ee29d5dd":{"kind":"REWORK","_number":8,"created":"2026-06-30 13:56:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/8"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 13:56:33.000000000","tz":120},"subject":"Mask sensitive keys in taskflow logging","message":"Mask sensitive keys in taskflow logging\n\nUse the new mask_inputs_keys and mask_outputs_keys\nparameters in failure logging and dynamic logging\nin taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/960206\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/066b28ecf132f5895d7ca062cbc4b940ee29d5dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/066b28ecf132f5895d7ca062cbc4b940ee29d5dd"}]},"branch":"refs/heads/master"},"e84556244afa374a0dec04ea7d00ee719783155e":{"kind":"REWORK","_number":9,"created":"2026-06-30 14:00:12.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/932844/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/932844/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/932844/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/932844/9"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 08:21:19.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-30 14:00:04.000000000","tz":120},"subject":"Mask sensitive keys in taskflow logging","message":"Mask sensitive keys in taskflow logging\n\nUse the new mask_inputs_keys and mask_outputs_keys\nparameters in failure logging and dynamic logging\nin taskflow.\n\nCloses-Bug: #2028546\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/960206\nChange-Id: Iceb1dcdc81f714b793aeef3fa41fdb03d68b44d3\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e84556244afa374a0dec04ea7d00ee719783155e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e84556244afa374a0dec04ea7d00ee719783155e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
