)]}'
{"id":"openstack%2Foctavia~932847","triplet_id":"openstack%2Foctavia~master~I97984cfab6eb29d15fb7b03da446914adcb7c524","project":"openstack/octavia","branch":"master","attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-04-05 13:32:04.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-09-18 22:05:12.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"38360":{"account":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"last_update":"2026-04-08 19:37:22.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-09-18 22:05:35.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-06-12 15:09:36.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-06-12 17:50:39.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I97984cfab6eb29d15fb7b03da446914adcb7c524","subject":"Improve HTTP 500 responses in Amphorae agent API","status":"NEW","created":"2024-10-21 09:03:28.000000000","updated":"2026-09-18 22:05:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4b1d0ef48cff5472fc17500012a183e48bbdb6d2","_number":932847,"virtual_id_number":932847,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-12 17:44:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2026-06-12 17:50:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2026-06-12 17:50:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-21 11:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 17:16:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-03 11:25:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-03 11:25:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2026-03-19 08:32:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-03-31 18:30:59.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-04-07 22:33:33.000000000","updated_by":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"reviewer":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"state":"REVIEWER"}],"messages":[{"id":"dd8b9568f2c0756250d90a430fb07287c847ac3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-21 09:03:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baf9ae49b710d8ed52556bc3ca2c821b4a40d6ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 11:44:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a117ad7e90d14b11a294cd67697c70b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bc496dffd0542b0be72a3a56d5ba85b : FAILURE in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babba0c563024c34806fd86e5d877845 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/255d86f79f4e47d0b8071d63ad13c7aa : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c18b6c7e9e1461ab53fb3fac1a23a8e : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30bc8379a96b4bc0825602cd5f9db1cb : SUCCESS in 11m 16s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e8f888c93dc146d9b1fd5908c2c2cec7 : SUCCESS in 4m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dbc01fd684af4412ad0e99303f3c44ce : FAILURE in 5m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e6b2e425b17f4105a47a255fb2f1ec8a : SUCCESS in 4m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/231cc86470224fffafa4ff3a5a94491e : FAILURE in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2cdc81831f584446b6431cdbf8e5838b : FAILURE in 5m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f65aaccda6443a299a6246da9dc0d8e : FAILURE in 5m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17aaf7e853f54d56bd3b2fc0c6b19ddc : SUCCESS in 49m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/31e8e1c16b3c492fafb250172a3d917d : SUCCESS in 1h 43m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/39bcb78291d14516b02d5120c73d6021 : SUCCESS in 57m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7a8f8ca0ef01405c891f4f45b684c2e6 : SUCCESS in 1h 49m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf40750cb5514fcca081d15253dbf9c7 : SUCCESS in 1h 19m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c0c29788cae4d27868f90c15869e31a : SUCCESS in 33m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/998fef9f64854a62bea07e9e869fce7f : SUCCESS in 43m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c3acc1d60f4140179c6f726f61f958da : SUCCESS in 43m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91f30028f5b44140b1e31ec0c68c43d1 : SUCCESS in 34m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba2c205b7e494fbdb7a34ac6e381f0aa : SUCCESS in 2h 23m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/decac6e97b944bb49f74e4c5829ae4e6 : SUCCESS in 2h 39m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eb53245b448e4913969fd9bd19e6a53e : SUCCESS in 2h 04m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2f38c9d1ff24b5baf5a03d508e02241 : TIMED_OUT in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88cf8179289f251905390efd76cb5ac81ff8b673","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-24 07:41:11.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"fc4d953fedc8c42af542ca427327b02ecc47bd2e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-25 08:21:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b57a71f28060fb69ad7f259f069a1ffd193180f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-25 11:28:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a02f194010684e3bb2432ca23a07cf82\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/125b4cd2334143f3b5d53e7ab14bd360 : FAILURE in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2522f969eea2426384e4e3cd8014cf8f : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbc17df0cf5f49f5b51ee005be2c4a59 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08bd277479e24679b478896357337bf0 : SUCCESS in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a205be2040e4a14a717d8c5c96e4d28 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e45f332367f4caa99486d9acc2dcd7e : SUCCESS in 12m 35s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/5cb6db86f5ef4586b025d2f69af9c01a : SUCCESS in 4m 21s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/3f8e2aa0ae06477fa7583ce3c27a7cde : FAILURE in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7704ac7f71ef4a4c99cb4566924ae44d : SUCCESS in 5m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/227d91748e6a43bd8cff51fb7973d964 : FAILURE in 8m 30s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ec9b53624ffc43a694d0d169c9ab1307 : FAILURE in 5m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e6cb7b8c6cd42b68e58c7ec1570e865 : FAILURE in 1h 12m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c0c82d90f4004c12b23d33ba812c8891 : SUCCESS in 1h 38m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4f115cc0c8dd4e79a95e5641d2f802cf : SUCCESS in 1h 28m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/83fabbb24c6b41f29dc7a6f4e74eed66 : SUCCESS in 1h 57m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d7ac5f62ea1c41d89fd7b9d4c322bfda : SUCCESS in 1h 18m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba837606ee944949b9a89fd62b5b58db : SUCCESS in 40m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29c3a66cf12c4ab88189a1055214ec2f : SUCCESS in 43m 58s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/bd69d3c0e2c04604bba6dd2abddeff05 : SUCCESS in 47m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d6046fe4d204f1d90f918676b719708 : SUCCESS in 39m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2143c915d56045b9a9ec0f0c81158873 : FAILURE in 58m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/45d62952fd2741ff952a8f7fabb76d0e : TIMED_OUT in 3h 04m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db744b79b8904534939d04a038bad09a : SUCCESS in 1h 34m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/916c29fc798547efb5144680da39ea7f : SUCCESS in 1h 35m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8e4c78e859594c3e9ec4749d99af8c79 : SUCCESS in 1h 01m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"132d79de6e9a06cabcf51faaa26a8ced9edb955d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-03-03 08:20:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"269726ccf0bab24c5b22745b62fdb8c5a5c49056","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-03-03 08:20:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"74e66e3b27bfb8789bc41bc70982502bde7b22b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 11:10:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51fc53d039b64252aed9effc03334d75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab61f54a78dc4e5696de1370dcdffb1c : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50f56beef354dbea08946e32dea50a9 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95c5cfca85784987bfd4d43e2e5233b6 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21bf2be560a34fe29ce9432c477ea161 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a6af796b9894ffd801127f6ce957f21 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9503eb04165646e29e507dc588843bcb : SUCCESS in 10m 39s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/924817763b244cca9d336cd87cad2cf2 : SUCCESS in 4m 43s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c559622ec0e74f2f92bdbe8c5e3ec874 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8b0511e6829f414dbe3edadbb706a9cb : SUCCESS in 2m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9f8e5c02823646d6b161f78c69b366e8 : SUCCESS in 7m 32s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bc3f43d104c9492db5da55a88b16a093 : SUCCESS in 4m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7715bf013dcb42d1906a600d3a6f9e2b : SUCCESS in 57m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b3e30cbc96714e8cbfb22b4787f6538c : SUCCESS in 1h 35m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/84f4fc444fdd4f3fb8dc484d82859eb1 : SUCCESS in 1h 14m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b5412f49891541eda012ec6354350a7b : SUCCESS in 1h 42m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/db52d468ae0c431d9b2c65c666207524 : SUCCESS in 1h 20m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6643a9d0be14470b077fcc0ecf096cd : SUCCESS in 36m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1b2f1b523d354cf2ae203a3bec86094f : SUCCESS in 45m 44s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9da14222bb1c40d58b66edf4302a67b8 : SUCCESS in 48m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/854fbe78d87044d483e28772fcdd22ed : FAILURE in 52m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/31ae968293084647b9042d962ce7ac89 : FAILURE in 55m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2a61f17ddc264fd0b5d90c644d5d9359 : SUCCESS in 2h 47m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/706bf575b0e44819bc1202188b433bd0 : SUCCESS in 1h 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/bd2adec73f3645dc9e8e329f9ab855f5 : SUCCESS in 1h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f8c69ea7e9ab4817906f45bf6cb63665 : SUCCESS in 1h 07m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"50b3bd0e65292d4bbbaa080752e5b3ee5aced271","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-03 11:25:59.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a8a3fedcf8efdf85ca734267ed31d241ca0a8ace","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-03-31 18:30:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f95720bd14d3cb60b5fa72152991852456b929ba","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-04-02 08:40:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9b25ef3ef20a52aa752e7a282ebe93c78d15e419","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-04-02 09:08:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"235c64aa50af06f818210e03030dd03989de0a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-05 13:32:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"344c5ca79633e655f4d0242c6e1b89b2cdaaa884","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-05 13:35:32.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"55de2fcba5b8b69f7d86df498bea0389fd7dfe3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-05 16:05:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1b6bd8d25e0431bba66bdfa8ef11168\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/becb6c8e3ac94cde8707e9ee81d7e2f3 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/178840609907424ea8f5969c678c71c4 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb6a6a15ab4e4e53998394c309690c00 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5dd73519d41e42a09a7dd9416ac719f7 : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9c7f9c73e6fe4f69a1914cd6416a9939 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d1d0a9d28f46a6865144b17c608d31 : SUCCESS in 14m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4445b2e1d9f47119e904bc78ba3cae3 : SUCCESS in 6m 00s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/152ef563353f4d1092584f9e23269392 : SUCCESS in 5m 10s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/3edd5277f349485ea9e4041cd3be00d2 : SUCCESS in 6m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e51eda9b3764512bba003963338e9a8 : SUCCESS in 3m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c99c5b48be04a6cad7ef7d81ca7712a : SUCCESS in 6m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/367af3f9d44f4b108e997a28bea2e90b : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4286dcfb893d4e25b55eb73bd73575fc : SUCCESS in 1h 03m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/df2d0d06de2847558886e45d4d6567aa : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c92904d1410b43afb4a0aa7bb13379c8 : SUCCESS in 1h 08m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4deb818a65204f92ac469328f89e81ec : SUCCESS in 2h 21m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b3b3d094682540fa997aefa7efb0330c : TIMED_OUT in 2h 32m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1a4f2d13bf6a4ac59999835ab1d1c3df : SUCCESS in 39m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f4eaa0474f894ea0b29257b08b7a3669 : SUCCESS in 1h 08m 51s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ec32eedcdd1d4253b8300aaee25179ad : SUCCESS in 45m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c5694d364dd4d9789021e537ca23287 : SUCCESS in 35m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af3699772ca04abe970bffa7ac74c271 : SUCCESS in 36m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5945004a4e884a1eb466045a4f115889 : SUCCESS in 2h 23m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/89edff54c8af4ae99eb0a87844f6b8f9 : SUCCESS in 1h 42m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/870c298123124bac81bc0b4783724a5c : SUCCESS in 1h 31m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/161ed8258859451d92589b6534cf20a3 : SUCCESS in 1h 04m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"499e9a34e77f50261def0697812a436c47fda29c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-05 23:04:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"88bebf828d876b8fb03edbcecf4cde9c3d207c8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 02:17:39.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32261af0d8d3450b9485f872436e61bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0b368befec345c88e74ce11a3ac6a33 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244096781af149c1abb1dc69d4fb79f4 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dafe60170e154d5e86e837c3a319b27d : SUCCESS in 3m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e2fd319f19b4ee78942d7c8cb10f59d : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d9a999606244199836047c17fb6ba2b : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4ac44d860d494c9a47ea8414a0b6f6 : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c0b26e5ea6d490d91ee57d9bfdd2cc1 : SUCCESS in 5m 36s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/9ac980b0a1bb4edeac3ba1ea762a60f1 : SUCCESS in 3m 26s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/6285837acd8147ae9c55e658bbddc643 : SUCCESS in 6m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e14cd79e63e64f9eb505e1cbcf3f0c97 : SUCCESS in 2m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00d65d204cb648a7988b679660f4e5ea : SUCCESS in 5m 57s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a77ae2a2a8124a9eb18ed75dfd37ab06 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34024cab14834016b1ba44aca40ed7df : SUCCESS in 41m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3cbe74a86df54f9385da46c609d035fe : TIMED_OUT in 2h 33m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9a2b9e909a144ca9a6bf3d0a5790e7cf : SUCCESS in 56m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/530693744ba944e393be7d350593bd7d : SUCCESS in 1h 37m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3d0ff1f216604cb29703aa88cf49b30e : SUCCESS in 1h 12m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f6cc8d9fc5d4308805d4769ea23cfc9 : SUCCESS in 34m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b0d5c90c6254984b989f87d2a5a2278 : SUCCESS in 49m 32s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/efd79c853f544ed0bf12b29e36eb3db6 : SUCCESS in 43m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7c9c882236c493eb3490e3674eb396c : SUCCESS in 40m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec43ff6d9dbe47e5b0577eb3b7161c25 : SUCCESS in 42m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/eca7c014f07b49c4b52ba7d659fe6421 : TIMED_OUT in 3h 05m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac404558f3d44dda95a23a53d0b84538 : SUCCESS in 1h 33m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/58514b0a04714cb7ae72fb628901c263 : SUCCESS in 2h 11m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bb6acdf722e54abaa3e4061d0b893e76 : SUCCESS in 1h 04m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dbb46f0c0a3691ef1ca54f2dc1b64e05c2dd1525","author":{"_account_id":38360,"name":"Zachary Mark Raines","display_name":"Zachary Raines","email":"zachary.raines@canonical.com","username":"raineszm","status":"Sustaining Engineer @ Canonical"},"date":"2026-04-07 22:33:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0f9d52d5579f3667d459314d3a1fbeabac3b27b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-08 19:37:22.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"dada26383f4f9661c6c589adeaa87c2b47abea9e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-08 19:38:26.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"efbd798e16e41abbd65ab2c3c60ec7b4a6f568cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 22:20:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86be04ab61f0466b9007508278dc81ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d186a732112545e58a91f6f40a46f4b4 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48e6b6a3c6641bdaef7773f24602e7a : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25d34bff191940fcbd144f09bba3f0da : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08f2cb4b315c4de6b74e4916430fead1 : SUCCESS in 4m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71d0b9993d54449980c57ac0738ea876 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae4232c1fdc481397881e3fac503460 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d4507c62de5447c906f0741a002dbb2 : SUCCESS in 3m 11s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/3780346944de4bfd8373837ac7a57a48 : SUCCESS in 3m 43s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/6229db6012f64e40b63b96196aba7c58 : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0a1a83d205648d8aef165dcc3d46ce9 : SUCCESS in 3m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/188daf7a626c489e9ddaa0fcfb322cf3 : SUCCESS in 4m 11s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b6922dc36e2e43668bc7d0d8940b24f5 : SUCCESS in 6m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7977b6e5ba8c4ca181c96bc1aab5bced : SUCCESS in 1h 48m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8791d1cf314d4c06ba0f8785de87ca2e : SUCCESS in 1h 35m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4595e12dcd2247bd8d279828dbe69c85 : SUCCESS in 1h 04m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3142e0314c6845488c8be78cfc571425 : SUCCESS in 1h 40m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/de732c126cfa4995b8e51bb082f24431 : SUCCESS in 1h 07m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb820a40c17d46f7b37b44bf1cc0c10d : SUCCESS in 36m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7cd7be5413e84be3ba5d3ca872aca212 : SUCCESS in 47m 39s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/59e5e8909e99409c9dd296f9ef854426 : SUCCESS in 45m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/deef065d40564d43ab836e1719b1569c : SUCCESS in 35m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c44cc4ae2e114b28909bd7dc9faf5e37 : SUCCESS in 36m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ca70b642781446b87741c547972022e : SUCCESS in 2h 42m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc8cdd0154c04391b8ae29362d65fc49 : SUCCESS in 1h 34m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c81c306e1e3d42eaafbc7aec9a04bce5 : SUCCESS in 1h 29m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/08d9e49380074a27be6762dce1a9fd91 : SUCCESS in 1h 06m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4dd9ed46549ddf9cae674a354d79930adf75b498","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 08:13:27.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"30594c02dda808800b203cf5fadf9b9caac761e7","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-04-10 14:01:20.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b2140a8bbbcf0c7026011a34d3a9ba994960d27c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-12 15:08:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"0c7dfaad7337833d821c107ff7eaccaae89c10b4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-06-12 15:09:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3eaaa759c3d2227b724a4f0a7d8e56d1e3d94d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 17:44:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/128a1eb185a944e5879ae4baca74b33c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd22c0f9acfb47bc9cd892f8d1dc5eb1 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55ae57d30be4a72811ce279dc5364d9 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c586ead19c454a8fb8b0b9b70817bcb2 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f544b1f7ee44da9889aff00756bbdfd : SUCCESS in 3m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d796a4f93f824024a094987aae3dc01e : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169be70c17b74c498c9a23fc0efd1ffa : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13c6953e29ca43e08bdfc7cfbc4844b9 : SUCCESS in 16m 58s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/81750bc94d5141deaef5e3636c42fcc7 : SUCCESS in 4m 56s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/bc2a9d004f68499fa6cc9c2eab417f51 : SUCCESS in 4m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cdc4b8f9035d403c8e7a512b7bc5ba07 : SUCCESS in 2m 43s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d3da93f644d74df9b6868bfc29b69224 : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5114bf45b22845fa9c291998c99264c0 : SUCCESS in 40m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/94666e6e98b54fb2b64268005619baf2 : SUCCESS in 1h 37m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/524bf2815f0045c2926ee081af014460 : SUCCESS in 1h 05m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/49f2ae32f9cf4c1b8145479867612b8a : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/59e8ef066bcd46f3a5e3caaec5f75944 : SUCCESS in 1h 09m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56f65e09019c42bc8f5f017cff79a7bb : SUCCESS in 43m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f5bfa5a4eef4141a5c2afa7f7e1eda7 : SUCCESS in 43m 03s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/45b3194b2c1a4ed5bb0361195e96f8d1 : SUCCESS in 44m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd013fd673414db5923a5321d2596faa : SUCCESS in 37m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/74fc4deae6f945e7ad3cdd41753a197b : SUCCESS in 35m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e11bd43b3f5e415fa2dc3bd0a912ac84 : SUCCESS in 2h 33m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/da5c972f1cb043baaeac4e90a91298ef : SUCCESS in 1h 30m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/c9a9d1f0fb374793a534aeab18e8815e : SUCCESS in 1h 25m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/38891cfa930a48d2b15a8d4e88824bfc : SUCCESS in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ed72435015ac9f24867f406a99408248e7dfafeb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-12 17:50:39.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"57f528515f6503b55abb0ccf175c6a583558b683","revisions":{"207fcb7b2f5e58a802cae7dcc616d912b6bf3f58":{"kind":"REWORK","_number":1,"created":"2024-10-21 09:03:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/47/932847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/932847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/932847/1"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 09:01:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 09:01:55.000000000","tz":120},"subject":"Return amphorae OSError exceptions","message":"Return amphorae OSError exceptions\n\nIf we catch OSError exceptions when handling files\nor similar we will just throw a InternalServerError\nwhich makes it harder to troubleshoot. If we instead\nreturn the exception back to the task we can get a\nbetter understanding of what happend without having\nto look around.\n\nCloses-Bug: #2085090\nChange-Id: I97984cfab6eb29d15fb7b03da446914adcb7c524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/207fcb7b2f5e58a802cae7dcc616d912b6bf3f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/207fcb7b2f5e58a802cae7dcc616d912b6bf3f58"}]},"branch":"refs/heads/master"},"085a04d07e7fb361364a3ae171960187beb33ac0":{"kind":"REWORK","_number":2,"created":"2026-02-25 08:21:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/47/932847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/932847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/932847/2"}}},"commit":{"parents":[{"commit":"225f28e469ef4d848cbfa5ae0eec303f46ffa8be","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/225f28e469ef4d848cbfa5ae0eec303f46ffa8be"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 09:01:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-25 08:21:33.000000000","tz":60},"subject":"Return amphorae OSError exceptions","message":"Return amphorae OSError exceptions\n\nIf we catch OSError exceptions when handling files\nor similar we will just throw a InternalServerError\nwhich makes it harder to troubleshoot. If we instead\nreturn the exception back to the task we can get a\nbetter understanding of what happend without having\nto look around.\n\nCloses-Bug: #2085090\nChange-Id: I97984cfab6eb29d15fb7b03da446914adcb7c524\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/085a04d07e7fb361364a3ae171960187beb33ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/085a04d07e7fb361364a3ae171960187beb33ac0"}]},"branch":"refs/heads/master"},"d3abdc276e1d82607f915666e8d9db5999847d0c":{"kind":"REWORK","_number":3,"created":"2026-03-03 08:20:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/47/932847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/932847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/932847/3"}}},"commit":{"parents":[{"commit":"225f28e469ef4d848cbfa5ae0eec303f46ffa8be","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/225f28e469ef4d848cbfa5ae0eec303f46ffa8be"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 09:01:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-03-03 08:20:36.000000000","tz":60},"subject":"Return amphorae OSError exceptions","message":"Return amphorae OSError exceptions\n\nIf we catch OSError exceptions when handling files\nor similar we will just throw a InternalServerError\nwhich makes it harder to troubleshoot. If we instead\nreturn the exception back to the task we can get a\nbetter understanding of what happend without having\nto look around.\n\nCloses-Bug: #2085090\nChange-Id: I97984cfab6eb29d15fb7b03da446914adcb7c524\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3abdc276e1d82607f915666e8d9db5999847d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3abdc276e1d82607f915666e8d9db5999847d0c"}]},"branch":"refs/heads/master"},"8b10db44c7a1a53e4946913cbbdd20b876e0996b":{"kind":"REWORK","_number":4,"created":"2026-04-05 13:32:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/47/932847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/932847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/932847/4"}}},"commit":{"parents":[{"commit":"c86b945c21b7d5d34cf13a4e9d52810d657682da","subject":"Add SKI and AKI extensions to amphora certificates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c86b945c21b7d5d34cf13a4e9d52810d657682da"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 09:01:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-04-05 13:31:21.000000000","tz":120},"subject":"Return amphorae OSError exceptions","message":"Return amphorae OSError exceptions\n\nIf we catch OSError exceptions when handling files\nor similar we will just throw a InternalServerError\nwhich makes it harder to troubleshoot. If we instead\nreturn the exception back to the task we can get a\nbetter understanding of what happend without having\nto look around.\n\nCloses-Bug: #2085090\nChange-Id: I97984cfab6eb29d15fb7b03da446914adcb7c524\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b10db44c7a1a53e4946913cbbdd20b876e0996b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b10db44c7a1a53e4946913cbbdd20b876e0996b"}]},"branch":"refs/heads/master"},"3b2f20159c5281adb87c3efa494cca22a892b6e0":{"kind":"REWORK","_number":5,"created":"2026-04-08 19:37:22.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/47/932847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/932847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/932847/5"}}},"commit":{"parents":[{"commit":"c86b945c21b7d5d34cf13a4e9d52810d657682da","subject":"Add SKI and AKI extensions to amphora certificates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c86b945c21b7d5d34cf13a4e9d52810d657682da"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 09:01:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-04-08 19:37:04.000000000","tz":120},"subject":"Improve HTTP 500 responses in Amphorae agent API","message":"Improve HTTP 500 responses in Amphorae agent API\n\nEnhance HTTP 500 responses returned by the Amphorae\nagent API to include structured error details when\nexceptions occur during request handling.\n\nInstead of returning an empty response, the API now\nreturns a JSON payload with relevant exception\ninformation.\n\nIf the raised exception includes a `webob.Response`\nin `HTTPException.response`, that response is used\ndirectly.\n\nThis change improves troubleshooting by allowing\noperators to rely on controller worker logs, reducing\nthe need to access Amphora instances for debugging.\n\nCloses-Bug: #2085090\nChange-Id: I97984cfab6eb29d15fb7b03da446914adcb7c524\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b2f20159c5281adb87c3efa494cca22a892b6e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b2f20159c5281adb87c3efa494cca22a892b6e0"}]},"branch":"refs/heads/master"},"57f528515f6503b55abb0ccf175c6a583558b683":{"kind":"REWORK","_number":6,"created":"2026-06-12 15:08:49.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/47/932847/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/932847/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/932847/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/932847/6"}}},"commit":{"parents":[{"commit":"9ff4683c8212e4c043af69f1b5ebadc21651dc58","subject":"Merge \"Drop Python 3.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ff4683c8212e4c043af69f1b5ebadc21651dc58"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2024-10-21 09:01:55.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-06-12 15:08:42.000000000","tz":120},"subject":"Improve HTTP 500 responses in Amphorae agent API","message":"Improve HTTP 500 responses in Amphorae agent API\n\nEnhance HTTP 500 responses returned by the Amphorae\nagent API to include structured error details when\nexceptions occur during request handling.\n\nInstead of returning an empty response, the API now\nreturns a JSON payload with relevant exception\ninformation.\n\nIf the raised exception includes a `webob.Response`\nin `HTTPException.response`, that response is used\ndirectly.\n\nThis change improves troubleshooting by allowing\noperators to rely on controller worker logs, reducing\nthe need to access Amphora instances for debugging.\n\nCloses-Bug: #2085090\nChange-Id: I97984cfab6eb29d15fb7b03da446914adcb7c524\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57f528515f6503b55abb0ccf175c6a583558b683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57f528515f6503b55abb0ccf175c6a583558b683"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
