)]}'
{"id":"openstack%2Foctavia~933025","triplet_id":"openstack%2Foctavia~master~I2e8ddaeb3f0cdc1bbfc5766fe13b45b5165526ca","project":"openstack/octavia","branch":"master","topic":"drop-direct-exec","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-23 08:51:11.000000000","reason":"Change was abandoned"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-10-23 08:51:11.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-10-23 08:51:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2e8ddaeb3f0cdc1bbfc5766fe13b45b5165526ca","subject":"Revert \"Drop direct execution of octavia/cmd/*.py\"","status":"ABANDONED","created":"2024-10-22 15:48:06.000000000","updated":"2024-10-23 08:51:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":907017,"meta_rev_id":"9e94f520edeb8353c84fb5e9d23eb36465b29205","_number":933025,"virtual_id_number":933025,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-22 18:54:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-22 15:48:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-10-22 15:48:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-10-22 15:48:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-10-22 15:48:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b74026edeb1271e03520c47f318d5aaddc540e17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-22 15:48:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db74062cd4c67ec7e75b6f12f4d68ba5eb051d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 18:54:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a7ab6b5097341efb44438cee7ef808e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60aa19e05d274d3cacb11d0bfc081bb0 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f120dd8585574b3e8bf06509f799f43d : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb418c6174b846bfa024c3157e0ed050 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ea6bef5cc2849949110063cc34c8ba7 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e774791cbf642f392679414b2ec72de : SUCCESS in 10m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/22166ff490b8489bbc489b5df5ffe6b1 : SUCCESS in 4m 05s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e10cc25e28354171b3de08778b9dd772 : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/614fb4347c9942babca641270622332b : SUCCESS in 3m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/55e7a95cab18441cb0ba0b30dc9f1628 : SUCCESS in 7m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/970479c2e5a24671b4e7eba875361362 : SUCCESS in 7m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/97b7bb185e564f869c619bb242240fbb : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/663ed7e8ee234640bd5faaabed00934f : SUCCESS in 1h 31m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/78da40dc434d42b9a08e9a107db991e0 : SUCCESS in 1h 40m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/83a8198de3ff4e9e9d16406a67d9eee4 : SUCCESS in 53m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8834bab63c1745ffbcaeb7cc1f63300c : SUCCESS in 1h 58m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ffc72d12599e430da39be73bc880fe1a : SUCCESS in 56m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4daec6b262d24197a13394fad41e5861 : SUCCESS in 42m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e09776b991ac4b24bfd9971f077357e9 : SUCCESS in 42m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cce14706de46437c9231827dfaf7f187 : SUCCESS in 56m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/586161f8ff334811a1764ae06831d900 : SUCCESS in 39m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0893b5b0b22048d78f88a9e9228b4cf8 : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cea585e6167d43e18c3eba6a6093f40a : TIMED_OUT in 3h 02m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa1da67a6a434cb096417b5bfa215929 : SUCCESS in 2h 27m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/38be57efe1a847dc978f8c8dcac93e80 : TIMED_OUT in 2h 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e94f520edeb8353c84fb5e9d23eb36465b29205","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-23 08:51:11.000000000","message":"Abandoned\n\nUpdated documentation instead: https://review.opendev.org/c/openstack/octavia/+/933032","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55e90ac522e365a16bcae64489cdb2d98b9fc035","revisions":{"55e90ac522e365a16bcae64489cdb2d98b9fc035":{"kind":"REWORK","_number":1,"created":"2024-10-22 15:48:06.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/25/933025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/933025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/933025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/933025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/933025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/933025/1"}}},"commit":{"parents":[{"commit":"1e0f59f55833d404d7ebdb5a5f05763e876cec95","subject":"Drop direct execution of octavia/cmd/*.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e0f59f55833d404d7ebdb5a5f05763e876cec95"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-22 15:48:06.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-10-22 15:48:06.000000000","tz":0},"subject":"Revert \"Drop direct execution of octavia/cmd/*.py\"","message":"Revert \"Drop direct execution of octavia/cmd/*.py\"\n\nThis reverts commit 1e0f59f55833d404d7ebdb5a5f05763e876cec95.\n\nReason for revert: The change broke documented workflow[1] and makes running octavia processes in Python debuggers much more difficult.\n\n[1]: https://docs.openstack.org/octavia/latest/contributor/guides/remote-debugger.html\n\nChange-Id: I2e8ddaeb3f0cdc1bbfc5766fe13b45b5165526ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55e90ac522e365a16bcae64489cdb2d98b9fc035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/55e90ac522e365a16bcae64489cdb2d98b9fc035"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
