)]}'
{"id":"openstack%2Foctavia~933084","triplet_id":"openstack%2Foctavia~master~Ia832ec2e206b2efac504cc81240ee834935fc48a","project":"openstack/octavia","branch":"master","topic":"rate-limiting","hashtags":[],"change_id":"Ia832ec2e206b2efac504cc81240ee834935fc48a","subject":"Fix more mypy findings","status":"ABANDONED","created":"2024-10-23 10:21:48.000000000","updated":"2025-02-11 09:40:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7b28a945025eec4157a7cd001fb455aed2cb0e46","_number":933084,"virtual_id_number":933084,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-10 19:29:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-23 13:29:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85469c48e9fd698afd544b1493b80bb10ade859e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-23 10:21:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bb20180a62fc568db3bf4123286b0968ba91309","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-23 10:22:27.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":1},{"id":"d8c048c4fd8533f30c40118fa872b44405d97911","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-23 10:22:37.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":1},{"id":"6d778fc7de48a4db6ed78738516ba3e679e72172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 13:29:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d842bbdaa1844dff9047175afc4ffabb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8380f8c52e144f62a19da338420d675d : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6bb40a143149f7bbafaec5e7d2894f : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ae087e3554548ff9ce36eb2abcdab9c : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf0f69ce51af4ff8948ff0ceb8ccdb47 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a4c87a159b41efa9cbc184e4fa9957 : SUCCESS in 7m 58s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6833e0535c4a41fabe0101562b852920 : SUCCESS in 4m 33s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8a2e5e65ac064f1389e85707312c6899 : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f81320b99e4b4b72aace18f32dcd1e7f : SUCCESS in 3m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dbd7d672a54b4781b0c6b9d294f2d91e : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e516811a88e44edb89e1389fb4333a9 : SUCCESS in 6m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5f2c3417272c478bb058558a7f2511d8 : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4413f6eadfd143ac9560633c0e7e741a : SUCCESS in 1h 16m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1c9a0bb363f6470bafa6f52b8df0f2bb : SUCCESS in 1h 56m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/870659eec34c470c86febd1fe2c59da8 : SUCCESS in 1h 28m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/634b018c386443b6810bef8014622b9b : SUCCESS in 2h 12m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7ec0f17f13884bc09818f9ce519c16d4 : SUCCESS in 1h 33m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0758c02410a4eb59e1b633a9f9e3063 : SUCCESS in 49m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0a79172510f245a6ac60faed5710d667 : SUCCESS in 47m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4c9940f048d542a58cfbb251fca46caa : SUCCESS in 57m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd6b1e64d0e6468da8c861659bfaf77c : SUCCESS in 47m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/08d07bc0d9d84e67800abb6a788f8b79 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf7b48d37be941aabe0663e7cbe112a9 : TIMED_OUT in 3h 03m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc66e1f4e27644dcb9a64d5fafa2c392 : SUCCESS in 2h 09m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/adca74ed21b349aaa3cbcda76961e727 : SUCCESS in 2h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44f465ecc807835220a2b2017150e8d9720adaf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:14:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"983427527aedca8a2edce34f49474b47e9456eb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:45:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e0d448a230ebc0afba5052884d802e4d3ff4333e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:46:42.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":3},{"id":"91288ebe52085ce22960b5e41f3066d5265e9e15","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:46:49.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":3},{"id":"b6f8699ad49efdb65a83dc993dba7c9629eda6a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 13:21:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"6fe01f4995810a53063a6aadbde0b1cc07991942","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 14:49:51.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":4},{"id":"252ff286d64b8ee641bea8e13b4c3b9bf646b3b0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 14:49:54.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":4},{"id":"20e571060b0af82deaf796b9b0cec98394caf06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:46:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d77ae33fc2a546f393fca83f2595437d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f05b322eddcf489983300b7a9a40cea7 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50f0b9c125394f86ab98e42b0867e730 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5646283a17dd4fd98237a0013671cebb : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d31bbf78137647f9ada07b0a27ed6cd4 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa5a0a0329d461ba3a95ba17b7a613b : SUCCESS in 11m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/07d58c1d94854e4da26d3d91e6e0022d : SUCCESS in 4m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e7441485ce8a4fb88168da3e34753f22 : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6eb0b564533d427da6f6a0ac8b3a6c75 : SUCCESS in 4m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4868e74db41c48adb222b8c2d6002390 : SUCCESS in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd0f12ec7a194de5ba9f22cb6e286b80 : SUCCESS in 8m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1ac7b8a376f0464c8f1430189396d6d0 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/538ee715ef2040f48835a30ad1979ba1 : SUCCESS in 1h 22m 35s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/23c04e10aa784a438175eba529152c44 : SUCCESS in 2h 30m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5093b27ed7754fc6bb38811e89aaf02e : SUCCESS in 1h 17m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5c9a5984da2140559762a29af40d1232 : SUCCESS in 2h 30m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0e4755ae92b4476f9c2074e138ffd094 : SUCCESS in 1h 11m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8469f7dfd8eb4d349c8d4504cd489a2b : SUCCESS in 48m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/376cc663786645aa9870f4804885e929 : SUCCESS in 46m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6d514db8dbcc4ff4b7c22abb483cb351 : SUCCESS in 44m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3515cba05ccc443582bc6150380e49e5 : SUCCESS in 37m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/41ea894f3b5147b4aff80cf21ce9732a : TIMED_OUT in 2h 34m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b3e426464b347d89a64f9f6920fcca2 : TIMED_OUT in 3h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/853d466d37e043f782b97287cad9babf : TIMED_OUT in 2h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c52f08ce1c8646ad84bda0af28aedf8e : TIMED_OUT in 2h 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6b66d3ccc68da78631393bd84182a817455484f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-20 12:07:07.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4e883df5f88cb6a44dd53dedbdf9d85316c6e4ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:58:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece698f0c34848ba958154976f0b9038\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba289ebae03f4edeb9387bd0b32c0fcc : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec120676d8154b3fb6c6875c34f547aa : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/173ca701799040c28126c4e17aa6f5df : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e018978b150d424f8f6644aec493e891 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0266455cbbbc4b6ab503b50eb9554563 : SUCCESS in 10m 40s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/89c7507a6eae4452beb669a6c707a5ec : SUCCESS in 4m 47s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a43b956cbcae4b09a176a691b8391fe0 : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/029329c3fd354ec6bdcfe2c159ef08c6 : SUCCESS in 4m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fb7b6155bfb4ebca202c7b991534c66 : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f5d32c7eae645f49bcdd2834185d537 : SUCCESS in 7m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3e7be9d118be4401aa66626fb90750f0 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e661dc7da994826a6eed00327ef7d6e : SUCCESS in 1h 21m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/16ae4670e1bc4a0293f1ffb1534f6be3 : SUCCESS in 1h 44m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e0d1256893984ec587447069f9478a2d : SUCCESS in 1h 00m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7a3a3a1a970848258ecb1c334644c6ac : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eeeb6249aa4b46ef9781fb7ece071dfc : SUCCESS in 59m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c085382477a46dbab9f649272da2929 : SUCCESS in 52m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7611379385c84e25b3a5c46d651b2e6c : SUCCESS in 59m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7a03350f0ae44e0f97520ed4bc8accef : SUCCESS in 1h 11m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b5627e88ac543cab1374a3bd37f3811 : SUCCESS in 45m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ea58e564463a4278bf95a502021fcfed : TIMED_OUT in 2h 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c097d1881b541d0bdf5dff5c8735713 : SUCCESS in 2h 48m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e2df4f1e3b064753ac6996374d2f4351 : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d9f8bef3f46f4b57a94df7ec96e094ba : SUCCESS in 2h 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d34f257a91bcc2c2166b83e4a68b583c4482bef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-10 15:47:30.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"62fc92fdc8db83b7f2fa6d6dc5b03f66be2caf66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 19:29:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2799e18f67d4890b21db74e3bb985ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/252a9e4ea91540b08edd28fca5f2bc31 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450535bb6ede449bbae2f1efb6097307 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d80c4e9745d847259a8d5ff5c51a36e9 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2f2630f3b9f4e82a4dbb70c93862376 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea345098e54346cea75d0a9b42fec921 : SUCCESS in 10m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/128699abd65e4d80b2ba4e1c7c73623a : SUCCESS in 4m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/20bfb07d6e7549328b9c0ec451a363c4 : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c1174785ea3430aae64cdb89c6ca33c : SUCCESS in 4m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/efff1e697e9b47bf99f16a15838a394f : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/34e1c24369234f04ab497a3686815250 : SUCCESS in 8m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/af32e5ba3ce943b693dd076461600483 : SUCCESS in 5m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d72bb2153782478f92743a79731d3ff2 : SUCCESS in 1h 19m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0bd2aeaa3d584c31b67e11c6a0847b45 : SUCCESS in 1h 46m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/16afa756b6e843c5b31361443282ba03 : SUCCESS in 1h 14m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20a0a3d38a464182a7d72dcb1f72e74f : SUCCESS in 2h 07m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afd2e02d795842899367bac9a3acfbf2 : SUCCESS in 1h 14m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3587e104b910491cb0481f888229f9e1 : SUCCESS in 2h 11m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/090f9c54108e48ecba1e3316f979120a : SUCCESS in 58m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0287c17cdca4c39949b55e7b1cf6e62 : SUCCESS in 43m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66e824f79a79466284f4b34da473f86f : SUCCESS in 45m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1727a81e2a544a0cbb8137dc65cf476d : SUCCESS in 1h 08m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6efb631f63fb43f6bfc632fa9a97aff2 : SUCCESS in 42m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71a73260feca476c803dbc6331d6ebe4 : TIMED_OUT in 2h 35m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/50e04f0a4da8444daa0729dd8f1c8285 : SUCCESS in 2h 30m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7fc9eb0c47b54c0cb36d865bb9b5dbe5 : SUCCESS in 2h 31m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/74c7502823ce400ebb665ce68235f1ee : SUCCESS in 2h 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7b28a945025eec4157a7cd001fb455aed2cb0e46","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-11 09:40:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c1bab686f26a9b8f8c67558650e93da42cff94db","revisions":{"de4a91298dca9b0258eb26f806b59351e1c9101b":{"kind":"REWORK","_number":1,"created":"2024-10-23 10:21:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/84/933084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/933084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/933084/1"}}},"commit":{"parents":[{"commit":"6e914a5b4c3464e378a7e71d18b38601afcde335","subject":"Add typing information to remaining octavia code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e914a5b4c3464e378a7e71d18b38601afcde335"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:20:35.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:21:38.000000000","tz":120},"subject":"Fix more mypy findings","message":"Fix more mypy findings\n\nChange-Id: Ia832ec2e206b2efac504cc81240ee834935fc48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de4a91298dca9b0258eb26f806b59351e1c9101b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de4a91298dca9b0258eb26f806b59351e1c9101b"}]},"branch":"refs/heads/master"},"62c93961b91370d404d3ecd0c97954505784ce7b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-28 11:14:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/84/933084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/933084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/933084/2"}}},"commit":{"parents":[{"commit":"e9935333e0c0b6aec657ae99d106e755400bc25e","subject":"Add typing information to remaining octavia code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e9935333e0c0b6aec657ae99d106e755400bc25e"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:20:35.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 11:14:24.000000000","tz":60},"subject":"Fix more mypy findings","message":"Fix more mypy findings\n\nChange-Id: Ia832ec2e206b2efac504cc81240ee834935fc48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62c93961b91370d404d3ecd0c97954505784ce7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62c93961b91370d404d3ecd0c97954505784ce7b"}]},"branch":"refs/heads/master"},"48df8a672b4730b7c32167a7b2596f9a608fb8df":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-28 11:45:44.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/84/933084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/933084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/933084/3"}}},"commit":{"parents":[{"commit":"5ccac5f12654a276e8547f6eb86dff4874e38ab7","subject":"Add typing information to remaining octavia code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ccac5f12654a276e8547f6eb86dff4874e38ab7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:20:35.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 11:45:35.000000000","tz":60},"subject":"Fix more mypy findings","message":"Fix more mypy findings\n\nChange-Id: Ia832ec2e206b2efac504cc81240ee834935fc48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48df8a672b4730b7c32167a7b2596f9a608fb8df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/48df8a672b4730b7c32167a7b2596f9a608fb8df"}]},"branch":"refs/heads/master"},"24c017a2c51e8d71796627c7c103abc188d5e790":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-28 13:21:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/84/933084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/933084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/933084/4"}}},"commit":{"parents":[{"commit":"e5500b5e03d09ca71a336bba199ba5407f9b94d3","subject":"Add typing information to remaining octavia code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5500b5e03d09ca71a336bba199ba5407f9b94d3"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:20:35.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 13:20:57.000000000","tz":60},"subject":"Fix more mypy findings","message":"Fix more mypy findings\n\nChange-Id: Ia832ec2e206b2efac504cc81240ee834935fc48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24c017a2c51e8d71796627c7c103abc188d5e790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24c017a2c51e8d71796627c7c103abc188d5e790"}]},"branch":"refs/heads/master"},"8d55b65351e867b9e239f9cc9cb1f64264a06e5c":{"kind":"REWORK","_number":5,"created":"2024-11-20 12:07:07.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/84/933084/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/933084/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/933084/5"}}},"commit":{"parents":[{"commit":"0b4a15144be58d9d890c1120dad054f0523eabc9","subject":"Add typing information to remaining octavia code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b4a15144be58d9d890c1120dad054f0523eabc9"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:20:35.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-11-20 12:03:24.000000000","tz":60},"subject":"Fix more mypy findings","message":"Fix more mypy findings\n\nChange-Id: Ia832ec2e206b2efac504cc81240ee834935fc48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d55b65351e867b9e239f9cc9cb1f64264a06e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d55b65351e867b9e239f9cc9cb1f64264a06e5c"}]},"branch":"refs/heads/master"},"c1bab686f26a9b8f8c67558650e93da42cff94db":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-02-10 15:47:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/84/933084/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/933084/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/933084/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/933084/6"}}},"commit":{"parents":[{"commit":"812e057f38945c836c5d5fcb1b8877b55b248a2e","subject":"Add typing information to remaining octavia code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/812e057f38945c836c5d5fcb1b8877b55b248a2e"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-23 10:20:35.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-10 15:45:07.000000000","tz":60},"subject":"Fix more mypy findings","message":"Fix more mypy findings\n\nChange-Id: Ia832ec2e206b2efac504cc81240ee834935fc48a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1bab686f26a9b8f8c67558650e93da42cff94db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c1bab686f26a9b8f8c67558650e93da42cff94db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
