)]}'
{"id":"openstack%2Foctavia~933254","triplet_id":"openstack%2Foctavia~master~Ia8b4612cd25a68b44d55b113f704756818eb8a5b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-11-11 16:04:40.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-11-11 16:04:40.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-11-08 12:54:35.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-11-11 13:56:12.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ia8b4612cd25a68b44d55b113f704756818eb8a5b","subject":"Remove amphora_health record on revert CreateAmphoraInDB","status":"MERGED","created":"2024-10-24 11:26:17.000000000","updated":"2024-11-11 16:05:50.000000000","submitted":"2024-11-11 16:04:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933254","meta_rev_id":"45ea2964dd506cc72d73e5e795435851909a3ed8","_number":933254,"virtual_id_number":933254,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-11 16:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-08 12:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-11-11 13:56:12.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-11-11 13:56:12.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-08 12:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 14:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-25 08:11:14.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-10-25 08:11:14.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-10-25 08:11:14.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"64b7405a046988be034f74212553219a440604fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-10-24 11:26:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"102fd8bd80261fe281f897223026b714a480d538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:26:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39e92267484a4987a6a4fb30eaa57313\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01b0f243bbd54c4189da9091eda07c73 : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3786234ec143a89c225261c98e9cfa : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1051247f695f48a58604269d9c19a012 : FAILURE in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6a90afcce27496c8c978953281f9803 : FAILURE in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26a4eb1c0604ed5815edf4d84149278 : SUCCESS in 10m 51s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/eacfd0408a8b4bccb2173a1bfe4f65ea : FAILURE in 4m 43s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/cabf565860bc4044bc4f9c62fbf95cc1 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/47c3ceae9ddd4ad69fe0a0092071e589 : SUCCESS in 4m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7bc6e81587342f782c6f41a62eeaba2 : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa9d1d13f9294d01bd24a1e2bc4e4670 : SUCCESS in 6m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/18c49456508d41409a19ca8979686d31 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ebac187e9424415cb895b089af96842b : SUCCESS in 1h 33m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/829bad44275e4ec19b082c27b3b71fdb : SUCCESS in 2h 13m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/01ef2f18cea44b759d3079850301ea43 : SUCCESS in 1h 25m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/88a39463d953456f857fb51e8883e6b3 : SUCCESS in 2h 19m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/97a8735ab7794687a771021eb4ef79b4 : SUCCESS in 1h 00m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2f6a6701955487aa631f560c5fbe360 : SUCCESS in 47m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a621791403f4161b46894775fb3764d : SUCCESS in 55m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6bdd77b38b1e42139c016b343f2d6ab9 : SUCCESS in 1h 03m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d416bdf4f9c4c2191c70d0b6f7bc2d7 : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/80c9a12b6bf841598a9c130c1acb3d13 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8be16e30f1514637819e26c24733f682 : SUCCESS in 2h 55m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7c51c7c8c0434b87933ac57d798632 : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7f529f76d354730b5dd39e9b0f848da : SUCCESS in 2h 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16a76c3198e8812002e74065f2eefb3d8bac507b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-10-25 04:44:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ec2aa3c7b6af12e0597ca1c5c1d8de307ede8882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 08:00:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/397552890ee54a17b13736b0ef05cdb1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b17428b0d1c403b956e9ee3ff5eb524 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d4ac456dd55497cbba7a3fb0b392006 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a630816c6f774e6e93a0f1c7db92fc3d : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33e8214f860e4fc8a720bb50d96d5808 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0f286a1d5a472e9d8897900bd628bf : SUCCESS in 8m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cba0d88b99a549e59191350519a300ce : SUCCESS in 4m 09s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/9f6cc3b16b2c4152a1fc1b6f1454da07 : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f1149bb090234e358dc08bf850b029ed : SUCCESS in 3m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b254c7884c77455bb5dd0cb7f762e9a2 : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/109793c6a48e405094945c5c39f1e512 : SUCCESS in 7m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16601b98f3f443689c474c3e1763f784 : SUCCESS in 5m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89213062025c4e8499978f2ec2d1f617 : SUCCESS in 1h 34m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cf19d3520764429296ff041d9a215e42 : SUCCESS in 1h 30m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6169f2fb06da499c923dc962ec6a41f7 : SUCCESS in 47m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/252f50faa0314e728f10b72044540a8e : SUCCESS in 1h 48m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cd0d1e46ee3c42b788853edfed72e730 : SUCCESS in 54m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82fabc992a8d46b29fc1d574cdf13226 : SUCCESS in 31m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0d9806433f94b4ba492c3f0b8f7da9e : SUCCESS in 37m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/55d5a040c19940668dec63bbda5deca1 : SUCCESS in 1h 20m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/640865b1b0ac4c9a9569564f7a9b7622 : SUCCESS in 49m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7749fb72d5c644f1957ce8f154da5909 : TIMED_OUT in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57e8fa0858614536913f7e861de79aed : TIMED_OUT in 3h 04m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a160a780fceb4a69ac3bbcb2b03a8381 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a36cae2d58b4a16b32167dd3ca96021 : SUCCESS in 2h 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4f805ed253261de47b5ef993835f02ba380b887","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 09:17:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e6db1112f6c97a87cd3496be0199170332062a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-10-28 16:56:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3c4b6ffb4dcda35fb0c52df0d231106d2557712d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-10-28 16:57:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03924f5f0dc4fbd6e1785064ae247877203807aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 20:00:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4d9e8b27b9546aaaa0b6f71bf1bec0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff1d0614b485442f9c26f3a68d8c9b30 : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/304be680fdb94417bdd110045117ecc1 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06e34c7ebf87418b90a166e8e8120d6e : SUCCESS in 5m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c60251892dd84739a52d8db521aeeb0e : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1bb24e8143c4f6e997462e256e40bf0 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d664d39a87b4bcb9bc64b8664b54917 : SUCCESS in 5m 30s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5fc9b69cdff9415dbd33b6297bbdb89b : SUCCESS in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/edd704acb2cb43b89034ea8055af0b06 : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/782be09a0ef248ce9da307059bb4eed9 : SUCCESS in 4m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/49d89f36e1df42bc8df2ec7004dd897d : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfd4a1bc142a44e8b063bfad9cd7e04c : SUCCESS in 8m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e3f1ba58087e426187b798428df7b28f : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b38ddbd4f77249f8ab87bdd1200a7a96 : SUCCESS in 1h 31m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fd05208a07e24a04a30949974684ea9b : SUCCESS in 1h 43m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f705cbd9f994467eb4dd57bd7ca3c8c1 : SUCCESS in 1h 28m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7bae073580074fa6ac835f74b3d285ab : SUCCESS in 1h 57m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a7b36ac93b1a4334bfe08cb55c9c8cc8 : SUCCESS in 1h 00m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/603bd94caf7d4c06b4202aa08299d2cd : SUCCESS in 31m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/167a7d0f6a4f42aab60fbc9eabcde9bc : SUCCESS in 59m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2c17625bfc67443b8bb320af3b4cd780 : SUCCESS in 1h 00m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c31aed33492f4635a21f02220453dcc5 : SUCCESS in 34m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3642c09b37e84185a12491d33a5b34f5 : SUCCESS in 2h 17m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42401ab5b7654aea84dfc7373e1ca1cf : SUCCESS in 3h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d1995856607f43f48ba92d7502acb60c : SUCCESS in 2h 10m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d116fe053a6b4b54943b3d4673a58ef0 : SUCCESS in 2h 08m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed7e79f72268d5e99ce3c1b96edaf1e9c87f112d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-29 16:44:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7859db52c0d53696a04e35099f2236d6c6e43cf2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-01 14:13:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4673d66db13ec76983dd3953ca0183f64574ca61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 14:55:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/571ebe298b7c4a739701e6541f356534\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b959f46ffa4f36856dc4213d845123 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3730e82c83f94579b2a0f00def7edc95 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89c3630b6d84402c887e30e6ecefdd93 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a31ecba9952d46338b6813d61ed84d49 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a82eb34a8534e07987d5f27b6f2feac : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1adc3b77ab754d44805df9f9ddf6a4a5 : SUCCESS in 5m 01s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/831d58052ad54e29895017605e54b5de : SUCCESS in 3m 47s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/25be804fe98540768868884610f98f9c : SUCCESS in 6m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7fbcd6cc3bf4466b841f17bf4a60357e : SUCCESS in 4m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/215ff25dfaaf4a0c80668cfb80096c65 : SUCCESS in 6m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea14aae648e2477a92c89fdf0e3e2f7d : SUCCESS in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/10f0716637f2427eb41a2ac617a74140 : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f79e1131d264a2680f72134c2318806 : FAILURE in 30m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/76e4cb5b4bfb4ecfa3c4a23232132b0a : FAILURE in 27m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e43e7db2407e4e27a6802217335bde19 : FAILURE in 27m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ecb5066f22f94249b5daf5671b9ca401 : FAILURE in 27m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c25388ff253a4ce3a614a127027a5bac : FAILURE in 26m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b0583f66d5964c548b2a7e4c92af9d6a : FAILURE in 25m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b178f5a8a0dd443a89922aba09af39e3 : FAILURE in 38m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4c696ad79c554a18ba7c00ee7b179677 : FAILURE in 40m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2cb9d39b14b42d1920fc34ebaf16cbc : FAILURE in 26m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c1ce0063283c461db134a622f5594e01 : FAILURE in 24m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6993449cfabc40a5b0045935e56039ee : FAILURE in 35m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55cb7306f9d94b44984371369b21ca83 : FAILURE in 23m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b62bae7ca6049f48301a632845dcc03 : FAILURE in 31m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"24b4d411c244bcbc71849230cc7256eaab10d65c","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-02 07:46:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36dd4ba5d78707b1ef024d09a2755a989ecc664e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-02 10:22:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fd1862933f14b31b3d95c356d021391\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bf818f547ab4bf5a335ccd5d4a839d6 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0630fedfbb034693a8a6aba3cafec666 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66613a554c2b4c64856e38107bd1de20 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81f842bfd5924bc18863df268a6ab30f : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981f9f4c2d4b4c8b94092a4abf2b9423 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f60b5b2f5ba641d188b232f76393660a : SUCCESS in 4m 45s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/826c785239194893b08edef544488a8c : SUCCESS in 4m 25s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/711f271450f54b61b0fd4d9710b836e4 : SUCCESS in 6m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/68cf91be35c9450291a9d5fec97715f6 : SUCCESS in 4m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa9eeeeb5b5a4905936a8dee781f11cd : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7de7a98405a1475593bb6261fb5e8bab : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2ca992bd3dd9410383dec53cd94014af : SUCCESS in 5m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/200e76bc736d4a7d83a065bbcd451753 : SUCCESS in 1h 19m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2a28d52e2db7487a8a08c2de4454ee84 : SUCCESS in 1h 35m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/47362f66ff8144918e5f0adeb91fbf3e : SUCCESS in 1h 13m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/df824d400a2d4cdebda4b747eef21bf4 : SUCCESS in 1h 38m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac2860f7d75446f5b7220a76e86c6a42 : SUCCESS in 1h 23m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/94039ca952b54f5d883fdf8b846abf7a : SUCCESS in 31m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8018e4cf05584aaf85f158b38944ad8c : SUCCESS in 45m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1fbe8e1eb7c24e85a714004b9602d79b : SUCCESS in 44m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bbaed43ec9a04a7c8157b9ea5676e309 : SUCCESS in 34m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e80b386bc0b540fd8afa3e25e30f21f6 : SUCCESS in 2h 03m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a1169759e74648cd91159e679d4688a1 : FAILURE in 1h 59m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/99ba92bbb5cd4015a6165c611c39e383 : SUCCESS in 2h 26m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/448596a15fe24c9cbbbe8997c25fbba1 : TIMED_OUT in 2h 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1946ce9118f3b49e715155d22f02dace8373fc04","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-05 07:54:00.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"693218c1f3ef8a2f4bcdef62f2d2710ad249e1cf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-05 08:00:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e255f611cf5146bafb549fb6565590e839fc31a","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-05 08:29:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0872cc5d6bd7634613fb48cf2749c65b092830cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-05 08:52:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d57cac5bc2adeb521bd19fee3be25aec0368a9c2","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-05 08:53:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6d016847fe69c9d3b118d06585cb945d993089ac","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-05 11:46:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"78d4e58f9450560d830c255dfb64229c48215c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 11:57:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b11e886e49041a6aefba4780886fae1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c743ffcc84948bdba4861246f616ab3 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d47f8e6aee04635925d9fa6ffe0a6be : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/165fd1173d8c4ccda1f5f0c8ce36f3e4 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03642ade41fd48a1921dfc7445067fc0 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35976f1f85a644298e1934c2fd6ea6b5 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/721fa3d2799f44dfa8abe2aa5a399612 : SUCCESS in 4m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/500af7c178cc44b3b5b0f216c75d3ac4 : SUCCESS in 7m 05s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6e1e38f59c404acd8e387ef9f7a1662f : SUCCESS in 7m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/12c06b8e3ec942d8943d0623c64ec312 : SUCCESS in 4m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/72f9a3f78e2c4809884e214180f79306 : SUCCESS in 10m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0f328e7aa924fb395c48165fa7deaa8 : SUCCESS in 5m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c2121d0f55894abdbe3e40896055a371 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e87b1d15af8c49b09035e0d35d3371e9 : SUCCESS in 1h 25m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/db297c5d877f4dfd910d9599feebcde0 : SUCCESS in 1h 47m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/145c28a4a2644546a651e175c27404c4 : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/58713c3769a2462d88a43e94921152d6 : SUCCESS in 1h 50m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f70ac32233df416fb11901be9fae600b : SUCCESS in 56m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b9b53c34baa45a5b0213076808c55ba : SUCCESS in 34m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c47715bdfb3e437a8257ff1e500cf993 : SUCCESS in 33m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b2f41a3b938140aaad03b116549af713 : SUCCESS in 41m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48ceb42514ee49829d6726b1ac7a64cd : SUCCESS in 39m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7ecb4a52a8154663bb98cb9be0f077fc : TIMED_OUT in 2h 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6120cea23ec5466482fb6bb373ac174c : TIMED_OUT in 3h 03m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bcc0535a87f54ba8a8a3dd9f8425e0a8 : TIMED_OUT in 2h 32m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/70733fa0ca7d4e09aa30444b3cc476aa : TIMED_OUT in 2h 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e75115d33a8ceceb48d0567966516c977009e8f9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-05 13:57:48.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3fbe337b4a10e45af7455caed0c77413e9630e09","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-06 06:39:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"17b24409cdce1d95501252f11d7ba924ad36180f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-06 08:28:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2a9d78fefaeb324557df0851c26d6375cc74e8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-06 08:45:33.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2c84c8525edd88ce974bd43367c66ce82fada00c","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-06 10:36:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0af1c4bf6328e63766fe437e104d2a615032d115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 11:56:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bd6fa489f1347d6bc588669ad2c488d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/462ef4a5b9d546b9b14e9e9cfcd1fa02 : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12852b1ef0294ac5862c1a035863136a : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50f829c654d4789942e5801df4d891a : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26918a8cf93641cc9b89cf1210752c4a : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c90974dec1284125be4eda067fd7ad22 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/860bb538698648d4bac048a5e10c4be7 : SUCCESS in 6m 18s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3614754994c54aa69b5f7af6075fc60e : SUCCESS in 4m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/59f6317322b94624b138c911cf57c1c9 : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d2df2d93483a41bcb084558e5eade51b : SUCCESS in 4m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ad0f0b81c544f9bbfffa36af34032a9 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e557ab94c2d142b8b056b7e4218df770 : SUCCESS in 7m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/17abed4388a545749aa52c5282197422 : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7321c598fd944197b22f2f98b30a29c2 : SUCCESS in 1h 19m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1d2f62f4729a47db94db65ebbe93d442 : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ae219400510b4c5984b0dcfd88595656 : SUCCESS in 54m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2cadf35e84c24eaa9eaf10c1feb2442a : TIMED_OUT in 2h 34m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7ba422dc2fb740d4b7eb66c675201a0f : SUCCESS in 1h 21m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce760b254e984b32ab1bae0be6035381 : SUCCESS in 41m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b1894f35409a43c594e9729cd6ee4080 : SUCCESS in 31m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e196fc6ebc214158a6a215b31f7345a7 : SUCCESS in 1h 09m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c278b7ec3e9440079d9b9b459b956358 : SUCCESS in 39m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9ab3173c27614e57ab7d0c01910780f8 : SUCCESS in 2h 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e71de8cc9d8f44afb2001d4cfe2fe0df : TIMED_OUT in 3h 04m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f89cd9ed99a342e69f7fdd7a5be050db : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/719264f94d9248628f43c8580684f06f : SUCCESS in 2h 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6a9af1d3ed1ab61a98332f2b0e81e67ca214e355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-07 04:31:32.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"fa1172ebe18393be8a0bda5676fd5f53a4052f6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 07:13:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60bc9db553d149929e641d73a98693c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1f978d945ed49ec86cc8df98a37fc27 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e816ba32ac7430dbdc3fa0a86de0a74 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69e1153efc0344a584a9538ffc4571dc : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72b1f45776c447dc911f580db8aaa9c0 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe1eb1ac8d44ac68c205778f8a2e474 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3f3b90031d04a7ea1a5d47143b3c21a : SUCCESS in 3m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/309b8416956e413c89301de90b8cbfd3 : SUCCESS in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0097b096ad5d476facafcad870666855 : SUCCESS in 6m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c595fba398cf4f4cb7ce2f8e0bc0964d : SUCCESS in 4m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0de518bf46794550ba19cd94af39151e : SUCCESS in 6m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/130f3184e730458d9863fd63f2716333 : SUCCESS in 8m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ac6f2b7aa5614fb1ae1308b4fa3f9629 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1d3a0bd49c649b3a65b5c632fcd4496 : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/10d8c6aff9bf4407a0152520ee8e52b1 : SUCCESS in 1h 52m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f976dd278b0b4e84b26dc75a34335284 : SUCCESS in 56m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/43c4d6fd658e4142b932fed115c94a64 : SUCCESS in 1h 33m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/26a946666a3e42a2a172311071720f07 : SUCCESS in 1h 02m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5da247a5380e4c37a67488b52b9f8e05 : SUCCESS in 28m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee9d6c4a01cd41e494080d77208aa780 : SUCCESS in 37m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5529f39f26aa437fa7bb5bb0a6fbdb18 : SUCCESS in 48m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09ed0c6b837341bc886e0b1e76a05707 : SUCCESS in 31m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2ce704667c4b475195a0fb0524d6742a : SUCCESS in 2h 13m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c2522c9f4544ba59227cb49896e9d1e : SUCCESS in 2h 40m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3c554919a84956b79a06fe2436c246 : SUCCESS in 2h 20m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd82dbcf89ab446ba9a9aa09eb8d0eed : SUCCESS in 1h 58m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c437955c5b4774397bd0128657b016672b3d7397","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-08 09:35:59.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e86f6fc9b94e9891cb0a64d909bb07ca3801b0ab","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-08 09:44:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2176be4d0ba22a9141688e2b4b5b4d8091714718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-08 11:11:32.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"ecf8172189c304ec715cecfabe449417345e885c","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-08 11:11:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7e3fb88ea64e36b6d214c679a1ce450c2f1fb7d4","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-08 11:12:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fb888a2796d1991d4359a3135cbd2416bf67e2b8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-08 12:54:35.000000000","message":"Patch Set 8: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8a56da3e5828c30d834addb27348f21c34da1661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 13:53:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f41c745b88c3477996f2ae1012e84bb2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fac15dc7d8243ac807d65c146ecd2f8 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecec1dbd49b54b5b8f155516fde0eadc : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83e5fb9663e14a03b4147c5607162895 : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b913248f659f49dfabd393cf15e6b22e : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2db07da5e2f4d31b98aea13ece8c987 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c2dde68006e43ae909ecc24e24ac69a : SUCCESS in 4m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fef42061d4ff4bf5a6165da10e7c181e : SUCCESS in 3m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f5ebf6f07a8d416a9e5b6f7b3ec4d101 : SUCCESS in 6m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/62bee60c99dd4ccab3cdafad7ed0e892 : SUCCESS in 3m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/179130af67d94d74a468e51ea22b453a : SUCCESS in 7m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8178ddf540b744c2bad163cde15ea669 : SUCCESS in 7m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c95cf65135e1488da7af1bf10c8d297a : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d538e3fbd434034bb7e10d633fe6746 : SUCCESS in 1h 11m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dace77ea655e472387677491c14488e2 : SUCCESS in 1h 54m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/071125eed715457fa08a43d3b90c1d59 : SUCCESS in 1h 14m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/226050ef0e8a48a88e0100153ad08ce3 : SUCCESS in 1h 51m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a1f5db2077794bd6a6585e6aa4ccc97c : SUCCESS in 1h 12m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/350766b0b7f8444f91ac8a2f90504638 : SUCCESS in 37m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3363214660454444b54931bdf7f159da : SUCCESS in 51m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/482c15f49be649acbaabc498550ca109 : SUCCESS in 57m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d892c86784842a69eb7074e5e2e2aa0 : SUCCESS in 36m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/045aefb16af74a6f802bc2a42d65fc5f : SUCCESS in 2h 26m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d27f65e5a644fb3a3eca8344a2ed116 : SUCCESS in 2h 40m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/965d388fd2dc44268d8823510ad3f56c : SUCCESS in 2h 32m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/feac565c873641edb79323f1cdfd323e : TIMED_OUT in 2h 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8389aa02163a36b8ed758b9f0eb1f60a5e7d392c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-11 13:56:12.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ac1c179364bd0cd5071110c56cd1bc5eed2e4963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 13:56:48.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f2b5d3fb1ce2109f374768bdb93e1ceca8b33e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 16:04:40.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e23805e231b44b23815ead446d031f06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7992a01571434ff8bf96bedbe77b94f4 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42def3800d174b7baa402280f873f9ee : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/163f2de09611451691d2218ffac7f917 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8874ac13043d4485b59bfe99ff16486d : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b391cbd25b6c44d988c23f3b1f82da83 : SUCCESS in 4m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/43804951a4f84b3a87a97480680dd25e : SUCCESS in 4m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e366d680f70a41daa0e8009abc462dde : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b746224c2acf40db883e006b08daacfe : SUCCESS in 1h 14m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b12b5040f264493c8fe9ec6e81dda8ad : SUCCESS in 2h 07m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f23fd7638cb946259fe4872d6ccb7007 : SUCCESS in 1h 02m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be3fb4a68c634d2699eebcfad388a563 : SUCCESS in 41m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/095c2a57f50c4b42a130f63b886bf960 : SUCCESS in 46m 30s","accounts_in_message":[],"_revision_number":8},{"id":"b59ed3e442322987ac7d9c0162d8ce8cf0b78a93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 16:04:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"45ea2964dd506cc72d73e5e795435851909a3ed8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 16:05:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc0cc0a2bcb7463fa3661886713c2ffd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/802830deb9e344a0a7534ea5037886f9 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdb2bb06b24a43c0be6d429e4bf94d8e : SUCCESS in 38s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"f76fa1ae330e933e86c666b857acfee06f675adb","revisions":{"ea961572aa1915546caf5060942a4b50d4717e7e":{"kind":"REWORK","_number":1,"created":"2024-10-24 11:26:17.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/1"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-24 11:26:05.000000000","tz":240},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea961572aa1915546caf5060942a4b50d4717e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea961572aa1915546caf5060942a4b50d4717e7e"}]},"branch":"refs/heads/master"},"bcd71636cf4b5e0d6bae0c402ef431ef86cdc98c":{"kind":"REWORK","_number":2,"created":"2024-10-25 04:44:03.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/2"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-25 04:43:39.000000000","tz":240},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcd71636cf4b5e0d6bae0c402ef431ef86cdc98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcd71636cf4b5e0d6bae0c402ef431ef86cdc98c"}]},"branch":"refs/heads/master"},"010b4bb15cf0cdfc03cc6097aa5303ace6d1520b":{"kind":"REWORK","_number":3,"created":"2024-10-28 16:56:46.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/3"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-28 16:56:33.000000000","tz":240},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/010b4bb15cf0cdfc03cc6097aa5303ace6d1520b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/010b4bb15cf0cdfc03cc6097aa5303ace6d1520b"}]},"branch":"refs/heads/master"},"798a3a95402dfb4420ce6ca4333528161cdc44e9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-01 14:13:56.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/4"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-11-01 14:13:56.000000000","tz":0},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/798a3a95402dfb4420ce6ca4333528161cdc44e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/798a3a95402dfb4420ce6ca4333528161cdc44e9"}]},"branch":"refs/heads/master","description":"Rebase"},"2996dd36ef0021d9cb3b9b3739fd2d2a14236460":{"kind":"REWORK","_number":5,"created":"2024-11-05 08:52:25.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/5"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-11-05 08:51:47.000000000","tz":240},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2996dd36ef0021d9cb3b9b3739fd2d2a14236460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2996dd36ef0021d9cb3b9b3739fd2d2a14236460"}]},"branch":"refs/heads/master"},"274de7d677d9f50dbb681b8a4926538e2575a571":{"kind":"REWORK","_number":6,"created":"2024-11-06 08:45:33.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/6"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-11-06 08:45:21.000000000","tz":240},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274de7d677d9f50dbb681b8a4926538e2575a571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/274de7d677d9f50dbb681b8a4926538e2575a571"}]},"branch":"refs/heads/master"},"483ce46f1128ae0341c76b07ea3a064375526497":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-11-07 04:31:32.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/7"}}},"commit":{"parents":[{"commit":"7c08b63a37e7ef990909d656683d098971228731","subject":"Do not fail on revert PlugVIPAmphora due undefined db_lb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c08b63a37e7ef990909d656683d098971228731"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-11-07 04:31:32.000000000","tz":0},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/483ce46f1128ae0341c76b07ea3a064375526497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/483ce46f1128ae0341c76b07ea3a064375526497"}]},"branch":"refs/heads/master","description":"Rebase"},"f76fa1ae330e933e86c666b857acfee06f675adb":{"kind":"REWORK","_number":8,"created":"2024-11-08 11:11:32.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/54/933254/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/933254/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/933254/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/933254/8"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-11-08 11:11:22.000000000","tz":240},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f76fa1ae330e933e86c666b857acfee06f675adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f76fa1ae330e933e86c666b857acfee06f675adb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
