)]}'
{"id":"openstack%2Foctavia~933422","triplet_id":"openstack%2Foctavia~stable%2F2024.1~Icac1764e818b9a600bc939cc5883c64f5adaed36","project":"openstack/octavia","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-04-15 00:03:52.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-04-15 00:03:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icac1764e818b9a600bc939cc5883c64f5adaed36","subject":"Fixed VRRP nopreempt option","status":"MERGED","created":"2024-10-25 13:46:11.000000000","updated":"2025-04-15 00:05:30.000000000","submitted":"2025-04-15 00:03:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933422","meta_rev_id":"29c2d1603681489aad01f12bbe944385c4b416c3","_number":933422,"virtual_id_number":933422,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-15 00:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-04-14 19:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-14 19:07:56.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-04-14 19:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 16:21:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 18:59:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-04-02 17:01:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"809a230d43364a2738b579bfd6a68fc0b1e8dd01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-25 13:46:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e716c366a9ddf52351bdf5c08140cde203295a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:21:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a95d443b05e040db93058207e37e3b09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/571ddf7270304a4a86ba9c735bb3f608 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad3b652dfcd4c2889c5d898918699c3 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23456ab2d4c44bc89f8fff52c353e041 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3491ad28c69b47d085b36c325083488c : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6843f0a9214494bbc8285595b5c4f71 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a2749ca67284e49b64c3fc46a378735 : SUCCESS in 4m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e66beeadcfec4c3f89d67fd259cbfbea : SUCCESS in 3m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd3f8dbf1c654e7990c26b837ac60cea : SUCCESS in 4m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9009fd3d17514e7caa809863446ac0ff : SUCCESS in 5m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6e7c70fedc814403a16c652f0a78104a : SUCCESS in 6m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ac702d2938942fa81707437a6f7037f : SUCCESS in 43m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c8c01a5fe40a457180584d2188dde4a4 : SUCCESS in 1h 38m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/866444e56cd2407cbf392ab792e3a1e5 : SUCCESS in 1h 01m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2d2e2857d210469e88c1320478e354d2 : SUCCESS in 1h 43m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf468e7ace264357893a0a63fc7655d5 : SUCCESS in 1h 00m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3ae9460b5a14a869749eea1830a137f : SUCCESS in 44m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e1577de4834b42f4952b7d0bcbce145b : SUCCESS in 35m 18s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/45d2f4341f064d8a887c6908b9e57813 : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/70a8421a55f14f24b98761a9aee86894 : SUCCESS in 48m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aec94ad6946d444eafa2e92eade4972e : SUCCESS in 30m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/489149494c3f4a78a8baec49469a50b6 : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db7f84f55a974d2897ee2b088cfb3f58 : SUCCESS in 2h 30m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/25e55d1a3a6f4bdbb37f74f5c489c5fe : FAILURE in 25m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9079b54095514256b48a876965fdf6e3 : SUCCESS in 2h 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0e3f310c92cdeef918369156f955f0b213ad477","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 18:59:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8894a72c043b5c9f7b7892e7b25a2031665138df","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-02 17:01:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c2ae058311102fa8950a78043ef20aeed0433ab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 17:02:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed8becee19a999347010a31b417734a7f4a8051b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 19:19:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f103cf0dd254e7b8608a4abff798fac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a608bb10f2b64654862ed57cb7e33cc0 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16d229634b7b4572b7b502308a2c84fd : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d5c6da992204d1989c2be21c80e759a : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af3c93bd89c4e32be243b2191e87c13 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bfa71ac23ae46e59134ebef2cc07a61 : SUCCESS in 4m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1431027549854765b281761aaca77e3c : SUCCESS in 4m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d19153692354c77ba564bb71490a9f2 : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3f0779d34e34f8abb3aa47d9ad70b53 : FAILURE in 11m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b82a83d76930458197b257ac4a3a4de9 : CANCELED in 12m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/308ca14f29784b69a12e29ceb77d697f : CANCELED in 12m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ccd9c994c6d427f9734752045b63b1c : CANCELED in 12m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1cd1d6954e324aa6af8b91dd5f5c6ab5 : CANCELED in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2adb8dfc41e811c0e3a088c5550787361dabc520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-14 19:07:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"db8e9ffa99d837fe32919e8a098df67d2b470d4c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-14 19:08:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"019fc73adcb14b8f9ca59ae8094a0ca10e259219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 22:01:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2793e87e104d450b93039cbd169320d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/716fa9192582465582a5c0d932ea1e13 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac50cd1b8ddc4025864b46217378de0a : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ae079c88d89458581d26f9e2e9c7b7c : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a5d2629b98e4d1fae78c65f27b8e6ac : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f89265dea042d084b0fc196f646d5e : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7af3b1dabf4411dad47a52b9381824d : SUCCESS in 4m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eeed0f3dd9c848fa87cdb3a5cadba969 : SUCCESS in 3m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/168f446e49c44500984e8b77c77a1e16 : SUCCESS in 10m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5cad1b383c944ebda96b41e259f22d60 : SUCCESS in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7747ad3c7d904ce09659bf9c22626410 : SUCCESS in 5m 06s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/afed2f8dab434d2f823d4f5e1aa7e401 : SUCCESS in 1h 04m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/1ce99c69ff8e4231b8c435ae40c5fd73 : SUCCESS in 1h 40m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/20f604c742cc458e8c8f029131c3b52b : SUCCESS in 52m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cf2f5b4b286544adaedb45d276875da0 : SUCCESS in 1h 44m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/fb0172bc7cbe4b39b83312978c04264c : SUCCESS in 55m 37s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3f249456b9ee48579b6f8df0d6ad3a3a : SUCCESS in 52m 36s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a0ae229a275b4a1099aed27bf9ccada6 : SUCCESS in 39m 12s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b0ad77fb2b8401bbe622967575264fc : FAILURE in 21m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e262ad5ef59b4ddf9c2793b2d9d42c1a : SUCCESS in 55m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d12db631cf7b4cc3bb69bd21212e9c39 : SUCCESS in 30m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/db18f9b8e0ce4a2fa76c0280ff626c30 : FAILURE in 2h 13m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3666514faeac4393883e31ab84c0f1c0 : RETRY_LIMIT in 3m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/51697a1aa7e142e1b45c372266b8f2d9 : FAILURE in 27m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/9254f46d91954968a3b89a55196620ca : TIMED_OUT in 2h 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f107add2e7084e35b695bc9ef1b54ac854ddfa9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 22:01:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83de714b25c5eec4d51ca3418677abc69de640a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 00:03:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bad8f5c0be674dadae633c17ab94ba39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f119a0711446409b9208a9d5f19d02d8 : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/787e30e41b5446fc90b401adcf870b5a : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60338aa99f1a4a12a8a99edea286014d : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15402bd0558440898093cfe3722ad65b : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21252082949140ff809760a371e54fe2 : SUCCESS in 3m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2790259238924ca7972a28d9134cff69 : SUCCESS in 4m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/be7ea6e0f8684724be7531b437972d0b : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/2bfc0db556094ca4bbf1a4cb85f32f6d : SUCCESS in 1h 07m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3119fa693b6a46b7b3d2d334bc22f105 : SUCCESS in 1h 46m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/e67636c163af4f50933805505249fd33 : SUCCESS in 1h 03m 37s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8902c8fde6a94a738319edad276669b6 : SUCCESS in 41m 05s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/514679776ac84b0186f6d56341168e35 : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"fffeb3daac7bd91754a9fe391bfa201b9a37e8de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 00:03:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"29c2d1603681489aad01f12bbe944385c4b416c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 00:05:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38aee10b3645481faac797ba1975cf70\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea876901e94b4cc696d82445bde0abf3 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/131c6d89f35a44ba9b7d045b4cc7e079 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c33604ae07b21c3120ee7fbf08822200a742397","revisions":{"c032046efc39be3c3ee1fa6b72eabbd005a9ce2d":{"kind":"REWORK","_number":1,"created":"2024-10-25 13:46:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/22/933422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/933422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/933422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/933422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/933422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/933422/1"}}},"commit":{"parents":[{"commit":"3af5401b81d7132af7939b64e7ef865629a47aad","subject":"Fix loadbalancer stuck in cascade delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3af5401b81d7132af7939b64e7ef865629a47aad"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 10:26:17.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-25 13:42:12.000000000","tz":120},"subject":"Fixed VRRP nopreempt option","message":"Fixed VRRP nopreempt option\n\nThe nopreempt option from keepalived didn\u0027t work as expected because the\nrole of the MASTER amphora was set in the configuration file [0].\nRemoving the default roles fixes this issue.\nNow the newly created amphora after a failover doesn\u0027t preempt the\nMASTER role from the other amphora.\n\n[0] https://github.com/acassen/keepalived/blob/master/doc/man/man5/\\\n        keepalived.conf.5.in#L2011-L2017\n\nStory 2010595\nTask 47457\n\nChange-Id: Icac1764e818b9a600bc939cc5883c64f5adaed36\n(cherry picked from commit 911faf26239840fd654c741a673c8bd657ba2c46)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c032046efc39be3c3ee1fa6b72eabbd005a9ce2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c032046efc39be3c3ee1fa6b72eabbd005a9ce2d"}]},"branch":"refs/heads/stable/2024.1"},"9c33604ae07b21c3120ee7fbf08822200a742397":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-14 19:07:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/22/933422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/22/933422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/933422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/933422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/22/933422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/22/933422/2"}}},"commit":{"parents":[{"commit":"e5e462c6ebfa6ea0fbb1c52017daf7fae45383cf","subject":"Update stable/2024.1 to use 2024.1 tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5e462c6ebfa6ea0fbb1c52017daf7fae45383cf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 10:26:17.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-14 19:07:56.000000000","tz":0},"subject":"Fixed VRRP nopreempt option","message":"Fixed VRRP nopreempt option\n\nThe nopreempt option from keepalived didn\u0027t work as expected because the\nrole of the MASTER amphora was set in the configuration file [0].\nRemoving the default roles fixes this issue.\nNow the newly created amphora after a failover doesn\u0027t preempt the\nMASTER role from the other amphora.\n\n[0] https://github.com/acassen/keepalived/blob/master/doc/man/man5/\\\n        keepalived.conf.5.in#L2011-L2017\n\nStory 2010595\nTask 47457\n\nChange-Id: Icac1764e818b9a600bc939cc5883c64f5adaed36\n(cherry picked from commit 911faf26239840fd654c741a673c8bd657ba2c46)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c33604ae07b21c3120ee7fbf08822200a742397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c33604ae07b21c3120ee7fbf08822200a742397"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
