)]}'
{"id":"openstack%2Foctavia~933423","triplet_id":"openstack%2Foctavia~stable%2F2023.2~Icac1764e818b9a600bc939cc5883c64f5adaed36","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-04-12 02:54:44.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-04-12 02:54:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icac1764e818b9a600bc939cc5883c64f5adaed36","subject":"Fixed VRRP nopreempt option","status":"MERGED","created":"2024-10-25 13:46:27.000000000","updated":"2025-04-12 02:57:04.000000000","submitted":"2025-04-12 02:54:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933423","meta_rev_id":"bd49c852c1219acd17266911cb933ad81db23eb9","_number":933423,"virtual_id_number":933423,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-12 02:54:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-04-11 22:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-11 22:15:14.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-04-11 22:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-25 16:20:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 18:59:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-04-02 17:01:20.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"6262fcb178299b17bc6b9ca2bfa646a7b404f3a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-25 13:46:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d556e0b9606ebe6e26a992f4229fe71a2fdf909f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 16:20:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9dfb086d77c43abb45bcb206c27f894\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/834529f54cf0430cad1705a4df2d0992 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7defa292a44d56969f451977e31bd1 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf8c2f27296142e3893a220a80de100a : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48f00d1b72bd4145a9de236d1d5ef1ab : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5168e2527d8a4dda8f2ce5c1c81349ee : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ede78619c1146458a8d975f4df323eb : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8418798bb95d4d4ba8260eabfb4290b0 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0214b2b7e9d4cfea0b2e3d917549a3a : SUCCESS in 4m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dfe862038c7747c4877a4ea959e5ed5a : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/595cc63eaa874c3e9733cd245c9f8f5a : SUCCESS in 5m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05d0741882244f849849493ae5a2603d : SUCCESS in 5m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d490332e99f846eb8eb8e023cdc7c442 : SUCCESS in 7m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/49f0398c8f86401bb319ab1127d61eb9 : SUCCESS in 1h 35m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/af981f1b210b438ea8fc83227e3cf59d : SUCCESS in 1h 56m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9b90058d6ccf428abfa1555d7db3c2ec : SUCCESS in 1h 08m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9d3acc00d5524250b74543988fd2d52e : SUCCESS in 1h 11m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b1136989acb547d29c935105899a0309 : SUCCESS in 54m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ce3cdd3f8ea14eb8924c8a449fa11696 : SUCCESS in 40m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf6c940d19f04632833801ab6b41e73c : SUCCESS in 58m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cb56eb4e6e7d442995723ec1238b7507 : FAILURE in 1h 03m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d51fee41920049f0a77a0cf7623894cf : SUCCESS in 47m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61f3264c7f3940e69fda36df9b852ab8 : SUCCESS in 2h 13m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e7752091b584bb48c05079c9b9bb301 : SUCCESS in 2h 15m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d53ef8d9c29c4227ba0b17bbc78f4f30 : FAILURE in 20m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3aef98e7e3443ce8ff889ab04710d16 : TIMED_OUT in 2h 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed16574668f43d208702b39b273bef68c19355c4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 18:59:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a2a564f88022c9ed4251f725c35bdd0bc1d68bc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-04-02 17:01:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7256c34a81b9474d134e8099e7ea48a552a372d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 17:02:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99c803e3ddb37ce9870fe0ae803ea32b0e8a82a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 19:19:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/078b1b4dbb9b4831b5ea4a30e9950c2b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11be6b3a265049238d2e268e83e91bf8 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a686e951821431e879971bdcb64d0d5 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3753111dae994988a132636ceb073faa : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9d99b97a8354ab2a2a788adffeff370 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9384e6026fb14ac3a5a6481ffadff514 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83733b836b8c49f1b956830897510172 : SUCCESS in 4m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/63c00fc13c7c435cb61bd77e1b92e541 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c3677b3756041b3b4afa5b6f318d141 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9edf11f7c80745638c9187edbbcc6c0b : CANCELED in 22m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d56427f499d64167a8d788afc0a733cb : CANCELED in 25m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f7111f7e74b64529b6320bc389f7b43f : CANCELED in 25m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a06785ed90974ba9a0656451ec4a019a : FAILURE in 26m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/073960cf657f439d801493be999d244b : CANCELED in 25m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f592460d69c9c7421baf0bb34435a8872a182b7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-11 22:15:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"e13fffba49ed68037a13cc78d489bb8a879df178","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-11 22:15:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"25cf4cfc1d17d4fd75a0b56e4a889df0d9605d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 00:56:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e0c2e3606647bcbd5215d215908772\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4565e384e368425986e259aa8e45c7b7 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e94c13e0d1490b8aae71d55e474d71 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2605ca12ff0c4e8bac774793e0ee7ae0 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f71fb0abc24d4863815817b4ce474d1b : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f1259a9fb8c4269bbee9934bd57b0c0 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42ef224ac0b7459197cb190f81572d4c : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5038ab0f87f4865b14a08ddacd4df32 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6123fe4cd86040dd988e10d1967bb05a : SUCCESS in 2m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d191e15d7e2b4f32b118e65307fb6af2 : SUCCESS in 3m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5324aca96ac54abe95650b5ecb24e086 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2baf5deda68145289d432c9985cf9f7c : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/de626404076243b9a6cdf06bca2d1c4a : SUCCESS in 3m 49s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4279abea960f4b059374792197612061 : SUCCESS in 1h 06m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/845ca6412ecb43a1a11a5c3cf5453ee6 : SUCCESS in 1h 51m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8161317cc9944ef9a4c06fc22c6878f3 : SUCCESS in 1h 03m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e2ac70927e8043a0883b4710d7e415a3 : SUCCESS in 1h 25m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/cde36d97445246a8907876e96c570fb4 : SUCCESS in 43m 04s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/cf98ec6dae6b432885ab288882283710 : SUCCESS in 37m 31s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2da456c54ac84dd0a4ccaa266d2b120c : SUCCESS in 59m 28s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e1043aec561144aebd80d6304ee17139 : FAILURE in 51m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/058aeb8f73a040d2be9b532161ed55b8 : SUCCESS in 37m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2023-2 https://zuul.opendev.org/t/openstack/build/0332f374a731431082d36ffd7f3edf44 : TIMED_OUT in 2h 34m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8f609285eb7a421a8de08bdbd51f12f9 : RETRY_LIMIT in 3m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2023-2 https://zuul.opendev.org/t/openstack/build/736b65f3052446a1a6c0d9c7d40f41e5 : FAILURE in 32m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b34603fa1ecc4cbe8fbe3c1c6d3648b2 : TIMED_OUT in 2h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2810d50666ce68a3a30fa645edcf4512efb07a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 00:57:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d2942a9c040e9519ea54c550219091a248b8481","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 02:54:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80f1bf25a97d41baa762887d66e56235\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adada2b99cdf4f83b024b446b113b5c2 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d681e719c5e34440a291558f88d81443 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e957f4666fb4d7aafae5a64c167eda9 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3329f7077b9e45bfa90e85b27da2bd6c : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/998423330de9470bbf8104ad0d58fc6d : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe2a9b34c58c4fe3be8b94a0e1100c73 : SUCCESS in 2m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f4aedb028ead4fdfafcdecbcf670ce89 : SUCCESS in 2m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/de3a2e56de8a4d24835a31f30c026ae3 : SUCCESS in 4m 28s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/738cb46e5f584bb99e92ef1dc64df45f : SUCCESS in 1h 22m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/79b7b0a518884e14a784290eeb2b56fc : SUCCESS in 1h 22m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6fbc144c48b24e8bb331ab071f6be112 : SUCCESS in 56m 28s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/767ea822e7274979991e9fee44cadc13 : SUCCESS in 37m 18s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e2943704247546ab860d0a08034a1a8a : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7d557349ccc0d617c22734b79449ea3160d53c00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 02:54:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bd49c852c1219acd17266911cb933ad81db23eb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 02:57:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5632fd3d7efc41fc9489ea61c05bbc8d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25835143ad844fd797393c65f65016b6 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/654f078bb87b41e5a0750a05c8444580 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a23a8ecaed3bd9d4bcd11b1d22526270a1c1423","revisions":{"4c1a405c8f02c04044b1e9d38e1fa51a852882aa":{"kind":"REWORK","_number":1,"created":"2024-10-25 13:46:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/933423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/933423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/933423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/933423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/933423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/933423/1"}}},"commit":{"parents":[{"commit":"09ea1e991a9acddece7cca0c9d1ef6c6a69be37b","subject":"Fix loadbalancer stuck in cascade delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09ea1e991a9acddece7cca0c9d1ef6c6a69be37b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 10:26:17.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-25 13:46:23.000000000","tz":120},"subject":"Fixed VRRP nopreempt option","message":"Fixed VRRP nopreempt option\n\nThe nopreempt option from keepalived didn\u0027t work as expected because the\nrole of the MASTER amphora was set in the configuration file [0].\nRemoving the default roles fixes this issue.\nNow the newly created amphora after a failover doesn\u0027t preempt the\nMASTER role from the other amphora.\n\n[0] https://github.com/acassen/keepalived/blob/master/doc/man/man5/\\\n        keepalived.conf.5.in#L2011-L2017\n\nStory 2010595\nTask 47457\n\nChange-Id: Icac1764e818b9a600bc939cc5883c64f5adaed36\n(cherry picked from commit 911faf26239840fd654c741a673c8bd657ba2c46)\n(cherry picked from commit c032046efc39be3c3ee1fa6b72eabbd005a9ce2d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c1a405c8f02c04044b1e9d38e1fa51a852882aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c1a405c8f02c04044b1e9d38e1fa51a852882aa"}]},"branch":"refs/heads/stable/2023.2"},"5a23a8ecaed3bd9d4bcd11b1d22526270a1c1423":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-11 22:15:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/933423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/933423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/933423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/933423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/933423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/933423/2"}}},"commit":{"parents":[{"commit":"ce0066f03fb605fca20fea06f1d0967e5957f10e","subject":"Handle undefined protocol field in security group rules correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce0066f03fb605fca20fea06f1d0967e5957f10e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 10:26:17.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-11 22:15:14.000000000","tz":0},"subject":"Fixed VRRP nopreempt option","message":"Fixed VRRP nopreempt option\n\nThe nopreempt option from keepalived didn\u0027t work as expected because the\nrole of the MASTER amphora was set in the configuration file [0].\nRemoving the default roles fixes this issue.\nNow the newly created amphora after a failover doesn\u0027t preempt the\nMASTER role from the other amphora.\n\n[0] https://github.com/acassen/keepalived/blob/master/doc/man/man5/\\\n        keepalived.conf.5.in#L2011-L2017\n\nStory 2010595\nTask 47457\n\nChange-Id: Icac1764e818b9a600bc939cc5883c64f5adaed36\n(cherry picked from commit 911faf26239840fd654c741a673c8bd657ba2c46)\n(cherry picked from commit c032046efc39be3c3ee1fa6b72eabbd005a9ce2d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a23a8ecaed3bd9d4bcd11b1d22526270a1c1423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a23a8ecaed3bd9d4bcd11b1d22526270a1c1423"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
