)]}'
{"id":"openstack%2Foctavia~933871","triplet_id":"openstack%2Foctavia~master~I0ff05a27265b32d7e37b162a3e82108253e25530","project":"openstack/octavia","branch":"master","topic":"fix_udp_garp","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-11-20 20:47:37.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-11-11 13:53:29.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I0ff05a27265b32d7e37b162a3e82108253e25530","subject":"Fix missing GARP with UDP listeners on SINGLE LB","status":"MERGED","created":"2024-10-31 14:24:32.000000000","updated":"2024-11-20 20:48:45.000000000","submitted":"2024-11-20 20:47:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933871-fix_udp_garp","meta_rev_id":"067cb328a8fcaa1fede558145f812ef423c3dbdd","_number":933871,"virtual_id_number":933871,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 20:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-11-20 18:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-11 13:53:29.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-11-20 18:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-31 17:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-05 10:03:42.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-11-20 18:21:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"53e21d3155518aedf9d48560c82b8ff4ed2ec584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-31 14:24:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48d0a2aae9c6147d22cf9b90c52162ed1e8888fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 17:50:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a44c0ad474df49fca977eabeb1a81563\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a9c7af7e2704e249fe0c9f4118e35f3 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a32e270f6184a0bb2a17acbc7765825 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c1cae032986441abfd1c411295f2c1f : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5104d28c276b4d0ebfd91cdfbf54222e : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b416bd93a94af8a56012057cb8791b : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc32759e2bdc480eb9d89dce0607ef41 : SUCCESS in 6m 13s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e955b34e374d453cada3f736915cb07e : SUCCESS in 4m 44s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1b4a5c1906314c789df67b1cc05548c2 : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8092e862f3e943beb616f2b2dcc9f722 : SUCCESS in 4m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b74687cf1cdc441bb13328b2e4a1f7c0 : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dcb0f02e3bb74565b44599409e6f6f75 : SUCCESS in 8m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cc22b0131575486db2792a4c8b2bdf3d : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26a9ef190c184a669f7133e48927133e : SUCCESS in 1h 40m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8ad802e3533b4bdaa6756f43af6a4179 : SUCCESS in 1h 43m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0f588a557f9248e2b681f112f072d469 : SUCCESS in 55m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/efe5fa198a8341d28915d72e0e106467 : SUCCESS in 1h 50m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/87ac1854a6dc492987994901da3b0db7 : SUCCESS in 57m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47c2e2e4525042deb97fa41ab2be755a : SUCCESS in 45m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0439266848e647ad8a19d7b2940554e5 : SUCCESS in 49m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b576066b72534dc7a08aef1463f2acb7 : SUCCESS in 1h 03m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4458d0d671a64afc986b8bdb2e29ee3e : SUCCESS in 31m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f6b56c4b37c54bfe868e1f39e8bfab62 : SUCCESS in 2h 15m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4abbe745f4444caf9301a2be71290a67 : TIMED_OUT in 3h 03m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a221cdf248c24ab6b143eb6596dfa188 : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7308894744d041cd9d939ac4333f08f9 : SUCCESS in 2h 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"215cc468d20552348687af8671145871639f6443","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-05 10:03:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6d2343d1496003dc0a163473f2e53a139afa42ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-05 21:53:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fca73c78fecd6f3a4564a2e81c4d175bf75fee2b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-05 21:58:29.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"189a8ebaa17b6eaff8702d735f394c11be1bbebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 01:04:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebaf630f1d874d21bbbfbfae9829a970\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3efa7e909e5f473b96c4c42d3ec83e72 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9261694fc4fb45c49294b3d5dbec0cce : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4238a4a216534286aa1ad3acf6a72dcd : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fee013e33bd4e7c9ac70b8246da8fd2 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a1f38f39aa440c8d9a4a8dfb4c88b4 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/591d1529447b4fc9b835bc55fc812001 : SUCCESS in 3m 46s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7ef81599fb07463e93bd89390d514a6b : SUCCESS in 4m 35s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/625bca2ddcc54f589414a930181950fc : SUCCESS in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/df17cbbc60814ef4a5e6a429f5a719d4 : SUCCESS in 3m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f79a8f304e2b4102990afec2f56278a8 : SUCCESS in 5m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53b58e8fb1604c1dbf1ddd7c2e7f7441 : SUCCESS in 7m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9fae52ca501145fd9b97c461c1d61380 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/013cb95a975049588919b3b1f9e69521 : SUCCESS in 1h 32m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d35732e4f96547c281a566f6e67e904e : SUCCESS in 2h 11m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8f7639032b594e24bc4300ac6dd5dfb4 : SUCCESS in 1h 36m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8096e6b882784a03887a05c2d6818265 : SUCCESS in 1h 41m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/124f276aa85e42ed872fc4b16abd6d53 : SUCCESS in 1h 29m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f03dae491acf413cbb30c0595764ee11 : SUCCESS in 35m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7087d1d373d64acdbf0b6bb3d5373ae8 : SUCCESS in 48m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0721a62c36ec4cde97c19f27598e50fa : SUCCESS in 1h 00m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8522ec0026074316a7c58ba1d04f7454 : SUCCESS in 37m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6a13edfc634c4cac985b7db408f6d7b4 : SUCCESS in 2h 19m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec554a7d48954ce68ba27d089e1ab183 : SUCCESS in 3h 00m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0bcba3233b1b46c8bb2b3d9e57f8d7d5 : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/79b1bd6eec774bed851f570e7f480674 : TIMED_OUT in 2h 35m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"526c2bed9a320c41252c7a74c98313fe3dffbc8d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-06 09:09:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a3842ca9e2b4231f99610f563faf2cf971978d5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-08 13:53:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83a57ad3eacb8daf171a75199024fddb585a126b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-11 13:53:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ae13d9acd496506c51c20a13ad3abb4cc977e11","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 18:21:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e017d320264af34fc304da3a4b4df48fb5497e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:22:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07438c85450d39e6abdb328ff79ebe14c7d2f5ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:47:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e131d93002644518e0becc3b337e10d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7c7379acc846fc9628a230c0e4cd4a : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a27b4e74e8c34475ac445dda39157c07 : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b6f270297974c51813de388c683e9d9 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f533b2b826804844b449d1b3c583c3b5 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad1811559f304c878d75a45d2aea5fc4 : SUCCESS in 4m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5e7ba6e049b54e20bc61d5d934a39bd8 : SUCCESS in 4m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7b44c700936741159fba9b72b05ff778 : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/81b67c5a38c1423b90394173add08b19 : SUCCESS in 1h 04m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e3eccc550ab44c239c636ee4a3973f99 : SUCCESS in 2h 23m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9ef9dc69fa6b491ab3895eba6649aafb : SUCCESS in 1h 22m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd41614b874849d3a3720b8f5ca86960 : SUCCESS in 43m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7b0d1b114cba4b6f8ccb48563228d457 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":2},{"id":"92bea175bec32f49499a1627b6f1a2149b433871","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:47:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"067cb328a8fcaa1fede558145f812ef423c3dbdd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:48:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a958d965129c4e45ae2ae783f04e17f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c500c83ffeb423eb9c95e8f46f1e120 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6912ac37fb8340738065646eacf69be6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e096d19d9d49581de52ebd10aaeb0bf4cb6a236","revisions":{"57e91282406ae38d629e256bbffc0a22f455de91":{"kind":"REWORK","_number":1,"created":"2024-10-31 14:24:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/933871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/933871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/933871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/933871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/933871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/933871/1"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-30 08:29:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-31 14:24:00.000000000","tz":60},"subject":"Fix missing GARP with UDP listeners on SINGLE LB","message":"Fix missing GARP with UDP listeners on SINGLE LB\n\nGARP was not sent when using a SINGLE topology LB with a UDP listener.\nEnsure that the GARP is sent when keepalived starts/reloads.\n\nNote: as the keepalivedlvs configuration is populated only when a pool\nis added, the GARP cannot be sent earlier (when the listener is added),\nthis behavior differs from the behavior of ACTIVE_STANDBY or\nhaproxy-based load balancers.\n\nCloses-Bug: #2086195\n\nChange-Id: I0ff05a27265b32d7e37b162a3e82108253e25530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e91282406ae38d629e256bbffc0a22f455de91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e91282406ae38d629e256bbffc0a22f455de91"}]},"branch":"refs/heads/master"},"7e096d19d9d49581de52ebd10aaeb0bf4cb6a236":{"kind":"REWORK","_number":2,"created":"2024-11-05 21:53:07.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/933871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/933871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/933871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/933871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/933871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/933871/2"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-30 08:29:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-11-05 20:57:36.000000000","tz":60},"subject":"Fix missing GARP with UDP listeners on SINGLE LB","message":"Fix missing GARP with UDP listeners on SINGLE LB\n\nGARP was not sent when using a SINGLE topology LB with a UDP listener.\nEnsure that the GARP is sent when keepalived starts/reloads.\n\nNote: as the keepalivedlvs configuration is populated only when a pool\nis added, the GARP cannot be sent earlier (when the listener is added),\nthis behavior differs from the behavior of ACTIVE_STANDBY or\nhaproxy-based load balancers.\n\nCloses-Bug: #2086195\n\nChange-Id: I0ff05a27265b32d7e37b162a3e82108253e25530\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e096d19d9d49581de52ebd10aaeb0bf4cb6a236"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e096d19d9d49581de52ebd10aaeb0bf4cb6a236"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
