)]}'
{"id":"openstack%2Foctavia~934079","triplet_id":"openstack%2Foctavia~master~Ib4d6659ea8861e17ca3fe5662bbdfbd6738f0a90","project":"openstack/octavia","branch":"master","topic":"shellcheck","attention_set":{"32118":{"account":{"_account_id":32118,"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","username":"thobiastrevisan"},"last_update":"2024-11-05 11:36:15.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ib4d6659ea8861e17ca3fe5662bbdfbd6738f0a90","subject":"resolve ShellCheck warnings","status":"NEW","created":"2024-11-04 19:16:54.000000000","updated":"2024-11-05 11:36:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bd17ab69a55af8262e86e69ec7664d3fa87b096","_number":934079,"virtual_id_number":934079,"owner":{"_account_id":32118,"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","username":"thobiastrevisan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-04 22:23:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-05 11:36:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-04 22:23:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-05 11:36:15.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"6fbf97367ec30b11524df1b9ffd6aa148183b17c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32118,"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","username":"thobiastrevisan"},"date":"2024-11-04 19:16:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"392f276ba913781786acacc7afd851455dcad4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 22:23:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1576a6f841b41a1a75395839dc73b9b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/def5326a322547f5a844a7841778b854 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87965004db74693817b5e50ea0dfe7b : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23e5e9d3a8174323bb6f547331300b21 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3495a62bf3da4c3e950fd4951bf134f4 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/044e1f780c9e49eda9f002ffe7c535af : SUCCESS in 11m 05s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bb7ee67f13434e9ba555f37cbe6ffae6 : SUCCESS in 5m 00s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dd4143aeebb540beb23cdf0b00a0f2b4 : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f37c8b442d0c4dc59356a13e16d0231f : SUCCESS in 4m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0a646a030708422ca5859b2fc93beadf : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/891e4180b08848878e6030be5997552b : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b426363da8cc4094a1f06d9c0eb8319a : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/318a5ff4ec6a4971b58aff8723891033 : SUCCESS in 1h 26m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b5fb5a7b4a4047afbef7f507f4a8a834 : SUCCESS in 2h 02m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/182254238bd1400a997c05944ca7d9c5 : SUCCESS in 1h 09m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/660e2929f9774a8bb37ec33c37dc14f5 : SUCCESS in 2h 11m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/433cf84bda3d4ee0914e7f187eabd389 : SUCCESS in 1h 27m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8820141abca442f3b57cbedaa4b1e6e9 : SUCCESS in 41m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0fac5ae89826466a9982075fa64680db : SUCCESS in 50m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c2ee5918f8d940249a7e3c336a8d4b5e : SUCCESS in 1h 10m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/510cc764bdd64960ad06750862be026f : SUCCESS in 48m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e77195cda3948c98ec905d2c772e707 : SUCCESS in 2h 09m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/08b96b306e1041759dd2bf1304126449 : SUCCESS in 2h 59m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b3ca16c22c04dad98515877888cdd91 : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/678de999c7e945b9a6917e742b9082f1 : TIMED_OUT in 2h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2bd17ab69a55af8262e86e69ec7664d3fa87b096","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-05 11:36:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9caca3267329a010fe33acefcaa1f58074c3fd1","revisions":{"c9caca3267329a010fe33acefcaa1f58074c3fd1":{"kind":"REWORK","_number":1,"created":"2024-11-04 19:16:54.000000000","uploader":{"_account_id":32118,"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","username":"thobiastrevisan"},"ref":"refs/changes/79/934079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/934079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/934079/1"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","date":"2024-11-04 18:15:43.000000000","tz":-180},"committer":{"name":"Thobias Salazar Trevisan","email":"thobiast@gmail.com","date":"2024-11-04 18:15:43.000000000","tz":-180},"subject":"resolve ShellCheck warnings","message":"resolve ShellCheck warnings\n\n- Fixed warnings SC2166, SC2086 and SC2181.\n\nChange-Id: Ib4d6659ea8861e17ca3fe5662bbdfbd6738f0a90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9caca3267329a010fe33acefcaa1f58074c3fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c9caca3267329a010fe33acefcaa1f58074c3fd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
