)]}'
{"id":"openstack%2Foctavia~934097","triplet_id":"openstack%2Foctavia~master~I090032566f329e7b73b0169ef133638a5f287a02","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-02-19 05:39:49.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-02-19 00:39:31.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I090032566f329e7b73b0169ef133638a5f287a02","subject":"Revert tempest disable due to RBAC changes","status":"MERGED","created":"2024-11-05 03:52:08.000000000","updated":"2025-02-19 05:41:50.000000000","submitted":"2025-02-19 05:39:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934097","meta_rev_id":"7e806cd9e34c6f40d2b2ffb7a07edc07746e42ce","_number":934097,"virtual_id_number":934097,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-19 05:39:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-18 16:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-18 21:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 03:55:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-18 16:21:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"7f2a7b1b5c3d3a26e7e1de7d3c42d8e685ee4aac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-05 03:52:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49eb0fd4d4d724f1a19cc792ed09ba8f0b9158fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 03:55:46.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926867 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"2991721d90c12d1b52b8822dea41fe713ad697c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-16 15:05:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7c8fb51dd2dc3f7be82f11954d3a2a31aaad4dd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 18:12:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b662d7afb8493a9b7b6738b6f24c9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dee1e2c2207c4a599039edb1422a798b : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2cdcee9b294702a626f99dd410267e : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6adfc89b77394a7c922635c7e100ad1a : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1327f87a9dbf458c8863eebd31b5e97f : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b85bbc9744c4c8f9048ed3ebff92071 : SUCCESS in 11m 32s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e9315d59c65141068e2baa2cb8c54d87 : SUCCESS in 4m 40s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fe7381afc7ff4e6aaa1c41c37995895a : SUCCESS in 9m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b5b48656b1d42d6acbb0e3799686117 : SUCCESS in 3m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40ccf2461b1745deb753e022d05ac35d : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/efff7d8db3f341339e0d0b90e7c88c2e : SUCCESS in 7m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bb04f4d94d9c4e5a970103f4f88106d4 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c5641f6b1214401bf9d26255ce7e177 : SUCCESS in 1h 23m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/346a188ee3cb4226885e2da319be1c2c : SUCCESS in 2h 33m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7f204282af7247359dc9333b3717a3ab : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ea5657737c449c9a07cee834f4c6a26 : SUCCESS in 2h 16m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5055ecfd241649199497c4ad90ac48bd : SUCCESS in 1h 15m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d016550f5c7c4a2f839b935ae3c12706 : SUCCESS in 1h 36m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0dce566051f94ab6b1f1203c9bee9ffd : SUCCESS in 58m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa2a10b76bf24fad89d097131713a823 : SUCCESS in 35m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fef0f5627ef46ec8d7e375b63fc540f : SUCCESS in 48m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b03493a9c32e418995896fac3d61f82d : SUCCESS in 38m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46121b5c19b44f618e5a36de3d15fe04 : SUCCESS in 41m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b4d08471b08b41c88500dbf4b3673af3 : TIMED_OUT in 2h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/afef9e87a78e4744ad16be8611bbde39 : TIMED_OUT in 3h 03m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f731430610eb47e382bed12511ae358e : SUCCESS in 2h 20m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d7350eaf4fb47c3a904569240c6e0f2 : TIMED_OUT in 2h 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65169aff2d400c7453a15c227af254d53ce7ae6d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-18 16:21:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c498f904cbe7b841d5a440806567a3f068376ab","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-18 21:47:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3975006667e4b5f6b9a884e91a889d0d119dead9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 21:48:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe7b743cb6d45f7ce98ef5d07fffddec6842a6a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 23:51:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f073c74c8dce4d9d99b0e8888c2134c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a737e4f791b494fa209b30d757e5967 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8bed76060044e999b5804d1ce4abc61 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ea70aa827ec4d1382f328d170743185 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1c8991db7b498abd9668fd70762533 : SUCCESS in 11m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3a73170587674be6ad37a14318f4cb59 : SUCCESS in 5m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/beab0af57c9846d08070472504ae6225 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96cb350cc3ba4015a26a2a8fd9e29e23 : SUCCESS in 1h 26m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ba9bce3475ec428e83ed3685c5ae79c3 : SUCCESS in 1h 59m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fa10a0451a744c1f9fe862688f0c5fbe : SUCCESS in 1h 02m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d946489b37cd4005a3f2ed6cd5d4416f : FAILURE in 2h 00m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06f1ad19f8a74821a71f42c33da3b8d9 : SUCCESS in 1h 22m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1784b5956ea742ca8800cff4dbb893b0 : SUCCESS in 49m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e59aa3c1c08847a9ad302f2de0ad5d8f : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0be1e3cf04c3e3708d655c19e05c8d4330684cad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-19 00:39:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"442c726fe8d329209cc269508b0c4f17b7fc8bb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 03:14:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/099ee7acba1147f18898f1295faf4c31\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/203c58081d6b44129a1c898b94a6150b : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3147e9b3e9cf4ea1be68a70473875278 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6e5d5dec8384791b28f76fd7fe19b3a : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e94d627d0164404ae1b418e150ac771 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff2c32daab54b5da745277eafb5bcab : SUCCESS in 10m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/718efe0e227e4ca99a8b4af85998f998 : SUCCESS in 3m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c6a339642b044480a95a4d9b80ceefc9 : SUCCESS in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d6f54035c1e84227ae3fa7490219db57 : SUCCESS in 3m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8e5ce243977a49dab1c5b1fc05d49f9d : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/915c696aef7544779f0827007c68337a : SUCCESS in 6m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4f76853e387345cd8b35dabb289fce8b : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/972b03512bcc4ac4a0dc4bada84098d9 : SUCCESS in 1h 31m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f9b103eaa3484962a945833748282a15 : SUCCESS in 1h 45m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f6253c6f30bf4e8da7de6e348767d16f : SUCCESS in 57m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5f300cb6efb4f8284c6c6943a619b10 : SUCCESS in 1h 46m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3807e915cc27497eac084629d28c3a34 : SUCCESS in 59m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b5ee8381c22e4faca748dcf24428c301 : SUCCESS in 2h 13m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/84dfd554306c44e5b58d9cfb49d49ef6 : SUCCESS in 1h 25m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/961b87377d6643d78e60dd3faaa67d2f : SUCCESS in 30m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f10e05534e6416abf1feae4fadc3e88 : SUCCESS in 49m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cbf090bd138541d79600997d7c50812e : SUCCESS in 41m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7fa852dcedf4e57a48adb101d493662 : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/86c51c2b542b4b85bf1403253de17a99 : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/42a4a79067f04f54bb36172400d350f8 : FAILURE in 1h 57m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/58c2f3e6c247448abb71bbe6e0e1e476 : SUCCESS in 2h 26m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dadc05683974f09a3f17f0086487edc : TIMED_OUT in 2h 33m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9faee26779585a015c3db4931385501d97959ded","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 03:15:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"04e41ae54685eed4f162ef30cb0a4242ce2a83e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 05:39:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0958a24c0e948018a7ab0cb454ea69e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c008b966c264dc3a48efc3d43e8c1b6 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6750fbd9ce264415b6588bbb2c05fec3 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cbc4342a49ee484c9a9fbdcddf202ff1 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e7f9e8e7b2f49d099f3f7cba829fc1e : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d3c08e4b3da849489a8355e422059ed3 : SUCCESS in 3m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/10ba8e56a3f145eebf85e4697a6db788 : SUCCESS in 6m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da19f56ead7e46fd94c855e8873a1206 : SUCCESS in 1h 20m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/09e732679e4c4610906db899d60ece5e : SUCCESS in 1h 52m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d0bc4333d02a457e843911e95885f6a0 : SUCCESS in 1h 13m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e34f3bf9f3a40ffac93968007568e6c : SUCCESS in 2h 22m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e04c95a83a9448e689f216d7bf54ad93 : SUCCESS in 58m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcb52e61ed0544f1ac9543306ef2cfc7 : SUCCESS in 37m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee6340ffdcd04a66988b69a8acc61d08 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ab47bedaaf0ff509f27adcb4bb4bbd63871387f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 05:39:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7e806cd9e34c6f40d2b2ffb7a07edc07746e42ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 05:41:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd68f1089651485ca800a5a23f40b022\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcea1935d900469d8ce9567664bdb89c : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a764d122aebc75cb515d20dc92cb022a7c819a7","revisions":{"89d647d86404e4e47f5086302869d0ad4b0d3932":{"kind":"REWORK","_number":1,"created":"2024-11-05 03:52:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/934097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/934097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/934097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/934097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/934097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/934097/1"}}},"commit":{"parents":[{"commit":"1d8bedc41879baac284c6be706497dbcf77e9dca","subject":"Make keystone default roles the default RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d8bedc41879baac284c6be706497dbcf77e9dca"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-05 03:48:16.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-05 03:48:16.000000000","tz":0},"subject":"Revert tempest disable due to RBAC changes","message":"Revert tempest disable due to RBAC changes\n\nThis patch re-enables the tempest tests after we have swapped to the\nlatest RBAC policies.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/926867\nChange-Id: I090032566f329e7b73b0169ef133638a5f287a02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89d647d86404e4e47f5086302869d0ad4b0d3932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/89d647d86404e4e47f5086302869d0ad4b0d3932"}]},"branch":"refs/heads/master"},"9a764d122aebc75cb515d20dc92cb022a7c819a7":{"kind":"REWORK","_number":2,"created":"2025-01-16 15:05:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/97/934097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/934097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/934097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/934097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/934097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/934097/2"}}},"commit":{"parents":[{"commit":"22929f654e6dd4ffb74f44e15514701449c4a664","subject":"Make keystone default roles the default RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/22929f654e6dd4ffb74f44e15514701449c4a664"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-05 03:48:16.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-01-16 15:04:12.000000000","tz":60},"subject":"Revert tempest disable due to RBAC changes","message":"Revert tempest disable due to RBAC changes\n\nThis patch re-enables the tempest tests after we have swapped to the\nlatest RBAC policies.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/926867\nChange-Id: I090032566f329e7b73b0169ef133638a5f287a02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a764d122aebc75cb515d20dc92cb022a7c819a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a764d122aebc75cb515d20dc92cb022a7c819a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
