)]}'
{"id":"openstack%2Foctavia~934111","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I97acfbd9b26fefbc38b61e36f72f483072390b23","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-11-08 11:01:57.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-11-08 11:01:57.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-11-08 09:17:09.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-11-05 11:39:20.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I97acfbd9b26fefbc38b61e36f72f483072390b23","subject":"Do not fail on revert PlugVIPAmphora due undefined db_lb","status":"MERGED","created":"2024-11-05 09:55:10.000000000","updated":"2024-11-08 11:03:52.000000000","submitted":"2024-11-08 11:01:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934111","cherry_pick_of_change":933101,"cherry_pick_of_patch_set":4,"meta_rev_id":"88054632dbcc3b2b80fc4c958f48a8fd1a3eb685","_number":934111,"virtual_id_number":934111,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-08 11:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-11-08 09:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-05 11:39:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2024-11-08 09:17:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 09:59:33.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-11-05 09:59:33.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-11-05 09:59:33.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-11-05 12:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1cbca6ab47d079e6f55fcfb7b35fb06c10f7ea69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-05 09:55:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"039efb4b48f2e11adad95959c360207dbf2b718a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-05 11:39:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10569aff1c52ee4d2e5f71a6108dd95b716ed3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 12:46:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b438e02b11c4f5b85874188a6fb4123\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/750cb4e067e448adbfde1891bc7d1328 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7836053ea564a479d3e1ad08518cc4d : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b86762c4598a411daf09b738baed80e4 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cd92f404b3040f1a404d74637c51312 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/685472f92ceb4c46970671263fba578d : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836847d86b8d46df9d1046ee3864d846 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c43555da59e9428ba368a894bedbbe70 : SUCCESS in 4m 36s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a64a4c14fd9e40a982847773318c8e6d : SUCCESS in 5m 16s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/23f05040a30245999e6b092d027cc831 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7c395cf1e89e4199b69e33ddf379fe06 : SUCCESS in 4m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f901ab6219cc420cb28960c0403e493c : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9abe65be1af24d0e9a645476621f42b3 : SUCCESS in 7m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fb59f9dacd67415ea381cd59a0f359a2 : SUCCESS in 6m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fadde72cefc42c98be1b75b7a921efc : SUCCESS in 52m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/82d667c5e2cf4e14895016d904a146f3 : SUCCESS in 1h 35m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/24a96b7e205047e8bbc5ce38d3649943 : SUCCESS in 50m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eb9ca74138854d5a8e1dc4978b778b91 : SUCCESS in 1h 53m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/73809117d296495ab3cb6866a4352580 : SUCCESS in 52m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8df268bdc0fb453e84a9468b3d238a27 : SUCCESS in 36m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/334f117015ee4974b3b2250324e2d946 : SUCCESS in 34m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0e33ce4624b742339508a15c1863349c : SUCCESS in 39m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83284dfff1494ac0bf7f1b12d4aee1ef : SUCCESS in 36m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54d0fb236e0a4caa8925c43cdc9ad87d : SUCCESS in 2h 26m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b151c659c0694cd383a8d99629414767 : SUCCESS in 2h 49m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ae8c90ac7b8546d8af326dac32859e98 : SUCCESS in 2h 19m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc50233d56224579a9ebc412473526c0 : SUCCESS in 2h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb528200f4c5ad9ce534e96ae5ac146353c195b3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-08 09:17:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"517449bba9b58ee40769575ae460664c78676e2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 09:17:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d70a39944539e7dbf326b4d67bf253d0add97817","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 11:01:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ef82a6e14454b4f8882923949a0a64a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcecb875811f47e59216f95b3ee1428e : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98ce7a73c75e43d7b33fb92e8a630df8 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03eaae780cb04de9a0e906c458ac1df4 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b4e5ddc6404d6dbba2f07be6937c1b : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42fbfe667a784a99a994505d3359f78d : SUCCESS in 5m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6622a421c37c498a8886698a240b8634 : SUCCESS in 4m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bacb47bf2a4746e3a581cee4c01abfc9 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c5272c11d2f44a5d9bfafa5a793097ed : SUCCESS in 1h 33m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/699cdf196e664f48997eb753265273ea : SUCCESS in 1h 43m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b830e3f9e0434b85a6b293c23e73464e : SUCCESS in 1h 03m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b5a58439cd04ae28d6c798199517764 : SUCCESS in 31m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3416fada968f48339e4b03ff244fa19d : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e687ea9c4bd6df69f26d4ccc1e0c53dd81a0c485","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 11:01:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88054632dbcc3b2b80fc4c958f48a8fd1a3eb685","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 11:03:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba3ddb3da618478e96cf3fde5924e8a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4ab37398d84c14abf94dac784a2640 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a8ac64538a74fab89e774650713f0c6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57ccd896ea31e9e5e78f356d44d8f97f058a14ac","revisions":{"57ccd896ea31e9e5e78f356d44d8f97f058a14ac":{"kind":"REWORK","_number":1,"created":"2024-11-05 09:55:10.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/11/934111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/934111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/934111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/934111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/934111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/934111/1"}}},"commit":{"parents":[{"commit":"df85248493f9f1a94de1f1c36a7b87fc6515a28a","subject":"Pin pylint on 2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df85248493f9f1a94de1f1c36a7b87fc6515a28a"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 11:24:50.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-11-05 09:55:10.000000000","tz":0},"subject":"Do not fail on revert PlugVIPAmphora due undefined db_lb","message":"Do not fail on revert PlugVIPAmphora due undefined db_lb\n\n- changed order: get db_lb before getting subnet info, to avoid useless\n  message without db_lb.vip\n- added default db_lb value for case, when DB calls are failed.\n  in this case error message will be generated without information about\n  VIP\n\nCloses-Bug: 2085427\n\nChange-Id: I97acfbd9b26fefbc38b61e36f72f483072390b23\n(cherry picked from commit 7c08b63a37e7ef990909d656683d098971228731)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57ccd896ea31e9e5e78f356d44d8f97f058a14ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57ccd896ea31e9e5e78f356d44d8f97f058a14ac"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
