)]}'
{"id":"openstack%2Foctavia~934579","triplet_id":"openstack%2Foctavia~master~I04255386e719e4354c7f94ddcaa61c0bab019a71","project":"openstack/octavia","branch":"master","topic":"poliyc-yaml-file","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-20 16:17:15.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-02-17 23:06:00.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-17 09:49:00.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-09-10 19:44:26.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I04255386e719e4354c7f94ddcaa61c0bab019a71","subject":"Remove default override for config options policy_file","status":"MERGED","created":"2024-11-11 05:44:12.000000000","updated":"2026-02-17 23:06:59.000000000","submitted":"2026-02-17 23:06:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934579-poliyc-yaml-file","meta_rev_id":"561c65d609befeb04b4c99daff27109d42c37df7","_number":934579,"virtual_id_number":934579,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 23:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-09-10 19:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2026-02-17 09:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-13 09:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2026-02-17 09:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 09:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-16 03:39:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-08-20 16:17:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-08-20 16:17:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-02-13 09:27:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"55b6ae29a0aee8016965e3580320333a7047cf23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:44:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"635b31ab90066fb09b311658498bafeaed87a847","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:53:53.000000000","message":"Topic policy-json-file removed","accounts_in_message":[],"_revision_number":1},{"id":"c0b540492aebb20b6964011b99133719bf272265","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:53:55.000000000","message":"Topic set to policy-yaml-file","accounts_in_message":[],"_revision_number":1},{"id":"5889794b755723f92425a795913164a9debdd531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 09:13:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c5d33cf2c0400b8f8f474bf15666be\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/305e8b3584e74604b8b4c925de0c0c58 : SUCCESS in 1m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/53844d9f9c9f4c2b9aa430f7acd82b72 : SUCCESS in 1m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2972825d11b44b6eaea435d5933ac10e : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4456485d194bd689b9663c5d850c38 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d45a123f16344bf9e4803c50a5fee91 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28b703b1957846498378f24ddad1772a : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31f9a0c9d0a472c94508287811ff281 : SUCCESS in 9m 31s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4412460e28da4a42b653a59340803963 : SUCCESS in 4m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8f1d04bd14634291b3d9e743d999c429 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cfcd5824243c43bb9c8db542b06510d3 : SUCCESS in 3m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27ec7bccb0174f038f2c6fd610b9fc94 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5b6e1a33aeb344ec86352f28846c83a2 : SUCCESS in 6m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7b631467fd6c4e639479436ce349c976 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b412288f4a9445e90868e76765569f1 : SUCCESS in 1h 29m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2520efda068348f1ad2da44068287feb : SUCCESS in 1h 42m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b604afb8133944328dd71fbdb0137816 : SUCCESS in 1h 01m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf5ae89a6a114a5fa37904ff2b48c1d1 : SUCCESS in 1h 46m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a365d4475c634de992eedd00cab12b8a : SUCCESS in 1h 03m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df5d0731bee04ed1b8e9cd97f67ce41e : SUCCESS in 32m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9948daf27deb476ebdb12feb156a12ff : SUCCESS in 37m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e84d43f5eb4542b1a83aff6a8e3c91b1 : SUCCESS in 1h 08m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21f1a71ddbba483f97f5a2824b12870e : SUCCESS in 38m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8e455151e6f4c26a4a05fb0e6f862ae : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e328fefd8f224ae392448e79fe693375 : TIMED_OUT in 3h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e2cc90b833b14179bc4e75f820b7e7ad : TIMED_OUT in 2h 34m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/35c15409891c4a72bdc6fe963f6a6bf5 : TIMED_OUT in 2h 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e91c369c579ae92a7978df4f98840363a2219825","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 03:39:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7ffda2979ca5f34195353891a76ca8fbc0b8b5f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-16 04:33:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"781b33b2b711a0485dd591732a415d912bae2b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-16 04:44:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c4b87e132cf762eff21106e6464dda23ce538e28","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-16 04:44:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f63b56f4d0da14ef8381da54aa9faad9479100f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-16 05:54:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3069e7f1bd232ed16b6b1615c83166ac5e4795de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 08:58:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e961ffc52db54d4dada0fdcd72bf09fa\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1c96dc70e1a14af0bf1619fb977a28a5 : SUCCESS in 1m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/496108aea02946f182bcf581b6c21f3a : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c53f298f9b84aa3a7705daf11fb065d : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1241daaa587d458e969e1f92ce878f8b : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e34a81065df4ec0bb4380bdfe9687cd : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec30deb5112f4d5f92f924b673c257ef : SUCCESS in 10m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4ddffae72ace464abc2ba0abfd27e8b1 : SUCCESS in 3m 35s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/04d493cecee2420d87b965f9457017ce : SUCCESS in 6m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a0133aefb51c4d03ab52647f71bf3f45 : SUCCESS in 3m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48f495bce8ad486a9a98136faf9a562a : SUCCESS in 6m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e0effa148318452b881dc0d70e71b147 : SUCCESS in 5m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8e319b26215c4b24a24c75a8ee317a32 : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0497c42461974ae2b91edbdcc4315953 : SUCCESS in 1h 29m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9fe76e12a47f4aeaae02b0595f9e9c69 : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3a9b07e909104ead9ed585666729c309 : SUCCESS in 1h 10m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7f3f98b65fc34a50915d2af8455384cb : SUCCESS in 1h 57m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c1717a6755554054b8de9055b1b953a7 : SUCCESS in 1h 04m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e7506a9cd94d42e38a9ec35526071275 : SUCCESS in 41m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fe60e547528949b7b7800d0620c16aff : SUCCESS in 46m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5c53f6c6cf3c47e5a062cfcd77b433cc : SUCCESS in 43m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfa909b2693f4b988d667e92868c7342 : SUCCESS in 31m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ed26802c9104513934cfec94b066d38 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/360ea6825cb8499abf465b1976f56c32 : TIMED_OUT in 3h 02m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3a340ed8cbbe44d2a34df8e6e101382a : SUCCESS in 2h 14m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/72c9d0d46cc148bdb90da79111f5458d : SUCCESS in 2h 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1805c70127e962de2d0bc2e5d7facd463ace4aed","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-09-10 19:44:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"55aafd89ac2a5082bafa60beffef32864c117105","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-13 09:27:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a59965fa66ec4f165ac093fbc391782904fb2b5b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-17 09:49:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"05cd8084bd6115ce0edd6661cb97468185029eda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 09:49:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"708036979337f0a3528272bb724f2a23db8ff160","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 10:26:41.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a09f43cf108f466197078b5c12d92893\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/215a30903a7c4f8ba263d2304709e25a : SUCCESS in 1m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b63c4830975413ab81c3bc4d6aa56e0 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/206f7b9004da4310bda3fdb1e3f6e9b0 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0cc63790d464dad9c064db73831d6a3 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd462f998b04186adc8da6b9e2add6e : SUCCESS in 5m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6148de2e44f542a99676b7e014fa0764 : SUCCESS in 3m 18s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/59c8f913e366427784225f7249d21317 : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/79157a103ab642e79ff926634ec70ef1 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d51e014e20b346f8ac18bfce177d4b1b : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ecf23ba476c8485b8ba9466a89879cef : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ee98c0de2974dc5a78417f6dd61f1e9 : FAILURE in 36m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd2fd641eea94b46b404888dc285868e : CANCELED\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d9bdeed88d194429906f59b1e3fdbdb1 : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"4ad2f299e0382c490361e8bec89bdd286633ce75","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-17 13:23:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ff3688d7ba08f918badc40866a6544546bb95f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 16:34:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbad99dc125f4694bbe660dc95d3e586\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd0c854c296a47e5a16b75ffd78c961c : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e241b617d124ff491ed55934aaae4ae : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189cf474b476438d9d07662991070101 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ca6609bc702428b87d94fa200977e83 : SUCCESS in 7m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9cc6b4e6d564daf9ab76d2216f37d6a : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410dd156fb934b3382f9e5a3b0d917d0 : SUCCESS in 11m 57s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/a37cf9ea6a1f4c55948b8e0e21a4b23f : SUCCESS in 5m 03s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/fcebc95be952478693af980eeef4a054 : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e427a63e1d5846fdadfd65a1e6cf2cfe : SUCCESS in 3m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f95c7105eb04006bfcfdd947187661f : SUCCESS in 7m 48s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2c5615bf61d04c4592714e0c4cb87184 : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b9b485ada4144a7b830f67584800fa39 : FAILURE in 1h 12m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3e0ae63cc23b4fe5a0436785bcd98421 : SUCCESS in 1h 44m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7863cf8423244b4aa43d9b8741402496 : SUCCESS in 1h 16m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/198e6317cea549a4b5efd459b19e1955 : SUCCESS in 1h 51m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9a85a89e108c43368c68e1516f3751d5 : SUCCESS in 1h 25m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e960a6dcd5244c483ef5c9b65d06dca : SUCCESS in 38m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/611cc3b55d4e441b82d5d294c981f216 : SUCCESS in 52m 32s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e336527ef6c042f8b819432c936980db : SUCCESS in 43m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf821eb0cdcb433eb9cabb6613de7cb6 : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9ab3934954fa42119a1961334ebe5bfd : SUCCESS in 36m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fffd3f5441de4904bba892a392dda726 : TIMED_OUT in 3h 03m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fd5a1efbd08f4507916b5837ea9bc897 : SUCCESS in 1h 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"264934ab355d33970ebfcc1a84cb2ae900aee9d9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-17 18:38:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1bd1edec4c4f2466061ac2a6662564f85fda2c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 21:22:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc4b3acde1145bfa2cef298f8b90336\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e60411d1803443d9d2b0be5c3f0aeb8 : SUCCESS in 1m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa3c327c157048c58768a8f2e8c38f6f : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0cb113f98b4027b58fb629f34d0381 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b9e5bd0ad6045299c699303c335be22 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a1987e9ac074abbbb1f75f1ea97f3af : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e76f4a4136b4c0b97869f4039b4c166 : SUCCESS in 12m 28s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/50a87e59db2941c09ca375795ad8014e : SUCCESS in 4m 47s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/c665734d3d8248cbb34ee21ab7c33848 : SUCCESS in 5m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef33e859650b40db9e7904fb755f0294 : SUCCESS in 3m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaeaa103d1d54e1198979e1e9cc1cd1d : SUCCESS in 7m 54s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a2c633de8613456fb19f1bcb15b188ed : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d9601fbfafd412ebe7cae7dee181df4 : SUCCESS in 37m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b472fdf4d849401392f3a31346d6d452 : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/02099f806e6747c49c43268194f5919c : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0cb7229f0bd04c1b83ccedc8703769b1 : SUCCESS in 1h 38m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8d8cdc3bebea4f7582aada2be851f1fa : SUCCESS in 1h 00m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/579148d205134b528fa479c6a7f8f7c8 : SUCCESS in 38m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b256d1a1b928457cb9c21ba8b6d677b1 : SUCCESS in 48m 42s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4abdaa3aace84c5088cd7e61e9df8664 : SUCCESS in 49m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f29352398ce45da9d6ee4c01bf8a098 : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1cf83b5be0e34fa2b5f33676ae18fbb5 : SUCCESS in 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/343a41376f834e42b600b6b89b954ee1 : SUCCESS in 2h 42m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/324cf4e935dc4594a3c98809f2b35346 : SUCCESS in 1h 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bb3e238301892b01a73f15a37660b6a8972f37ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 21:22:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8f9037dd43c2e0ec4d2b02b325533358b4b08c53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 23:06:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be802d63b6c74f8e841f015966267250\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6aabc1459aca44b9a94ea801411df4e3 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b704f79e28447cb20de3c2061b59e0 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1fbfdc5a31942eb9a8d46f3f88ea396 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52b0dfcd6e7649b987855e33849475ea : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbba97b4f4db4ada978208f32e6fb9af : SUCCESS in 15m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f98f6c18b7c4c78ae6ec25fed8a3eef : SUCCESS in 5m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0327b8c6ce1f493fa5dbecd285e75438 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/37a01aa2b75d4a17af8e35cfdb97cea3 : SUCCESS in 1h 41m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d903d42c6af4433991416314936bcb8b : SUCCESS in 1h 36m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4fc80b657450485181fdbd17f3451c55 : SUCCESS in 1h 06m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4b5b9b02c7544b49f2ebb93d45f877f : SUCCESS in 37m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ef10482e47d43779c6283ffb088560d : SUCCESS in 43m 26s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2e90b377d12740afb8168aed28fe5ca0 : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8774327a0fe3261f5b7e23bf32199103905a3424","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 23:06:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"561c65d609befeb04b4c99daff27109d42c37df7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 23:06:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2cf67de81bc476594a6c7ce73c9777f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a44143fc60d4462a8b26ebe33780234 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0fbf085e123e8bd8a376f2c4b8570a2db0a34ec1","revisions":{"84161558d033f7a6e2ae87a0eea3f920888eec3a":{"kind":"REWORK","_number":1,"created":"2024-11-11 05:44:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/79/934579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/934579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/934579/1"}}},"commit":{"parents":[{"commit":"0c64c51275d0a4f5a3b7409393d288618eac77be","subject":"Merge \"Document how the stale amps get processed internally\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c64c51275d0a4f5a3b7409393d288618eac77be"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:06.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:06.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: I04255386e719e4354c7f94ddcaa61c0bab019a71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84161558d033f7a6e2ae87a0eea3f920888eec3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84161558d033f7a6e2ae87a0eea3f920888eec3a"}]},"branch":"refs/heads/master"},"f7e52fd11dea5ea95e7b477ba424692eb7ec5f20":{"kind":"REWORK","_number":2,"created":"2024-11-16 04:44:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/79/934579/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/934579/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/934579/2"}}},"commit":{"parents":[{"commit":"0c64c51275d0a4f5a3b7409393d288618eac77be","subject":"Merge \"Document how the stale amps get processed internally\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c64c51275d0a4f5a3b7409393d288618eac77be"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:37:06.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 04:44:35.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: I04255386e719e4354c7f94ddcaa61c0bab019a71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7e52fd11dea5ea95e7b477ba424692eb7ec5f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7e52fd11dea5ea95e7b477ba424692eb7ec5f20"}]},"branch":"refs/heads/master"},"0fbf085e123e8bd8a376f2c4b8570a2db0a34ec1":{"kind":"REWORK","_number":3,"created":"2024-11-16 05:54:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/79/934579/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/934579/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934579/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/934579/3"}}},"commit":{"parents":[{"commit":"c6cf74f5139d0bf82e50290eaaba4381aa400085","subject":"Use overridden defaults cors options when generating a config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6cf74f5139d0bf82e50290eaaba4381aa400085"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 05:54:13.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-16 05:54:13.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: I04255386e719e4354c7f94ddcaa61c0bab019a71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fbf085e123e8bd8a376f2c4b8570a2db0a34ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fbf085e123e8bd8a376f2c4b8570a2db0a34ec1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
