)]}'
{"id":"openstack%2Foctavia~934679","triplet_id":"openstack%2Foctavia~stable%2F2024.1~Ia8b4612cd25a68b44d55b113f704756818eb8a5b","project":"openstack/octavia","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-11-14 10:00:33.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-11-14 08:01:14.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-11-12 10:11:27.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ia8b4612cd25a68b44d55b113f704756818eb8a5b","subject":"Remove amphora_health record on revert CreateAmphoraInDB","status":"MERGED","created":"2024-11-11 17:18:45.000000000","updated":"2024-11-14 10:02:27.000000000","submitted":"2024-11-14 10:00:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"934679","cherry_pick_of_change":934678,"cherry_pick_of_patch_set":1,"meta_rev_id":"a8ab7b5a6c8968883fd0670bb550853a6e33fdee","_number":934679,"virtual_id_number":934679,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 10:00:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-11-14 08:01:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-12 10:11:42.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2024-11-14 08:01:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 20:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-12 09:21:42.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-11-12 09:21:42.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-11-12 10:11:42.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-11-14 08:01:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"374a643eef344a2953a57ac33b8de788dfaa2563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-11 17:18:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"3def35655911edd29abcc89cd51f05759e8e73eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 20:00:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef896236f4443d1a815123aa153bcdf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd6647a4253349af8d2ce60936d7cfa1 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad674ff354c14f1cb4848d7f30f75311 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12680b377d404455853973584f4d892f : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dba1dfd699104e5bbb78046497c7feb0 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de0655ec9b144ada7954bfbb19cca13 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e2ea06bf5214c06b58d7d14a4d5026e : SUCCESS in 5m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eb26d131618e47689a9a591d5e4dd4ab : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b6657bc4f6a5479db973c69dda4ef2d6 : SUCCESS in 7m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/646344d742574fd4a4521eaf0a510d83 : SUCCESS in 6m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15aa2d6547354fb8bc0b9329f502d634 : SUCCESS in 7m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0268e012c80048b3b9ac8d2f3af055fb : SUCCESS in 1h 20m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e9652b25250746b19d657631b6b709f1 : SUCCESS in 1h 44m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d288724b168448b9a4e4e759b51a25aa : SUCCESS in 57m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ea9c934a6dce4065bd884f2bf9334647 : SUCCESS in 1h 50m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41c9f401ae7d431dbe8d65040002211d : SUCCESS in 1h 15m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83ab1fb3f8774657a892b8e6d9209130 : SUCCESS in 39m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7b8918e8d97406baf8ce8243661e61f : SUCCESS in 34m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b28361a67d0a49b4acaec364023a173d : SUCCESS in 41m 53s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dee65c84af874960b58281866e5c1eae : SUCCESS in 1h 03m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e14689a7e1b40be9dbf3a809d3b43fa : SUCCESS in 36m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c61e02abf1d144659347edfe690de820 : SUCCESS in 2h 09m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57d3e90e3eed428b8e4b06d82fc1e1ff : SUCCESS in 2h 40m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a09f67a4144643399cf4e14d9801b895 : FAILURE in 22m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/44096f03965049a59488c759289f7f57 : SUCCESS in 2h 06m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0a65c593cd7e1022b61795b6550016d0d9e5a28","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-12 09:21:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf727ceb14849009d692758c3d6533494051a013","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-11-12 09:39:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0d8b2891696e1f7e03b1857497a1aad132efe45","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-12 10:11:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ff45c0df81afac76ce525610d95d1bca061c093","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-12 10:11:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f34d7c1a58d07346a106a6efc3b046f7c73c8776","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-14 08:01:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4000a7ad7bd3b9d251bde06d87aba2392f7c276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 08:01:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13b8435fad45018d197993566008d23fa0faa651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 10:00:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6699d59f4ec94a54a9fa357f40c661f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93f909b619a2432889b1b0a23d4eb544 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d1e0683d237413a8750aeba886fb129 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27e58d663e744bf1810c2bb16edcf9b7 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776843be98a3458f9734a9de3b1323be : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11a6d4f9b46a4cb3a6bc31c3b9b8657d : SUCCESS in 5m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8725b6eaedd948bfbf6f868bf66b0d72 : SUCCESS in 4m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/183adf1c6a7549e6bae38a2a998f82de : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef58b2be0d7245ccaa6a920f8e92e443 : SUCCESS in 1h 05m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9314b3bc8e434167b50368651ee362e3 : SUCCESS in 1h 37m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3a0d47e9ddc840768eb718965fd3c578 : SUCCESS in 58m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9877c01391bf4ec788dd128f5a99ac91 : SUCCESS in 34m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7cd56f0e50b44b4ea47a94724ea0b321 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e951b4b4e16d070635b5a7f7c8287ce9024d7e2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 10:00:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8ab7b5a6c8968883fd0670bb550853a6e33fdee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 10:02:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54690f3cc0a340ad81c31b4b0d081f2b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44eea58cc6045f6ac381573079f6751 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21574334798047dd94aba6dd6529fb39 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e4a03e5c85556b19a019108022f399341950360","revisions":{"2e4a03e5c85556b19a019108022f399341950360":{"kind":"REWORK","_number":1,"created":"2024-11-11 17:18:45.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/79/934679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/934679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/934679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/934679/1"}}},"commit":{"parents":[{"commit":"5a039fcd9fe6927d24cb00485869edf6cb1de629","subject":"Do not fail on revert PlugVIPAmphora due undefined db_lb","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a039fcd9fe6927d24cb00485869edf6cb1de629"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-10-23 13:57:44.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-11-11 17:18:45.000000000","tz":0},"subject":"Remove amphora_health record on revert CreateAmphoraInDB","message":"Remove amphora_health record on revert CreateAmphoraInDB\n\nRecord in amphora_health created by first UDP request from booted\namphora.\nIt\u0027s possible to run creation/failover LB where Amphora will be booted\nand reverted on the final tasks (e.g. AmphoraPostVIPPlug) as result\nrecord in amphora_health will be created, but not be removed on revert.\nThis patch adds logic to remove amphora_health record on revert flow.\n\nCloses-Bug: 2085530\n\nChange-Id: Ia8b4612cd25a68b44d55b113f704756818eb8a5b\n(cherry picked from commit f76fa1ae330e933e86c666b857acfee06f675adb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e4a03e5c85556b19a019108022f399341950360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e4a03e5c85556b19a019108022f399341950360"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
