)]}'
{"id":"openstack%2Foctavia~935658","triplet_id":"openstack%2Foctavia~stable%2F2024.1~I87a26faf6c72640d8916bc3452123a9b5a74d39f","project":"openstack/octavia","branch":"stable/2024.1","topic":"rock9-curl-stable/2024.1","attention_set":{},"removed_from_attention_set":{"290":{"account":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"last_update":"2024-11-20 20:49:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I87a26faf6c72640d8916bc3452123a9b5a74d39f","subject":"Don\u0027t install firewalld in the amphora","status":"MERGED","created":"2024-11-19 14:29:35.000000000","updated":"2024-11-20 20:49:56.000000000","submitted":"2024-11-20 20:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935658-rock9-curl-stable/2024.1","cherry_pick_of_change":921424,"cherry_pick_of_patch_set":1,"meta_rev_id":"d3c2d415bb1747ce017f074afc6954f536faf81b","_number":935658,"virtual_id_number":935658,"owner":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 20:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-11-20 18:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-20 09:23:17.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-11-20 18:49:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 14:29:35.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-11-19 17:10:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 09:23:17.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-11-20 18:49:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4870e21239b269ce0235a213faae6cdf3f1867f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2024-11-19 14:29:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c558a50d6ae06bc727e9a40571cbbb437e246e85","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2024-11-19 14:54:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a38075ff5450d669d7948d2547861c4cf2f136c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:10:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ac9b40d5a943078a4ac29251b0cf5e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc5d7b363b3b4f11bc7020e95747036f : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9acb362854b144768560875c9145e15e : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06e7698accf7467387f09290cb7dccb3 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6afb6c23cc474df2a9fb5d7a2149cb24 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f7e2ad900a4f108314d5e0b57fe9cf : SUCCESS in 10m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fdfafbc475574caa89f474418fa18535 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e120abbd831a47eaa114f2ae105c0188 : SUCCESS in 8m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6e646959e97406e880e1445cb261185 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73a4f30e7b4347c791fa1717c6367557 : SUCCESS in 8m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b3e106c40424ebe887daf771af0292f : SUCCESS in 1h 37m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9d2d89188c3c4e81a74ba5457c10873e : SUCCESS in 2h 05m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bca3cf660f6940ab884346939126584b : SUCCESS in 1h 11m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/72f505b1ef0049719bc89b9ba49fd32d : SUCCESS in 2h 04m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/be09d14da63c4873b6f2ee8335dc5f83 : SUCCESS in 1h 14m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7655f5330c604bbe9dddb800c784b362 : SUCCESS in 45m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/79766f9e00804916a7d5c3781800c120 : SUCCESS in 43m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b2e50891d5894c0f97c47f8385e410f3 : FAILURE in 9m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d2ca925efc49495493c161f8b57457b4 : SUCCESS in 1h 06m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2e552cfd05c4c12b8f68d9aa783c181 : SUCCESS in 35m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ca58e3fb63ea4ef888ffd1b232dc7bd1 : SUCCESS in 2h 13m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7a6324d025b24dfb8f3f2d10d77ea8f9 : SUCCESS in 2h 35m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fb764ce808cc40d2be1fb541427db013 : FAILURE in 24m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a37215f6b1b4a2887f3d89a8a41c625 : TIMED_OUT in 2h 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bef06beebc58f8b4c45eb5626677b5f8901e7af","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-20 09:23:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30e1ee2ba2e9d5010b617241af5ad7b77ed190f3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 18:49:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2681ba7283a3d55a41da07c40cd8503e4645dd99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:50:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8108e4fce86b9e1b639189b46caa96c2f5e4575f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:49:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c60b6502b7e4788808b6b6322e8a3b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98c8e80d942451f8a58283e5c241caa : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1a4eaa1114441bbbcda7851442680fd : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c92e62510a640938f31c9cf17a4e383 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2414ce270ff24055a74ff175c95a0c13 : SUCCESS in 11m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c524dc63c4fd4229ac0f6b0af1577989 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5648535942b45e2810a88f06ca52440 : SUCCESS in 4m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d240e0b886a84d43881c12d462a56020 : SUCCESS in 1h 33m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/92d480043e9f4af395122cfc672685e6 : SUCCESS in 1h 38m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/702d8a3068ff4d81b5e6420a690f46dd : SUCCESS in 53m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7eb59a8ef9ed4c29b97c56c25b929f05 : SUCCESS in 33m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff8d4a05d07845c3a4421986265c5a96 : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d3c2d415bb1747ce017f074afc6954f536faf81b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:49:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e1301ff8a748be008fc8a690d4e28ba1fd8b92a","revisions":{"1e1301ff8a748be008fc8a690d4e28ba1fd8b92a":{"kind":"REWORK","_number":1,"created":"2024-11-19 14:29:35.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/58/935658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/935658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/935658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/935658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/935658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/935658/1"}}},"commit":{"parents":[{"commit":"f7c818b7dc80f5e79072aca9179275b0f12802ba","subject":"Merge \"Remove amphora_health record on revert CreateAmphoraInDB\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7c818b7dc80f5e79072aca9179275b0f12802ba"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-06-06 08:18:18.000000000","tz":120},"committer":{"name":"Antony Messerli","email":"antony@mes.ser.li","date":"2024-11-19 14:29:35.000000000","tz":0},"subject":"Don\u0027t install firewalld in the amphora","message":"Don\u0027t install firewalld in the amphora\n\nWhen building rockylinux images, DIB install the Minimal Install group\nwhich pulls firewalld. By default it allows only SSH traffic, so we\nwould need specific rules for the management and tenant traffic, but as\nwe are using security groups, it\u0027s safe to remove firewalld.\n\nChange-Id: I87a26faf6c72640d8916bc3452123a9b5a74d39f\n(cherry picked from commit e8fd3b7843b4971796aac69e5b1627d55451d082)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e1301ff8a748be008fc8a690d4e28ba1fd8b92a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e1301ff8a748be008fc8a690d4e28ba1fd8b92a"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
