)]}'
{"id":"openstack%2Foctavia~935659","triplet_id":"openstack%2Foctavia~stable%2F2024.1~I444293e0d6fcf125d2c3b12144bdf92c00f49dce","project":"openstack/octavia","branch":"stable/2024.1","topic":"rock9-curl-stable/2024.1","attention_set":{},"removed_from_attention_set":{"290":{"account":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"last_update":"2024-11-20 20:49:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I444293e0d6fcf125d2c3b12144bdf92c00f49dce","subject":"curl-minimal conflicts with already installed curl","status":"MERGED","created":"2024-11-19 14:32:11.000000000","updated":"2024-11-20 20:51:07.000000000","submitted":"2024-11-20 20:49:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935659-rock9-curl-stable/2024.1","cherry_pick_of_change":920989,"cherry_pick_of_patch_set":1,"meta_rev_id":"218c920a807a4c847f2d27c81b52fad2a97bcd05","_number":935659,"virtual_id_number":935659,"owner":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 20:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-11-20 18:49:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-20 09:24:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-11-20 18:49:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 14:32:11.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-11-19 17:16:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 09:24:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-11-20 18:49:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c34f65c66da308f3c435fd4722afd0882efcc125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2024-11-19 14:32:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5db27f4b6a7f57f7e71683521160b7bc391bf58a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 17:16:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e078a06f55a4fb88a9e3c8672504fcc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb49b418e57f486d82bd02ee660b340e : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1f50823c13471aacfd1afe7e84de42 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83851875245b445680abb1aa9d798d3a : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5281805ee8c74b888aca7e4ee5f89e72 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/352a34f5c4874775916abdc6219e1363 : SUCCESS in 12m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fcd320b264c54f03abe99d3b050cfc9f : SUCCESS in 5m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e0eb0f92fa24bd5987bda6c4e4e0e98 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/86bd07dd608d4ad994df657aa185f3f8 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28d8fec25d4c43458498f18279bf2e55 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9859edc55a1a4510bb0e354e3ae48c2f : SUCCESS in 1h 38m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3c2696d4276749d3aafa9b16a49079ef : SUCCESS in 2h 06m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e8a7bd398c8a41ac88a3025febde6c0c : SUCCESS in 1h 24m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0a7771a6db434dab9bee39b280d23be3 : SUCCESS in 2h 10m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/24d2bff312964a75a0e7476e65ec2e36 : SUCCESS in 1h 30m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cff860cca3ae4410aa5e326b7204c658 : SUCCESS in 44m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9cc41e73781a49da94604857e5f05350 : SUCCESS in 1h 00m 07s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0b5d646ac6e849f692267827f1e7c25d : FAILURE in 9m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c52d1f886cf845c3b62fdfa96cc8b4f6 : SUCCESS in 1h 13m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be7b07eac6954e98a1c76eaa4aca2074 : SUCCESS in 39m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4825502841b7401f82d1d1d4585231db : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/70a4c543567f4e378da0db2bf1204d91 : SUCCESS in 2h 36m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d68533aa6a1e4b8d90641ab7079426e9 : FAILURE in 34m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc18952e3551428cb2e435737f8be31c : TIMED_OUT in 2h 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14f6ddad1659bdfb7b649cca234b270787c52b61","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-20 09:24:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ccac33cc0df96bcbb3b9f9f10428955f75ecbc99","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 18:49:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc53e33036db75dfc3d0c1887eaf7bbbbc789f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 18:50:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf5ea286d857bd3351412f2d2f50e8b7b6a3d559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:49:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c5aed3de484cb8ae4a14eb570bf171\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db4c44fd3404cdd96edf6b2e72413dd : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc55621571bc48e5994edee73449e521 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6739e41684294994804d729bed503a1c : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678811ff26a2490ca1421390e3988bf8 : SUCCESS in 6m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ad6c33fec6844318969b30b2eef3f632 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bf060ff27098425f81c80a99d954477d : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/edd9616b530f498680a499a74c4afd7e : SUCCESS in 1h 47m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1425baab65604c73a5e38f4290150481 : SUCCESS in 1h 59m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/094e78de92cd4696a10ddec8fb9bac93 : SUCCESS in 1h 07m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/36ddb80afce74b8789d8c5bde85edf1c : SUCCESS in 41m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/941b26c6efe44d0f8ffaf37a461667ad : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5dc3a6db6f934adf9dbd21111e4fe9eeaf311bb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:49:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"218c920a807a4c847f2d27c81b52fad2a97bcd05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:51:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b3ac91eade4f37881121d806ad3ccb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab166d1045fe4ba594effcef7c05f414 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a97dcca946307921d9feabd82653e5930d18e7fe","revisions":{"a97dcca946307921d9feabd82653e5930d18e7fe":{"kind":"REWORK","_number":1,"created":"2024-11-19 14:32:11.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/59/935659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/935659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/935659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/935659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/935659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/935659/1"}}},"commit":{"parents":[{"commit":"f7c818b7dc80f5e79072aca9179275b0f12802ba","subject":"Merge \"Remove amphora_health record on revert CreateAmphoraInDB\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7c818b7dc80f5e79072aca9179275b0f12802ba"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-03 08:20:58.000000000","tz":120},"committer":{"name":"Antony Messerli","email":"antony@mes.ser.li","date":"2024-11-19 14:32:11.000000000","tz":0},"subject":"curl-minimal conflicts with already installed curl","message":"curl-minimal conflicts with already installed curl\n\n... on Rocky 9 during DIB build.\n\n2024-05-29 14:06:50.789 | \u003e Error:\n2024-05-29 14:06:50.789 | \u003e  Problem: problem with installed package\ncurl-7.76.1-29.el9_4.x86_64\n2024-05-29 14:06:50.789 | \u003e   - package\ncurl-minimal-7.76.1-29.el9_4.x86_64 from baseos conflicts with curl\nprovided by curl-7.76.1-29.el9_4.x86_64 from @System\n2024-05-29 14:06:50.789 | \u003e   - package\ncurl-minimal-7.76.1-29.el9_4.x86_64 from baseos conflicts with curl\nprovided by curl-7.76.1-29.el9_4.x86_64 from baseos\n2024-05-29 14:06:50.789 | \u003e   - conflicting requests\n2024-05-29 14:06:50.789 | \u003e (try to add \u0027--allowerasing\u0027 to command line\nto replace conflicting packages or \u0027--skip-broken\u0027 to skip uninstallable\npackages or \u0027--nobest\u0027 to use not only best candidate packages)\n2024-05-29 14:06:50.789 | returncode: 1\n2024-05-29 14:06:50.789 | install-packages failed with returncode 1\n\nChange-Id: I444293e0d6fcf125d2c3b12144bdf92c00f49dce\n(cherry picked from commit 5c0b011eacc9b5de1812bf2b52a663b192ff9d4a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a97dcca946307921d9feabd82653e5930d18e7fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a97dcca946307921d9feabd82653e5930d18e7fe"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
