)]}'
{"id":"openstack%2Foctavia~942885","triplet_id":"openstack%2Foctavia~master~Iadf100ab45499b59421425bc6874012a8bf0bde5","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-05-07 11:17:42.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-05-07 08:45:25.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-04-16 16:37:19.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-05-07 11:17:42.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-04-11 09:52:49.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Iadf100ab45499b59421425bc6874012a8bf0bde5","subject":"Support rotating server_certs_key_passphrase key","status":"MERGED","created":"2025-02-27 09:37:08.000000000","updated":"2025-05-07 11:18:38.000000000","submitted":"2025-05-07 11:17:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"942885","meta_rev_id":"9206383fadd0af98a8e87f07296c9ba845d1162f","_number":942885,"virtual_id_number":942885,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 11:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-16 16:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-05-07 08:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-04-16 18:42:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-05-07 08:45:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-27 11:23:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-10 21:12:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-04-11 07:26:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-11 07:26:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2025-04-11 07:26:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"c610592227130214a1514d383945abdab8ee7519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-27 09:37:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0ab4cbe2f07bde5ecb7decc85781af2d18b927b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 11:23:23.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dac035905f24a989841558807493cf6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/109910e91f1f46b79a7d77a715ff1b86 : FAILURE in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ea0a5081ee41f08208ecee0b0b356a : FAILURE in 13m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19d47b203e8f40db89df6544ae59b8ac : FAILURE in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b083ff6ace6b4278b5de170f7d3df5c9 : FAILURE in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/298c8143efc14ed79bbc1c00c46f899d : SUCCESS in 6m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/645ea44b970c487c982c54492434050c : FAILURE in 4m 18s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8f1ec14dc9fe43728cac096d28d18ed7 : FAILURE in 6m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/722435e439d34312aa775cc4b5e8bcf1 : SUCCESS in 3m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7075da1841b49f89bb9b58a8739d9f1 : FAILURE in 5m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4878d364989f46bfb7eefbc24fb4faa5 : FAILURE in 5m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d995ec362df94361b90907042bf3099c : FAILURE in 4m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bffc2d18576b4095b51c7509cee9eb00 : FAILURE in 35m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/09b554b0543b4d41938f16e1359b4dad : FAILURE in 27m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/60f94716a0e74c05a7e800fb29e8656a : FAILURE in 25m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97efcf23d6e24aebb7df89f5c8126c8d : FAILURE in 22m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/594a6938b6174d8b81979d8af708ad62 : FAILURE in 22m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2ed76a24c41c45bfb1967fe5ce13252b : FAILURE in 24m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3ce7dcf1ec5441748204ef819754f133 : FAILURE in 26m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da8c8b90a0074932af0a3373bdf20edd : FAILURE in 28m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f02a5c6790f54fc98c81a447bb157540 : FAILURE in 39m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a84c669f638744cca3c8e9d61ecf601b : FAILURE in 52m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3475272d44f24613a8dedeeb59ac8562 : FAILURE in 23m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1331c839619b4b3c8e9e98fb4b2de3b3 : FAILURE in 29m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e853a8d5dea4af9bcbfb00bbbb6bd16 : FAILURE in 36m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2a5aa7846059454e8f9d674fce6a0aab : FAILURE in 28m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5ecb30b7de54a1ebf3a0f8e11afc3b7 : FAILURE in 49m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"874f8b99e5ad0d84c29107e328e84d2d69d40df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-27 13:10:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7080e64b1cde9a70e158cb0fe882a27e9c93c294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 01:45:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/767c8133192c41279ddfc8c2dd12e1b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d909703ce00342388f86d340504bb13e : FAILURE in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ab66372e4a4e159202da28575f9096 : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1b7c45dd5740838484ae686bcb8905 : FAILURE in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d3252ba285e04d6c9288ecf606d256a9 : FAILURE in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6593cccfe4a34909b60893944e9ce87a : SUCCESS in 5m 32s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/71591bf5a6c04e0eac772573cd58121b : FAILURE in 14m 42s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dbf6642b17f94037b363b4ad6bd2cb0b : FAILURE in 4m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e40a0104479e4336a5153c82f4052aa7 : SUCCESS in 2m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bfda5691193e49c0ad3c1f7427cf9197 : FAILURE in 4m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f29518cef46845a2838e49f5aeeb31b7 : FAILURE in 3m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c626ea1c3e404e5d8744b7494b4281b5 : FAILURE in 14m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20af03f3053b4765887ed54b5099c4a4 : FAILURE in 26m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5cc4e31363264a3a8b8b7cf097423bf0 : FAILURE in 26m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/29139e23d5b6430bbd1ea55146cd6728 : FAILURE in 26m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f11ee8a35e38476c80dbd8b14f61dd7b : FAILURE in 23m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7bf781a26f340048f95e7df23b123f6 : FAILURE in 24m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b77d84c8a28e4865aabe32d9cbe4a6ba : FAILURE in 28m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c6a104c755514493afe5e6a05e8c0c40 : FAILURE in 28m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7063169275942c1bb547f126448e749 : FAILURE in 30m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1e8c248d1094e4c82690ca68281ec37 : FAILURE in 39m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5bbcdffcc6054f72a148b8da3f2afefc : FAILURE in 1h 02m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/839ad26213414956b32db40ab79f5c32 : FAILURE in 30m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c52920ea0ca1496e8cd3b99913cbf88a : FAILURE in 30m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73e079ae07e44ae584b3ecf7902cd946 : FAILURE in 38m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/68067a8e5dc64283be86c92c474e0bad : FAILURE in 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fc3b476171545db99eafc40123f98d2 : FAILURE in 47m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86bdd887aaccf9674142f5a494caada1827b0393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-28 10:36:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7581dda332e2c6aafabd509216c5994c4a49599f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 11:33:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eae1606a59df42988cd21fe8135c5dda\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eb4257ed0ba4eaa841d1cc7fb8133a0 : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de21ae4dead48fcab6ad240dc496708 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0822427b234b48b41a4ce2acb24e57 : FAILURE in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7786ac41a48b412383df48aec4e5b671 : FAILURE in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57b37a87fe4c4d8987f9788c52e9a69a : SUCCESS in 3m 35s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3ba3026628a14896982db69c51ea077a : FAILURE in 4m 26s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/7431ee9639aa4b2f959287f4d93767ec : FAILURE in 4m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/05ae5295a9bd419f83b1179bdadd357a : SUCCESS in 2m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a2cca9c2b4914c34888870119997ee96 : FAILURE in 6m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad0aee3ee12c477db9aa26b28ae2acc4 : FAILURE in 4m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d1060902f32645639bbb9d81a35f7b1e : FAILURE in 4m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5282aff1b754708bbaf5310d53d4170 : FAILURE in 24m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/48c26b6bb8cc48b0b9ff56ebf95940d3 : FAILURE in 24m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5b17c5bb00834ef7a97df4dbe1d3cca8 : FAILURE in 23m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90ec2e72f69f4cfc96da2369c42fce0b : FAILURE in 31m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c69f0339a24404f8bb88c0b425389c1 : FAILURE in 32m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3dc503e60e054573a878d7a2004a540e : FAILURE in 34m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d0f1801314cf430c84a3134310fc3d1e : FAILURE in 36m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f11ca24c3d944055b2ee1b23da45dd6d : FAILURE in 35m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f3ef6fc8547b465dbd8b155935cd227b : FAILURE in 46m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/859ea7532a8b438490d81be088c9bc20 : FAILURE in 48m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c22873ff4a24fd08239cc046233de54 : FAILURE in 34m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/19c92d27f2d54666a6d21df638b7b90d : FAILURE in 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edc819d8cd7747249c98ea39e347484d : FAILURE in 40m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/738ad7c41534483e95d10919217503ff : FAILURE in 26m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/67a8a78e943240f59c807545e1875439 : FAILURE in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6dc47be60ee6e106d775aa1d385059d97aedadce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-02-28 14:39:27.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9c4a87b53996e761065d0ab8c2864aecb89d3534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 17:55:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ec945f233be4c7aafbbe7b75b62e5c7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03edc07e2073436f9d7a5e206d79aec0 : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec62f16fcca4242be64ff83b50cd726 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c218321ed604aefa1abd0cd3aab4571 : FAILURE in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81956594b6d741cc876968254618e6af : FAILURE in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b6ef205dd847eba5c77626c6f77536 : SUCCESS in 4m 52s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c467b5cf9ba24b3fb0cd7ee812bd1953 : FAILURE in 5m 11s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/bf1818ae3e274719812cdd18189f300b : SUCCESS in 7m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b1b6b20829154e3681b3349ef554ba40 : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/416e9ec5f7974421b4059363fdf02869 : SUCCESS in 10m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da8edbe9c4254e5bb3297e5be844d442 : SUCCESS in 8m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68aad42be36c4dd0b10f974fce1883b3 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f4a82b1fb08408d98bb093d077fdbf8 : SUCCESS in 1h 49m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e45bbced568e4d9190646ddf560a4f76 : SUCCESS in 1h 48m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1522bc19df1b4216b4867fda963bd5b1 : SUCCESS in 1h 02m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef19d6508f0049d98c3c0e5f54a86652 : SUCCESS in 1h 52m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b965858cb4674e39a3e85be7bd947a51 : SUCCESS in 1h 34m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a544c742f1334bb5aa03eb8e123ee9eb : SUCCESS in 2h 23m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/59252043273c48748638297f7747dbfd : SUCCESS in 1h 28m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17f9b5364ce94b5db2b81ea390a1bb70 : SUCCESS in 58m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bdef64ef4b9140c296dd048d0e7f8357 : SUCCESS in 51m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/156f472149a24c0081e7364cef38ebe0 : SUCCESS in 43m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/826e8b3ea82748338a7adc2ec1eb0f7b : SUCCESS in 57m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2b00d0237c13432ab48d5de0975bef29 : TIMED_OUT in 2h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ceda6152f1f349e1bac57c5bfdfd634c : TIMED_OUT in 3h 03m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/47d9b6ec01e5460d9c5b3b535fc31f35 : TIMED_OUT in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d477b81f446147ac969118d655f56803 : TIMED_OUT in 2h 35m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51982bcada95d7e1e7dd183fc59aa823ea85be07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-03 09:43:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c7a5bbae202c22b1feadb0135427640834668a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 12:50:10.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/010ff356defd4f0cbdd3101d9957712f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e324906fada14291a69eb6e3c9818405 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4c6ce3b3c7475ea9d5cd01a39a015a : FAILURE in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50e077b687ca44dba4f627516818a69e : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10a5304c282e4bdbbda3c51f7ac84841 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b80534677144a3a8b92599090e4aff3 : SUCCESS in 6m 22s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b998ecf0da644df6829ecfe288941374 : SUCCESS in 3m 52s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3804478779664d528eeb3d1dd0b082fa : SUCCESS in 7m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4cbf6a81cdd248c1b4fee18d8c4046e0 : SUCCESS in 3m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e82a3be58d745729d6fd0b6f53e8013 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c406b978b8a4631879b4fa4d2e9d8d3 : SUCCESS in 7m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bd6df430c72747398a3e3d69652d678c : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/79c59a21e7ee43d4b9052db691381382 : SUCCESS in 1h 39m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8da18e51efcb45568c4d34d092e20bb1 : SUCCESS in 2h 33m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/401e98529e4a4486ab19c6b85fa467c3 : SUCCESS in 1h 12m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db58b61b08b84214b2f9afb1d430d7a7 : FAILURE in 2h 25m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbbd1fc0bd2147e8af3fa7d2d983f2ab : SUCCESS in 1h 10m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a8800a859f274b5ca200ecf7ee176bd3 : SUCCESS in 2h 18m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5250b0ad6c6c44adb6dcd6f3ec88878c : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c545defd8f944deabdfbd8945c175e5 : SUCCESS in 47m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7eb981fb6a2a4d4eac82a8b9bcd58596 : SUCCESS in 56m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/22697372494147be8fcdf8cf269ccb61 : SUCCESS in 1h 06m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53e433bf0dab47a5ac8d8e8ac615caf8 : SUCCESS in 44m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/381bd2e82e4448c584daef6f11d3e4d9 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/621e8bd5083a4b2d83df3efacf559584 : SUCCESS in 3h 02m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e5e633beb2764f53ae96003249b5a36c : TIMED_OUT in 2h 35m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bca7344a0cd4f4395424375b8f88024 : TIMED_OUT in 2h 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a7771cfc1ba36806edcb84effe0eb4496a583411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-03 13:44:04.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d7d1bf4d005cd1f263ddcb645313d47b52a38e73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 16:49:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae9da2483d4f4c2e9723c7c603e58beb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/881c3c25546e4418be5d84023e3e683b : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda3124dc46e4e8e97503fedd0b602d4 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93261928bf8d456f80bc99e86b28e578 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccb5d074440e48b4863d8b4ab6e091ea : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c4baa1369384d01b483c0f284623544 : SUCCESS in 5m 51s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/aa646fc529fe41d7af5b358103dd47e7 : SUCCESS in 4m 36s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/aa67a8cb9ed84e1c88a9897dd41c0b5c : SUCCESS in 14m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/34fbdd789d9844119dd0e4f540e05c03 : SUCCESS in 3m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1aecc333cd484fc0970dd99ca14e9aa0 : SUCCESS in 15m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a70d261acf894ccab776336e4921c2db : SUCCESS in 7m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1d450c5d68474094b042cff5b80e6492 : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed839edb025f4ccfa440563a4fc91ce1 : SUCCESS in 1h 14m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b66308dcf4364513b745a1334a8ad561 : SUCCESS in 1h 47m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dbe6a1a57c1541bdbce5b142b9482c86 : SUCCESS in 1h 03m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d766d2d4ae734aba89a322734d4d0f80 : FAILURE in 1h 46m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99e5bef400774d2fa2b4aa32110c980a : SUCCESS in 55m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a870228c4b084744a35d90f82b5299a5 : SUCCESS in 1h 55m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/893ef7cd1541455d97d2e9b8372b1cdc : SUCCESS in 1h 16m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/387d117ded994e489c74fe9539e2b3f4 : SUCCESS in 38m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b64623cc1c5b4d79a74dfcb32d7c71b4 : SUCCESS in 45m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ee7acf7b679d4a5ba9e13e2bcd80d6b5 : SUCCESS in 1h 12m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b1ee1e569af48909299daec9e573a64 : SUCCESS in 37m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2826217fa8de4779bec1c52dcdd8c17d : SUCCESS in 2h 12m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1bce667468c44f5beb6686c614612ac : TIMED_OUT in 3h 03m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e68cca3e7a124562a60af3f1a89614eb : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0c10f2d45d049e9b5806cd44739cf44 : TIMED_OUT in 2h 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"23d7f2ebbb3cb902c18e3741972f72d940277737","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-04 10:21:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"392f055bce4e2d4c10b8678954f94c46b32abf14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 13:50:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f60d0c5703146e28125bf872464a73d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/226fb787d42d47d1b0e95f7a8dbd3244 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a27665da66d45d28ca1ff4c47b927a8 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60149074148d415fb2587460185cedcb : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7fe845793c34c589e5a241a713100c6 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfa64db8396e4113abfc72bc54a47a86 : SUCCESS in 5m 37s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9f0342350cdc4d0c8cfa330a8c65bf6d : SUCCESS in 3m 50s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fe3ff56aa4654c3096369e1d87ccead9 : SUCCESS in 9m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d0d729bc677b4b36b77d7e62dd6e8894 : SUCCESS in 5m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a107ec097cf448edb7d612ae1fb5a32a : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b69a4c96a71d4aea92fdc03f7c89d89a : SUCCESS in 4m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/40570bb7446e4c1fabbd6f788f3f4475 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/33463eb1014a473c8127c6deec20033a : SUCCESS in 1h 23m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e0e6df7c53264c0d92bf38464eff4248 : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/42014ebc6dc84586aae9adc3e6d977d1 : SUCCESS in 1h 07m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ff6a21d7cb642bc82bafda66429a533 : SUCCESS in 1h 37m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ded7039bfe6e43ab8b86bc67847d42ea : SUCCESS in 1h 34m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/adcd7a62456d4638b236e8e6440eb755 : SUCCESS in 2h 32m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9f6ab9932dbc4fb9bf732c7d8419d409 : SUCCESS in 1h 06m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8fde6b53dc84400ebf8f6b3da657cbdb : SUCCESS in 55m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a9a2fd347e1440fa217ca9f60bdc67c : SUCCESS in 55m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6e27e6121b7c4e28afc8b0e9dc2e1d7d : SUCCESS in 1h 06m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cef98cafa9343e8a8dca6e08b34ad09 : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/40977923fec14ffcb22406d34dde52e0 : TIMED_OUT in 2h 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00f07b83125148029ed8e4ac7623a8c7 : SUCCESS in 2h 56m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/feb90823f5d641a2986c8d26ff1d0f3b : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/450b66d2271e4f31a43ca0e50ca3f824 : TIMED_OUT in 2h 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9a03b029d15474cda1e8099274c4b6e8f4567d0d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-05 15:43:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b009b0d480879b74fa395f2404bb5cfcd740e025","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-10 21:12:16.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"11e7f993435dedb5d0b1aad4d3bd79841a493af7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-11 07:26:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f2e5e0251e185bd3d57e9fcde90d6ef037d35ef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-11 09:09:53.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3db372dcffcf680c2ff2fed171130028f98669f9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-11 09:10:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f5d33baa79864e4551fa9395335edda1b7a940a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 12:11:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4653a97e167641b0bbdfbdbb6d306783\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2564957a5b2b49178944edae793f0d15 : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efb9fe7918a44060aa857830bfb40270 : SUCCESS in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b08405d30c2a4178a10a77c3c26b6e43 : FAILURE in 6m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83780c6789c9478583a201342a80f8ee : SUCCESS in 7m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a39f8b5d75244f4bbd324dead8f4d136 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98ca0eb6aab4aafa2f1d5556abbed8d : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73f4f7023d60448f97d37a221de39f12 : SUCCESS in 5m 10s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d600d76fb0294f4d8a6fdfd5d95504f4 : SUCCESS in 6m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e881475dbb7c4bbbaaf9e61a913083b7 : SUCCESS in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6830b5a123854b648fb96ffac630ec4c : SUCCESS in 2m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b3ac376c1da54bec953f313f13f2e24d : SUCCESS in 5m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4bd143efdd85462f89614ec5a2b7c124 : SUCCESS in 7m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b35f1e8ba9d049078972331a0933645a : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ebf598c62384070adea759011f03c5c : SUCCESS in 34m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/77e8f4558c864734aa7ddac2a5a8cb7d : SUCCESS in 2h 30m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/685a93d49b174f41adefef74b97528e3 : SUCCESS in 50m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/416db66f9b2842e99cdf182c964426d7 : SUCCESS in 1h 24m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7aeaaab7985d4140a17d4bc2b4d0df8e : SUCCESS in 1h 16m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4c14c73e817e4344a5154a375f559def : SUCCESS in 1h 26m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/34cb5b55addb4f4fbcb486683397613f : SUCCESS in 1h 11m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d2ad549a4e34c4b8e5b2a4ebf5e1d90 : SUCCESS in 41m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b22d276d9f6d48b48e61c47797172919 : SUCCESS in 50m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0281644e53a3480aa288de5c4d398dd5 : SUCCESS in 1h 09m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62d7fd7669c74557a00f4e8ed1b28eca : SUCCESS in 40m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22b7010665584ecb8623702ba5924b59 : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/44f1c56f1c21434bb14accddd2ec3bc7 : SUCCESS in 3h 00m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8393a1c1926a436c8e1d8cf50e160f5f : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f9503a6ded44827b10c9106c524c6a1 : TIMED_OUT in 2h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"32e7c46e26cb4a5ba34d2026125c8101079d4019","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-14 06:02:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"9e36fb9c4a8b24d00bbaccb8ea1d2a713dc010cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 08:19:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d558794dee664e6d96513045eb7831bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46b34ef0f29c4589a1247e7f03d810b3 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6481cb2218b84f41aefb4059a58be850 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f40b8777571f453abaa66e91e448a62c : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/372bdf029e7f48808f1b97b4fae3349e : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a51739f86874469af86fc4fa425721f : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6b45646fdb4446982458c6c2bb467a : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60cd2ae331e049949b6f407f31f1cc04 : SUCCESS in 5m 14s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cb9f3efdf66241858d4c6a4cbec1ecd9 : SUCCESS in 3m 59s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/474fd6e3ceed43d9a8f2241563cebbcf : SUCCESS in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/83ccc24775334facabec72a4049d6973 : SUCCESS in 3m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1a8041014b2a4793b52be7b46e1633cb : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d60c0ec92184cd9b699f480435a490b : SUCCESS in 6m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bb8c023b320a487a8bc0ca4972b12e01 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7069aa0b7c54b1d931503e588f849a0 : SUCCESS in 1h 21m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3351323488e44485aea8f00551d44800 : SUCCESS in 1h 43m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ec3822a6fa054cb0bd586850bdd5d481 : SUCCESS in 1h 09m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2544d07bb9d140628eb137d49b84ecca : SUCCESS in 1h 41m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a33745fb36e2482f808f800d3784135f : SUCCESS in 1h 13m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4810f50c06d84a8b853f458c6ccfc5ac : SUCCESS in 2h 03m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/227ae1f991404a18b988497dbe6bce8b : SUCCESS in 1h 02m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abb8af2b303c47d9b52d8f86a510297a : SUCCESS in 38m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/11808acfa2fa4cdc9aa6eb779caec1cc : SUCCESS in 48m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/398c74c27e2744129ac2374b394215f7 : SUCCESS in 50m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/238c46dfdac44706920d986b8bb029c4 : SUCCESS in 25m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b76b48b08d694d7ea5bdbe6d4e5be6b2 : SUCCESS in 1h 53m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d92e0aef61404c32a59ef59f65c219d1 : SUCCESS in 2h 15m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0fd25acd44e14abeb12905e233707810 : SUCCESS in 1h 47m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f6074478b7e4f52bedc85fd7b067c0b : FAILURE in 1h 44m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"25725fbf6257423d8b235dbdbf931904bb05b086","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-16 16:37:19.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8cb3d9c4fceac68c827e21857c95416151a0c501","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-16 18:42:17.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e3b15a9500ccdb08824f212d501bbdcfc7c1adf2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-05-07 08:45:25.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"b19acd4c6cd7d706f92f1beb40071cc40f0834be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 08:45:51.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"9e50834e285b74528569885323c14bcecadf3f08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:17:42.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5f84de5735347f082c72f373e742441\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806f871aa09c419eba7913eebe44f24d : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75533f9aca5f489da609d01a10b38f6d : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fce9551b7c2f435f8e7cacf21b5abeb8 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f91bb768394548b24a24bd37b310c8 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/679cb08ca1214ed6ab6716c356d511a4 : SUCCESS in 4m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ad2dc0f86404b6da18ecc5e0b4bbc1c : SUCCESS in 3m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0310bf0df2d4454b80901de76facfc55 : SUCCESS in 9m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/931b37ce63ee4c7184842db2b9bee0c2 : SUCCESS in 1h 15m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dd416248611d439c96a230777e337ef6 : SUCCESS in 2h 23m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/618e1972be6b4390add0d268478c946a : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d279166edb124962839a4873deeea5fd : SUCCESS in 2h 27m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f28fba78a9714f78b12d26058fb5fce6 : SUCCESS in 1h 32m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db892b4a8d90498f9cee6f635f39104e : FAILURE in 30m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d4014d88e09c437d930faf0367f7a370 : FAILURE in 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"74562c69ce9c11ada210933bd40e999cd46d225a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:17:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"9206383fadd0af98a8e87f07296c9ba845d1162f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:18:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0512bb7f1fd48d1a5a9b2b4dfda7758\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d727c09e1bc459f9dce5339ae143e30 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40f6fc32f46e432ab697a3519ef85486 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4dfcd861cdf86622c7d85e04d328a0a2fabd7938","revisions":{"cece3882764bc6bbe7a2834f9a0ac796e3a99de3":{"kind":"REWORK","_number":1,"created":"2025-02-27 09:37:08.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/1"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"subject":"wip: Support rotating server_certs_key_passphrase fernet key","message":"wip: Support rotating server_certs_key_passphrase fernet key\n\nTODO\n\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cece3882764bc6bbe7a2834f9a0ac796e3a99de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cece3882764bc6bbe7a2834f9a0ac796e3a99de3"}]},"branch":"refs/heads/master"},"a25adc9fb0f1657dfc37f744f66ffd7eda63f0ee":{"kind":"REWORK","_number":2,"created":"2025-02-27 13:10:52.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/2"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 13:10:41.000000000","tz":60},"subject":"wip: Support rotating server_certs_key_passphrase fernet key","message":"wip: Support rotating server_certs_key_passphrase fernet key\n\nTODO\n\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a25adc9fb0f1657dfc37f744f66ffd7eda63f0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a25adc9fb0f1657dfc37f744f66ffd7eda63f0ee"}]},"branch":"refs/heads/master"},"60635e292673d1ebf9f18c91264ca0c7933fedf6":{"kind":"REWORK","_number":3,"created":"2025-02-28 10:36:10.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/3"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-28 10:36:04.000000000","tz":60},"subject":"wip: Support rotating server_certs_key_passphrase fernet key","message":"wip: Support rotating server_certs_key_passphrase fernet key\n\nTODO\n\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60635e292673d1ebf9f18c91264ca0c7933fedf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60635e292673d1ebf9f18c91264ca0c7933fedf6"}]},"branch":"refs/heads/master"},"eaef12dfe1b174eab6c9e865922eecf0b9e1c8b2":{"kind":"REWORK","_number":4,"created":"2025-02-28 14:39:27.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/4"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-28 14:39:27.000000000","tz":0},"subject":"wip: Support rotating server_certs_key_passphrase fernet key","message":"wip: Support rotating server_certs_key_passphrase fernet key\n\nTODO\n\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eaef12dfe1b174eab6c9e865922eecf0b9e1c8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eaef12dfe1b174eab6c9e865922eecf0b9e1c8b2"}]},"branch":"refs/heads/master"},"4d80cb6a222d1a490498847c653db6cf5d5b85e2":{"kind":"REWORK","_number":5,"created":"2025-03-03 09:43:16.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/5"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-03 09:38:09.000000000","tz":60},"subject":"Support rotating server_certs_key_passphrase key","message":"Support rotating server_certs_key_passphrase key\n\nThe server_certs_key_passphrase was introduced to encrypt\nthe stored amphora certificates and the configuration option\nfor the passphrase has a default value.\n\nThis default value cannot be changed once assigned as there\nis data stored encrypted using the passphrase, if the\npassphrase is changed all amphora needs failover so that\nthe certificates is stored encrypted using the passphrase.\n\nThis wrap the existing cryptography.fernet.Fernet logic into\na utils helper function and converts it to using MultiFernet\nso that the first key in the list is used for encrypting data\nand all the others is available for decryption that means\nthat the the first key in the list can be changed and the\ncurrent key appended to the end of the list and when you\nhave organically (over time) performed amphora failovers you\ncan remove old keys.\n\nThe server_certs_key_passphrase config option is changed to\na ListOpt which is backward compatible with StrOpt.\n\nWe introduce a new FernetKeyOpt item_type for the config\noption because ListOpt of strings (or if we would have\nused a MultiStrOpt) does not support the `regex` kwarg\nfor regex based validation.\n\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d80cb6a222d1a490498847c653db6cf5d5b85e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d80cb6a222d1a490498847c653db6cf5d5b85e2"}]},"branch":"refs/heads/master"},"d57aac9a418b4bf51702403f2654c3f5a55b132e":{"kind":"REWORK","_number":6,"created":"2025-03-03 13:44:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/6"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-03 13:43:57.000000000","tz":60},"subject":"Support rotating server_certs_key_passphrase key","message":"Support rotating server_certs_key_passphrase key\n\nThe server_certs_key_passphrase was introduced to encrypt\nthe stored amphora certificates and the configuration option\nfor the passphrase has a default value.\n\nThis default value cannot be changed once assigned as there\nis data stored encrypted using the passphrase, if the\npassphrase is changed all amphora needs failover so that\nthe certificates is stored encrypted using the passphrase.\n\nThis wrap the existing cryptography.fernet.Fernet logic into\na utils helper function and converts it to using MultiFernet\nso that the first key in the list is used for encrypting data\nand all the others is available for decryption that means\nthat the the first key in the list can be changed and the\ncurrent key appended to the end of the list and when you\nhave organically (over time) performed amphora failovers you\ncan remove old keys.\n\nThe server_certs_key_passphrase config option is changed to\na ListOpt which is backward compatible with StrOpt.\n\nWe introduce a new FernetKeyOpt item_type for the config\noption because ListOpt of strings (or if we would have\nused a MultiStrOpt) does not support the `regex` kwarg\nfor regex based validation.\n\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d57aac9a418b4bf51702403f2654c3f5a55b132e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d57aac9a418b4bf51702403f2654c3f5a55b132e"}]},"branch":"refs/heads/master"},"2df5142420b9c2b303401b4e95f2a09f09c6787f":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-04-11 09:09:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/7"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-11 09:09:53.000000000","tz":0},"subject":"Support rotating server_certs_key_passphrase key","message":"Support rotating server_certs_key_passphrase key\n\nThe server_certs_key_passphrase was introduced to encrypt\nthe stored amphora certificates and the configuration option\nfor the passphrase has a default value.\n\nThis default value cannot be changed once assigned as there\nis data stored encrypted using the passphrase, if the\npassphrase is changed all amphora needs failover so that\nthe certificates is stored encrypted using the passphrase.\n\nThis wrap the existing cryptography.fernet.Fernet logic into\na utils helper function and converts it to using MultiFernet\nso that the first key in the list is used for encrypting data\nand all the others is available for decryption that means\nthat the the first key in the list can be changed and the\ncurrent key appended to the end of the list and when you\nhave organically (over time) performed amphora failovers you\ncan remove old keys.\n\nThe server_certs_key_passphrase config option is changed to\na ListOpt which is backward compatible with StrOpt.\n\nWe introduce a new FernetKeyOpt item_type for the config\noption because ListOpt of strings (or if we would have\nused a MultiStrOpt) does not support the `regex` kwarg\nfor regex based validation.\n\nCloses-Bug: #2106810\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2df5142420b9c2b303401b4e95f2a09f09c6787f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2df5142420b9c2b303401b4e95f2a09f09c6787f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4dfcd861cdf86622c7d85e04d328a0a2fabd7938":{"kind":"REWORK","_number":8,"created":"2025-04-14 06:02:34.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/942885/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/942885/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/942885/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/942885/8"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-02-27 09:36:20.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-14 06:02:24.000000000","tz":120},"subject":"Support rotating server_certs_key_passphrase key","message":"Support rotating server_certs_key_passphrase key\n\nThe server_certs_key_passphrase was introduced to encrypt\nthe stored amphora certificates and the configuration option\nfor the passphrase has a default value.\n\nThis default value cannot be changed once assigned as there\nis data stored encrypted using the passphrase, if the\npassphrase is changed all amphora needs failover so that\nthe certificates is stored encrypted using the passphrase.\n\nThis wrap the existing cryptography.fernet.Fernet logic into\na utils helper function and converts it to using MultiFernet\nso that the first key in the list is used for encrypting data\nand all the others is available for decryption that means\nthat the the first key in the list can be changed and the\ncurrent key appended to the end of the list and when you\nhave organically (over time) performed amphora failovers you\ncan remove old keys.\n\nThe server_certs_key_passphrase config option is changed to\na ListOpt which is backward compatible with StrOpt.\n\nWe introduce a new FernetKeyOpt item_type for the config\noption because ListOpt of strings (or if we would have\nused a MultiStrOpt) does not support the `regex` kwarg\nfor regex based validation.\n\nCloses-Bug: #2106810\nChange-Id: Iadf100ab45499b59421425bc6874012a8bf0bde5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4dfcd861cdf86622c7d85e04d328a0a2fabd7938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4dfcd861cdf86622c7d85e04d328a0a2fabd7938"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
