)]}'
{"id":"openstack%2Foctavia~943397","triplet_id":"openstack%2Foctavia~master~I6a826b05175cf4ceb1a051b2ad1b665a78d514dd","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-03-05 16:05:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I6a826b05175cf4ceb1a051b2ad1b665a78d514dd","subject":"Re-enabling doc job","status":"MERGED","created":"2025-03-05 14:27:50.000000000","updated":"2025-03-05 19:24:22.000000000","submitted":"2025-03-05 19:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943397","meta_rev_id":"b3cc32c79501300a4444b1cb5c5790fade41f868","_number":943397,"virtual_id_number":943397,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-05 19:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2025-03-05 16:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-05 14:50:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2025-03-05 16:05:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-05 14:50:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2025-03-05 17:33:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9b3ae11f3b0350a3fd78af7b8b12f683cdeda87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-03-05 14:27:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f12d77a7026f780303f184f6ff0aca875ff328d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-03-05 14:50:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"643aab4e62351402a1164fdedafb6f7a97d0852b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-03-05 16:05:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d49b2d540c774f3e6619dce47317f1b0931c94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 17:33:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70ea571e0658425a9042c94d6d2b7c98\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/997e0b0a0e054ea989fed6bb80f0b24f : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36b17647b3f4fdaabdd9b6dacb6dbfd : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/067aa9be69ad483186eab17a65651d13 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a00a3dcd3905426b86e01fab9f5ba40e : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d4b53c7a41432097e52e579eaf30b9 : SUCCESS in 12m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ca5b168881f043f0b39e929b53c70f37 : SUCCESS in 5m 11s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/9846d2d54dec447592b56dda32c9bb93 : SUCCESS in 7m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f989e5c12c64720abd964f9ec9905da : SUCCESS in 5m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d355fe4c4854ba6a7b24f3120c3c98a : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62c7a4e6532b4be19f266f941ec0c9d3 : SUCCESS in 6m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3f48da70be724702bb53ab754115756b : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/390b791dd7994f119664b6d8c57dd4eb : SUCCESS in 1h 36m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d699f2207e4049efa1b4d364f24f2579 : SUCCESS in 1h 53m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1ccf8c26b5ec40a4846a7ef15695905c : SUCCESS in 1h 02m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/431fc4977d9b41c4a2f4903403dd7278 : SUCCESS in 1h 49m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74fa5ba2a4754ab18f3a3df57eb555c2 : SUCCESS in 58m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1290938eb97349fa8bcbe0c0702251c7 : SUCCESS in 1h 58m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/301472fc8db444db8fe1ec24d7d4da66 : SUCCESS in 1h 17m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b25ecfef42c449b49244cfcdd2e6d087 : SUCCESS in 34m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1fa1feac43074636b067593d1a48ab52 : SUCCESS in 41m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/64ee45f8d8d34020932e6288308b33df : SUCCESS in 51m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1aa400715a74eafb2ff458197765e61 : SUCCESS in 34m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/78a49d0fc2b442cba19829da0222f007 : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/351a841eab544aba8674ff32d0b97a5c : TIMED_OUT in 3h 02m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc244d2f9d2b4610af6ad33b8e16d282 : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5920737100f4b74a2d99a0d3e90a310 : TIMED_OUT in 2h 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"585567611bd8619baa4c13490d42c15acde3458f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 17:33:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf4761f62a2c545f264f4122df9896ddb7057200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:22:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f917c1f5a2742c9bb1b19b3992383a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebb2e24df75f434b815eff66db821d39 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/097d9e6b1e1a42589df3308af1ba78ef : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a373f5a3614a419fb260fdcd773b70d3 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02622dbdf6444cead4f52c1610336b1 : SUCCESS in 7m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e2728f4272604e62a6edbf4478cc607f : SUCCESS in 2m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/69984ef9862749edaa86b0c5354020f0 : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24dc8a80d04c49ac9e2b779fbff0a5f7 : SUCCESS in 1h 42m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4cc20820b0b34caaa748ea2aebda2890 : SUCCESS in 1h 46m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/462d77c277a5408190cc0d6042a627fe : SUCCESS in 1h 03m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6c7962ca07546c28f536809040fb3e7 : SUCCESS in 1h 41m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa00b3e4951a4cb2b6503de8a6b65fb0 : SUCCESS in 1h 30m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/22ddaa27c7384328b0c6ea6dc305d7c5 : SUCCESS in 36m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/21b0ac72a2a84bf98206f21fabeb618a : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"34827659f8fb1c5420109e3560d4763c1a321324","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:22:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3cc32c79501300a4444b1cb5c5790fade41f868","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 19:24:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df06010c87f14ca290da76165b17892c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee98edb87c449d5b97e2863af7682f3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d555babf25874f62032ef7ebc80891f528a7cd0","revisions":{"4d555babf25874f62032ef7ebc80891f528a7cd0":{"kind":"REWORK","_number":1,"created":"2025-03-05 14:27:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/97/943397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/943397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/943397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/943397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/943397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/943397/1"}}},"commit":{"parents":[{"commit":"083dd12b086d521343940d92ff1886cc114a8673","subject":"Merge \"Added RBAC for vip_sg_ids\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083dd12b086d521343940d92ff1886cc114a8673"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-03-05 14:27:39.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-03-05 14:27:39.000000000","tz":60},"subject":"Re-enabling doc job","message":"Re-enabling doc job\n\nChange-Id: I6a826b05175cf4ceb1a051b2ad1b665a78d514dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d555babf25874f62032ef7ebc80891f528a7cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d555babf25874f62032ef7ebc80891f528a7cd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
