)]}'
{"id":"openstack%2Foctavia~944004","triplet_id":"openstack%2Foctavia~master~I8acc040c5e834c1f68aed6ec15acd86742a8af1f","project":"openstack/octavia","branch":"master","topic":"migrate-to-noble","hashtags":[],"change_id":"I8acc040c5e834c1f68aed6ec15acd86742a8af1f","subject":"Testing doc job on Ubuntu Noble","status":"ABANDONED","created":"2025-03-11 05:29:42.000000000","updated":"2025-03-11 10:58:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aab7288617cbfae0a54f47828390883cc3404d8f","_number":944004,"virtual_id_number":944004,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-11 08:17:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-11 08:17:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81e546bc25621b3c1e78bce2262ba226f3cdb436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-11 05:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7fe8c152b772b41671b388166839940f2e1cd83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 08:17:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6579fac44232420591deb9b441f90a9b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af2467bcede74e39aee9340d83248fe5 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4ef355067f2435ba7786c4b319b8d2b : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af10f3534eb541c3b78641904670db4a : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88da0e0ce49e45bc86aa6e5f6625461f : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d2fd6813caf4476b7519f99d66cbee5 : SUCCESS in 12m 14s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cf704fb3a9934b71a916869916225972 : SUCCESS in 4m 07s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8cbd6c73c7ca4bda9397d0fc6a7c3180 : SUCCESS in 7m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/42f6f45b45404408af5f27eb9cb8096d : SUCCESS in 3m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18a5a6cc983f4f898880976b1581d94f : SUCCESS in 4m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8dfaa54d8bd432d91076e3a0ad058cd : SUCCESS in 6m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e015c3c997004a8b872033741cd278b8 : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d37a49c06e1463a9e2b92c4023db358 : SUCCESS in 1h 22m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/421b7cdceb3f42d1afe81efbcc2b4c7e : SUCCESS in 1h 55m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c18ab68afe9c4f9aa9fb2e0064f2cdbe : SUCCESS in 47m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d1a36357ed1407f90b6945d86a817f4 : SUCCESS in 1h 58m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da3d25125e1d4d34b70cb25a5c8cb751 : SUCCESS in 1h 27m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7ffdfe6b21234adfb4ca2d52035fdbfa : SUCCESS in 1h 27m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9609b1f418bc4714a187aa9d6f0a2104 : SUCCESS in 54m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2349b7068b844408a1454a5325b1fb4d : SUCCESS in 38m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97a4f5f9c379471d9f89776a08812e9d : SUCCESS in 1h 00m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/70c5f22122ef48739a8efd8ee064621b : SUCCESS in 53m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91312873a01c4864a59b534296bc7304 : SUCCESS in 37m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ed7d159dc7d4691a38af9841462b935 : TIMED_OUT in 2h 34m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec23d2f358444f03861378cd4737b6c1 : SUCCESS in 2h 44m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bfc028f535814aa5b95c1e5fc5337eae : SUCCESS in 2h 05m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/75b327db4ccb472abf6d350dd39fc13e : SUCCESS in 2h 02m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aab7288617cbfae0a54f47828390883cc3404d8f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-11 10:58:05.000000000","message":"Abandoned\n\nall good, doc job is passing on Noble","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5764d7ddc70334735cc8df4bf25271357059dd2","revisions":{"c5764d7ddc70334735cc8df4bf25271357059dd2":{"kind":"REWORK","_number":1,"created":"2025-03-11 05:29:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/944004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/04/944004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/944004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/944004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/04/944004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/04/944004/1"}}},"commit":{"parents":[{"commit":"27f5dd211d95679b57193b27ba74015d2e1bfde5","subject":"Merge \"Drop VIRTUAL_ENV\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f5dd211d95679b57193b27ba74015d2e1bfde5"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-03-11 05:29:19.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-03-11 05:29:19.000000000","tz":-420},"subject":"Testing doc job on Ubuntu Noble","message":"Testing doc job on Ubuntu Noble\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/935459\nChange-Id: I8acc040c5e834c1f68aed6ec15acd86742a8af1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5764d7ddc70334735cc8df4bf25271357059dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5764d7ddc70334735cc8df4bf25271357059dd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
