)]}'
{"id":"openstack%2Foctavia~944315","triplet_id":"openstack%2Foctavia~master~Icb950b8f06db4bb3a354308b46f80db58c09b656","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2025-11-09 06:44:42.000000000","reason":"Change was abandoned"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-11-09 06:44:42.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-13 01:07:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-11-09 06:44:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icb950b8f06db4bb3a354308b46f80db58c09b656","subject":"Add validation vip_address in vip_subnet","status":"ABANDONED","created":"2025-03-14 07:58:09.000000000","updated":"2025-11-09 06:44:42.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0656e2ec7fbbb7dc390166efa50a3fa3795d80c3","_number":944315,"virtual_id_number":944315,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-19 00:42:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":-1,"date":"2025-03-28 20:51:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":-1,"date":"2025-04-15 15:18:33.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 15:35:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-16 08:11:57.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-03-16 08:11:57.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-03-16 08:11:57.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"7989917ee12e33eeff3761d620d97c71664233fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-14 07:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54da78f46501077c5557ddc8008c117943cf6edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 15:35:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66fdc7cf9380412d88812d9a64d1810b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17ec245d16e2479890151932d8cf3407 : FAILURE in 2m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35432f169884477cbc2a3c498f9c24c7 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b8b9af97f4f4e1eadecad6fd6ea5e72 : FAILURE in 1m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e441e641a7a4995b5ad901593b1c873 : FAILURE in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd2f0591fd0428ea14989e8087766c1 : FAILURE in 4m 26s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3dfa0b3f3caf4517b547789ad6d02f8d : FAILURE in 2m 26s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5e6e12eb783c4f8fb82380317beb32a8 : FAILURE in 3m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/971a67992bf4493799055df51415d13b : SUCCESS in 3m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/73053b4ea004414b99ddfd50bb907def : FAILURE in 1m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08c0cbdba29145e5b900fd127cb2efc9 : FAILURE in 2m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2286208f474a4c89913e754e11c6680a : FAILURE in 4m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d03aa79ac0fb42db88c2891a177bb99a : SUCCESS in 1h 17m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f949f4915e4043b6959845fadbb4dd39 : FAILURE in 46m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/de4ac65d9cb64848976a0626f454ce01 : TIMED_OUT in 2h 32m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99d8991f6bdd4e0ab2f2f03b793d1c18 : FAILURE in 45m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58e8e729c47f42a0b4e18fa27a3c74e6 : TIMED_OUT in 2h 32m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9eb04118da26475989faaac910c9ea94 : FAILURE in 42m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/76e7cfd845044761a284b3659efccb05 : RETRY_LIMIT Host unreachable in 2h 31m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26ee0d34f9bd4897a686ccb1247a204a : FAILURE in 37m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/55fd7fae387c4f25955ca83b07baae05 : SUCCESS in 55m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e0941e6b72c5495f8c4db7033f1ced91 : FAILURE in 1h 05m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07387dcfc47a42bf8269b99315e26df4 : FAILURE in 43m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/83b8560abc4b44ed914e925190c47f9d : TIMED_OUT in 2h 32m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b4ba3bc073e64f77b67dce6559ef52f1 : FAILURE in 2h 54m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e65da313ec5b4756b4ec6cb772c51d10 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/49d8c83d64894ab3a4b2209cd77fcf8c : TIMED_OUT in 2h 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4d3964d8a69534e21e2b35422e4966c2a559b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-14 16:09:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6f68990b5cf2eb4bc4ec194427985a4f3e0ccdd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 19:20:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3abd2d273b614cbb9df403f7285a741f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f40d7520451e4f3c95f36db80fe4d90f : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df5390c27b6434888a03667dbed3ea6 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e6e5d6a482b4dd596273d5e5219acb4 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73b545af0a9148908f7526e7f160a6d2 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b807365d7e4dc28a670903cf6d82dc : SUCCESS in 13m 05s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f32cfa6249ec4dd0a1aee807de2c085e : SUCCESS in 4m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/02d47c6d7a6d475fafe03207128c67eb : SUCCESS in 10m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/96a083fd527d49ebab8c03ffe19252f9 : SUCCESS in 1m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7082799e0b56479db7e4bd4f78908e86 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36b74b2ae1224d2f871f245b436f8b35 : SUCCESS in 4m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ed665b5a6825478eaaa2299f7d6a8077 : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65015c04817a45308d61eda363a5e5a5 : SUCCESS in 1h 46m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a0ca18ca223e4b94b151ae9b7e4a29be : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6f51e1857c564b759e556f05ba0fdf48 : SUCCESS in 1h 21m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a869ce9b3525422ea25a38c44d944137 : SUCCESS in 2h 32m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1741a28568164daab998de718e278caa : SUCCESS in 51m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b5183f2da50b4b069d88b56f09572cb3 : TIMED_OUT in 2h 34m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5fcb5c5ffbc3478dabaa3fa73242dfcd : SUCCESS in 1h 25m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1b51e9a519d49ee929154979dd85356 : SUCCESS in 54m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/092c88993c5e4bdea6c95c437dfb25a8 : SUCCESS in 1h 03m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9d02f6dd99f54822859b5de7d86f6f29 : SUCCESS in 1h 14m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76be95a1b1cd4090afb959bfc9bd125b : SUCCESS in 45m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ebb51be37d2b4367a328026b94afac26 : TIMED_OUT in 2h 34m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47ad1a77126a42a1b900d6856ce28ef6 : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/09e27d27b1d748afb5820df2d100001c : SUCCESS in 2h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2028a2bb69347bb9ad92ea2cac73a85 : TIMED_OUT in 2h 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e61edce3a33a274c1bf27b63f2d02361b11f3fd2","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-15 07:02:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ca3de3ac64c84449106c958382a6d99dfdb82ea","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-15 07:10:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"938b1b9949f8f9c687581613f6a0a845f1a97a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-15 09:56:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df8f81d94b04f8b8cafb926f0e4fcde\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c80e3f9cdb34115b1d15b0ef59be57c : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4d603cd73842a3868f42bc7ed1fdfb : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac195cba25fd4364bdac6c6f52055416 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4fe7d8845254e1ab0cdae413b4c0cf1 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bdd0c265650455fa86b20a853cc6919 : SUCCESS in 5m 38s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bfb51e29fa324e558f3a29ca89e4c59b : SUCCESS in 3m 38s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3b08c08f86e24d0785c83598b6cf38aa : SUCCESS in 3m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c28c1b1c47ea4bfdbea69b53ae589054 : SUCCESS in 2m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17b493c542a8427c85be7a32ba01cdc4 : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4575a9d56338438bbfd4b347507dde07 : SUCCESS in 3m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/48544c74e91c44b1b46c9170e0bd3bda : SUCCESS in 4m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b48b57aee61434e939d2fbbd4f7c5da : SUCCESS in 1h 37m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9efe629133ab4fd2913cf09a6ea98012 : SUCCESS in 1h 50m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/daabb80f517f4c35bdc6a01e400cf361 : SUCCESS in 1h 05m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cad286196c64d71ab34afd8b707fb0b : SUCCESS in 1h 46m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad82ba3dc384f51b88307bfa683faed : SUCCESS in 1h 01m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b57ef98a7a1548929bd9f4472089c522 : SUCCESS in 1h 49m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/77dc9d5834dc44d4b91c00bf290c4433 : SUCCESS in 1h 02m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8552be3a0494f81af66f57f8fc705ba : SUCCESS in 35m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2ba6e7c30a434e5083512fde734ac992 : SUCCESS in 47m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1da7998842034dd690a69145afa36ecd : SUCCESS in 49m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/228bd910188f46a4aff5b06d9fd4bc80 : SUCCESS in 32m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/018d72a4759d48eab2ca9a3f4fa1a352 : TIMED_OUT in 2h 33m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00734dadf6d4430f8107581f9b9090b0 : SUCCESS in 2h 53m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bd692c01820a4e389cfe4fdb724b3def : SUCCESS in 2h 14m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf418075dc3a40dca17d6e67e52f590f : SUCCESS in 2h 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"028ee1335ac16faf64f63a59434efc8179d1f2cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-18 21:50:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"01ea0228fb84f1557d187365bcc8d58cd38a323c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 00:42:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/644636253ce240b39a63a9ad2e418358\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58b2ae29b9514ae2b1f961613368e488 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833fb174263241f6ad12e1ec6fa5bf27 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70aea8b1602145f7a5a5f3ea43da9879 : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b6d0bff19eef400ea9a8c91d82bda250 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66bb5e1b132347629fea6b44c969f189 : SUCCESS in 9m 30s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8048d4d0b9f04fa3b64c6bfd5a865474 : SUCCESS in 2m 34s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/9017d31879874e3ebe3b79a115fdf5f6 : SUCCESS in 5m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c60e87ca968f4577b42a20f60b214b38 : SUCCESS in 3m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/960f14680ba14ae486d34922b52ed3fe : SUCCESS in 5m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5683e795e38b4de589e3d73d7a4223a0 : SUCCESS in 3m 47s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7550ceecba8d4e20b64c92785989da08 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/914dd35cd0c84ff9ab3b5fcf4cff373b : SUCCESS in 33m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d98ca649c6494d30996654caf7364a4e : SUCCESS in 2h 05m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/58800dcbf2e54c4d8d791956abc99db4 : SUCCESS in 40m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f45e6f9f64d1428ea54aa5bd044e231a : SUCCESS in 2h 03m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/980baee3edb24c728d69e819ab93e6ea : SUCCESS in 46m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c96eb5ae2898473cb8de062c6c1f9094 : SUCCESS in 2h 06m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/932ce5359ca244e1993ac7b318cd7d82 : SUCCESS in 1h 16m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c49066310b04bcc846b09b18136fd96 : SUCCESS in 37m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/617a4afc7e014be3a9064b93b62696f6 : SUCCESS in 56m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3a020eb5dffa442fab22e8f1152c029d : SUCCESS in 33m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c94dc6ab7bb64649a2493c2c84f4fb85 : SUCCESS in 36m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/771df43728014dbd9174fecb91458d06 : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8cabdc2e74d4f9191b807e6f69a2ba4 : SUCCESS in 2h 50m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/48e665430175443e9a65b5d84ca4144c : SUCCESS in 2h 29m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f817f76237e44b489d085643b978ef8f : SUCCESS in 2h 07m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0487ca3fb67831f121d983e8bf5cc85770d4e385","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-03-28 20:51:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2b3107d15a440e515aa27da9eb32b1a44d20b440","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-03-31 07:09:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"141b3f52b925c07e8daaf8b0341cd0ef606fa993","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-04-15 15:18:33.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da1c4f2d91454265cc2a13106402becf09b71625","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-13 01:07:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0656e2ec7fbbb7dc390166efa50a3fa3795d80c3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-11-09 06:44:42.000000000","message":"Abandoned\n\nanother fix was proposed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9fd5737afe729ed63e82693c0362e77d7c497a17","revisions":{"471337a5a4477bcee01001865da93a38da7bcf90":{"kind":"REWORK","_number":1,"created":"2025-03-14 07:58:09.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/15/944315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/944315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/944315/1"}}},"commit":{"parents":[{"commit":"27f5dd211d95679b57193b27ba74015d2e1bfde5","subject":"Merge \"Drop VIRTUAL_ENV\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f5dd211d95679b57193b27ba74015d2e1bfde5"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-12 20:50:46.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-14 07:57:34.000000000","tz":240},"subject":"Add validation vip_address in vip_subnet","message":"Add validation vip_address in vip_subnet\n\nValidate 3 cases:\n- when specified ip_address and subnet_id, that this ip_address\n  belongs\n  to specified subnet\n- when specified ip_address and subnet_id, that this ip_address is\n  not\n  already exist in subnet\n- when specified ip_address and network_id, that this ip_address is\n  not\n  already exist in subnet\n\nCloses-Bug: #2052682\n\nChange-Id: Icb950b8f06db4bb3a354308b46f80db58c09b656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/471337a5a4477bcee01001865da93a38da7bcf90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/471337a5a4477bcee01001865da93a38da7bcf90"}]},"branch":"refs/heads/master"},"d508ec2138d88faf7d6b41ac16bcbedfa4c0a153":{"kind":"REWORK","_number":2,"created":"2025-03-14 16:09:21.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/15/944315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/944315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/944315/2"}}},"commit":{"parents":[{"commit":"27f5dd211d95679b57193b27ba74015d2e1bfde5","subject":"Merge \"Drop VIRTUAL_ENV\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f5dd211d95679b57193b27ba74015d2e1bfde5"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-12 20:50:46.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-14 16:09:11.000000000","tz":240},"subject":"Add validation vip_address in vip_subnet","message":"Add validation vip_address in vip_subnet\n\nValidate 3 cases:\n- when specified ip_address and subnet_id, that this ip_address\n  belongs\n  to specified subnet\n- when specified ip_address and subnet_id, that this ip_address is\n  not\n  already exist in subnet\n- when specified ip_address and network_id, that this ip_address is\n  not\n  already exist in subnet\n\nCloses-Bug: #2052682\n\nChange-Id: Icb950b8f06db4bb3a354308b46f80db58c09b656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d508ec2138d88faf7d6b41ac16bcbedfa4c0a153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d508ec2138d88faf7d6b41ac16bcbedfa4c0a153"}]},"branch":"refs/heads/master"},"9fd5737afe729ed63e82693c0362e77d7c497a17":{"kind":"REWORK","_number":3,"created":"2025-03-18 21:50:56.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/15/944315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/944315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/944315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/944315/3"}}},"commit":{"parents":[{"commit":"27f5dd211d95679b57193b27ba74015d2e1bfde5","subject":"Merge \"Drop VIRTUAL_ENV\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27f5dd211d95679b57193b27ba74015d2e1bfde5"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-12 20:50:46.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-18 21:50:38.000000000","tz":240},"subject":"Add validation vip_address in vip_subnet","message":"Add validation vip_address in vip_subnet\n\nValidate 3 cases:\n- when specified ip_address and subnet_id, that this ip_address\n  belongs\n  to specified subnet\n- when specified ip_address and subnet_id, that this ip_address is\n  not\n  already exist in subnet\n- when specified ip_address and network_id, that this ip_address is\n  not\n  already exist in subnet\n\nCloses-Bug: #2052682\n\nChange-Id: Icb950b8f06db4bb3a354308b46f80db58c09b656\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fd5737afe729ed63e82693c0362e77d7c497a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9fd5737afe729ed63e82693c0362e77d7c497a17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
