)]}'
{"id":"openstack%2Foctavia~945171","triplet_id":"openstack%2Foctavia~master~I7a3d7fb6ba1d5b32071b590a79fb039a465fe5fd","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7a3d7fb6ba1d5b32071b590a79fb039a465fe5fd","subject":"Bump octavia-lib requirement to 3.8.0","status":"MERGED","created":"2025-03-20 22:10:47.000000000","updated":"2025-03-21 02:43:53.000000000","submitted":"2025-03-21 02:42:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945171","meta_rev_id":"672a726dc4661ac64a5f53dd7e229eb78e0019d2","_number":945171,"virtual_id_number":945171,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 02:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-03-20 22:12:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-03-20 22:12:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-21 00:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"541653fa5370ed01aa16cac53a4037a66226a87e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-03-20 22:10:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76e79fbecbb13ff1a7d3048dd17ad07ff288465b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-03-20 22:12:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aefc7440d58746c34731dd481c45ed09a0ed5d34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 00:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efe552758a634bb58376b51bae11d2c5\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e2e0e639177d43389ad2da8d131f5819 : SUCCESS in 1m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21e578a3ffd441a884c82bc800b35dff : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7bf99a4a40642b190873367b1f169d1 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6893a0f1b7904282ab4e30e88cd9fb08 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/095226ce7f6441f6810bc7b8a176b123 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23c512b92184577ab41031b04b7cb5f : SUCCESS in 7m 31s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/82fbda7a84f54fa5865b49a652b198f1 : SUCCESS in 2m 34s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/439566b6a0ef40d5b7f7def41f925a99 : SUCCESS in 5m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af54199786d943ff9133be6b85f6bded : SUCCESS in 3m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/427b1ad49ad8453086f8b38fd6b8eb1d : SUCCESS in 3m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e345fbe3aa74ea3956b47602046414b : SUCCESS in 7m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0707d6b7ad24409ba22c3f9d81a06b89 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4334c2619b1241ddaf4ab2e06667e241 : SUCCESS in 1h 31m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0db59378f7c84efca20821137d99585b : SUCCESS in 1h 46m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/81f41e2fb14b4e4eb62ba5d2bcb9f29a : SUCCESS in 1h 04m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d003c12502646df88b17d2c0cf925f7 : SUCCESS in 1h 47m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/131ec04e13294e5ab8760914078efa55 : SUCCESS in 1h 14m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c0b3e8f97a3d4852bbe4c79e2479d91d : SUCCESS in 1h 32m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4da6e5a1bf69420da710e9d814574eb3 : SUCCESS in 1h 03m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9dd73b9d596d46418c55b220df6c6e75 : SUCCESS in 41m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/75eb2017105545149c2d2105262d6120 : SUCCESS in 57m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13077af733064c2e863c11bbf6076db6 : SUCCESS in 34m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8b7429ae0ec43b2b3b433e01b7b5269 : SUCCESS in 2h 20m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/db5905319908466f96ea172c8c5ba9e8 : SUCCESS in 2h 10m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3e948ecae59d488bbed8229c1bc2eeea : SUCCESS in 2h 23m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a90b8bb52b54325a041518d55407809 : TIMED_OUT in 2h 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae83f804ca483c991cda1d69919a957822fe0a9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 00:49:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58110798870e058550375df381136ce1ab2698cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 02:42:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f623a23bb94a04adce21db41a404d6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0e3e1dfb5ca54528945062162fdfd53a : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae882bd1df9a4c44bd1ed310a94774d2 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80aa1a569dd948658e6377478055764c : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7ceea9d69c346e08d3cd4a963ee04c0 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792eb44ee09f4eb69041cf079f9894b2 : SUCCESS in 9m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8400f5dc542942039960342b6ba74f9c : SUCCESS in 2m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c820b0135e674954aeb53b46e33a9a22 : SUCCESS in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9028cca9442e49fea4477050e49e1e50 : SUCCESS in 36m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bcbb1f22c96741bda5bb5a21ce060371 : SUCCESS in 1h 51m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/07f99e1db28d47038401490aa97afd10 : SUCCESS in 1h 06m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/711271bfd6f346279ef6ea8e6d546202 : SUCCESS in 1h 49m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da796bc6fcdd41ffbebeab2c135d953e : SUCCESS in 1h 02m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/74cba3b3e5164a8fa3fefca951a33908 : SUCCESS in 23m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/deda340c6e9f4948a7480a2cc212f86e : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0c196393668f4988cdd82ad4e4e4a07338430209","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 02:42:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"672a726dc4661ac64a5f53dd7e229eb78e0019d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 02:43:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b44515967244c2b96894053108b71a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e13576332a428f8e5a3c304f0d4b8a : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b5aca4e2fc644e40eee93fea69c3dc687e0d616","revisions":{"0b5aca4e2fc644e40eee93fea69c3dc687e0d616":{"kind":"REWORK","_number":1,"created":"2025-03-20 22:10:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/71/945171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/945171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/945171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/945171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/945171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/945171/1"}}},"commit":{"parents":[{"commit":"f59da708b15d8801ab3217583a696250ffc51707","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f59da708b15d8801ab3217583a696250ffc51707"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-03-20 22:09:35.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-03-20 22:10:45.000000000","tz":0},"subject":"Bump octavia-lib requirement to 3.8.0","message":"Bump octavia-lib requirement to 3.8.0\n\nTo support the VIP security groups patch, we need to bump the minimum\nrequired version of octavia-lib to 3.8.0\n\nChange-Id: I7a3d7fb6ba1d5b32071b590a79fb039a465fe5fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b5aca4e2fc644e40eee93fea69c3dc687e0d616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b5aca4e2fc644e40eee93fea69c3dc687e0d616"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
