)]}'
{"id":"openstack%2Foctavia~945172","triplet_id":"openstack%2Foctavia~stable%2F2025.1~I7a3d7fb6ba1d5b32071b590a79fb039a465fe5fd","project":"openstack/octavia","branch":"stable/2025.1","hashtags":[],"change_id":"I7a3d7fb6ba1d5b32071b590a79fb039a465fe5fd","subject":"Bump octavia-lib requirement to 3.8.0","status":"MERGED","created":"2025-03-20 22:11:28.000000000","updated":"2025-03-20 22:45:37.000000000","submitted":"2025-03-20 22:44:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945172","cherry_pick_of_change":945171,"cherry_pick_of_patch_set":1,"meta_rev_id":"4a9e0e97259790f81d180da9b7005d5bb002e9cb","_number":945172,"virtual_id_number":945172,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-20 22:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-03-20 22:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-03-20 22:12:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 22:30:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a17236670b9ad434bb355fb6c6b9ca82a8ea63ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-03-20 22:11:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"24ec6ee8a714ff8b592dc88d049f5fdd915aa867","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-03-20 22:12:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6d6502cd963bece8cd4e6e84467a03769ea87e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:30:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7eaf8e2b4a84987b591660420d82f06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b5d8580e04045ba8b816d514e2dfd82 : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b513ed99964445b7fbb5e7f67e90f2 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de7c1544652a4e52b87651f6ee413ed3 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b2712431509434babd5f641fd5cec2a : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6c4229bc824aefb642676b78f68ec2 : SUCCESS in 13m 01s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b3eaf1c3e5f64f65b75d5056488435ef : SUCCESS in 5m 26s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a972e981b837423aa8a75bf9d741c73b : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c2748a894fe4a5eb722728738b22737 : SUCCESS in 3m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a56059cae4b48ff9ef6ea6b4508a6fd : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f38b4690ac3f4358bb208c791af2c82a : SUCCESS in 8m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/07785fec55164bcbbc6e29d72a3171f9 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a262b61f3546cf562534346bb19f54dccc8717af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:30:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6107e3a07489bb1264a1062f54b3592695e9bfba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:44:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/973155c1f4f1425e9d7edaf82d3d2772\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1158b38aea034cb7917f255ece189cd1 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a25cbaa5989476aa1f7c01495809e76 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0ffd0ca025243f794f0298e7c835e61 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ecefe666b347f7ba63bee916d12096 : SUCCESS in 11m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29cd6202e7b24e018c8b563994cc5888 : SUCCESS in 4m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/32ea270aae2041a2abcbbcf09e2b9bd3 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6c0c49d921913adbff587b397c00b71a3bdf9ac2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:44:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a9e0e97259790f81d180da9b7005d5bb002e9cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:45:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aa9a6e9b6c3474da62bc99bce1fabda\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2192d29dbae4492afc6de0310b0e62b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b236b25477ac471db4afcde0b7989eaab8fe408","revisions":{"3b236b25477ac471db4afcde0b7989eaab8fe408":{"kind":"REWORK","_number":1,"created":"2025-03-20 22:11:28.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/945172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/945172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/945172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/945172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/945172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/945172/1"}}},"commit":{"parents":[{"commit":"2374e72be0ad6e6500fc497c6c1bc2c71aaf0fa8","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2374e72be0ad6e6500fc497c6c1bc2c71aaf0fa8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-03-20 22:09:35.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-03-20 22:11:28.000000000","tz":0},"subject":"Bump octavia-lib requirement to 3.8.0","message":"Bump octavia-lib requirement to 3.8.0\n\nTo support the VIP security groups patch, we need to bump the minimum\nrequired version of octavia-lib to 3.8.0\n\nChange-Id: I7a3d7fb6ba1d5b32071b590a79fb039a465fe5fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b236b25477ac471db4afcde0b7989eaab8fe408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b236b25477ac471db4afcde0b7989eaab8fe408"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
