)]}'
{"id":"openstack%2Foctavia~945316","triplet_id":"openstack%2Foctavia~master~I3f6b8709a1fff9715961425c85b5f0d4b9359012","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-03-02 09:21:20.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-03-24 23:07:32.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-03-24 23:07:32.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-03-24 08:48:32.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I3f6b8709a1fff9715961425c85b5f0d4b9359012","subject":"Catch SubnetNotFound from network driver","status":"MERGED","created":"2025-03-24 13:55:40.000000000","updated":"2026-03-24 23:08:36.000000000","submitted":"2026-03-24 23:07:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945316","meta_rev_id":"27b75025208172a02469d12eda31b52bd39a26bd","_number":945316,"virtual_id_number":945316,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-24 23:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-03-02 09:21:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-24 08:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2026-03-02 23:10:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-24 08:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 17:02:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 12:41:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-07-11 21:19:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-03-01 00:42:04.000000000","updated_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-03-02 23:10:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"eda02de75510642884d9e8a55b09e6919204a620","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-03-24 13:55:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a7adc97538562b8f4bea4bcff1f324586151b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 17:02:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d051323d5a84be79aeac8ca7e334e8d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/451b552f4a3a402b880d8e32b67331a9 : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fcc4241715c46b1aa18ba59b16b3941 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/818d99eb58454151932d4982f219d07e : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bee414e2fff2472b9de14b06cd608079 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43ae036550b4b9aa2c665cb3ade8405 : SUCCESS in 13m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6456f1a82db047ff8119e76dc004a9d4 : SUCCESS in 6m 23s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1c438de4df7348c6a6f957e944459ed3 : SUCCESS in 11m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f31ff81b0fa54af5a6486db41106f760 : SUCCESS in 3m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4db8dd406d4b425caa3598b8b8942971 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a333da875354fae9d8d32e0116a07ee : SUCCESS in 8m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a3f6e8e4f6b44faf99533eec60ba8f44 : SUCCESS in 7m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/171333f4333141c4acf9880e16cef31d : SUCCESS in 47m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b770e8311bcf402cbb162b285c25e718 : SUCCESS in 1h 46m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1c7d9a00ac7c40f09fcd349a3d24225c : SUCCESS in 59m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b7f09e916574086b4db09cba3f7f5fc : SUCCESS in 1h 45m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10dec88cb4be481f93e6b4ed732066dc : SUCCESS in 1h 33m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1ce62f721c4d4b85b0502808f0f79fbe : SUCCESS in 1h 52m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7d1abc9eae084e6bba5f467dc554104a : SUCCESS in 1h 05m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/873c3dcebe7e401b9b0db50e29e0befa : SUCCESS in 28m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78cb2d20ef4d41dd9ab99d57b3bf630d : SUCCESS in 45m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c32db9f6bf944233b0ff3eecabd42af7 : SUCCESS in 39m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e72ed92af904e8d99b3f18d67a95cc9 : SUCCESS in 26m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/937a44651fab4dc08cdfa5881992d2f2 : TIMED_OUT in 2h 34m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a937eb163e34e27adcaf9ff1f9ee487 : TIMED_OUT in 3h 04m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/80e4db182915437a924a1aa0eade1180 : TIMED_OUT in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e15999d32414aa29a5e970eaec1bdbb : TIMED_OUT in 2h 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"508f399fae5d32430f9c613567e31755346605b8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-07-10 12:41:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f96c0782db1400e0f18270618910afc11b3dbadb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-11 21:19:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f92528c9541714a6d230c5fa017df1f8ebe4aac8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-19 07:24:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f1bfb4fe7218c47274b742355b5859c329e4800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-27 14:30:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"122905e5ae2d271292c264fa74a51cebf69be6a4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-27 14:32:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"385f8af8169354cf2aa37844396bf9366a857858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:53:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2207328c82904ef6bb0b9df78235d80d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b31e52d054dd4c50ac45bb480ef8ff8e : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ace44241e1143a890c22da293a2663c : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5c0c3e0c1ae4c7d9ef67806954c9fd5 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/529c4c71e21647928bd6ee7a66d91974 : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4202397e13904a12ab3dcbacf9ef3002 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc94093b2474be0bf8d993024f9835b : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08c37525b5f14da597e48838554c09e1 : SUCCESS in 6m 22s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/befe8eda34ff49e0aa53bef28209d58c : SUCCESS in 4m 57s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/59abfd34168d43a49e1bc743c2126a26 : SUCCESS in 9m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/56f723f69c2d403da0e3b4bc1e1a0f07 : SUCCESS in 2m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a58ace636b254b3d89de38c3bd805544 : SUCCESS in 7m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8fa39bcdddaf4bc08db90c658a7c3e27 : SUCCESS in 10m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7eca86a5ca404d00b14d5119402f0401 : SUCCESS in 1h 52m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8572ce3492474d53ab496dfd081514b1 : SUCCESS in 1h 36m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cde962a1baff4e0caf5375e83e502266 : SUCCESS in 1h 27m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/78849326b2214aae81d18ca22ae0e37a : SUCCESS in 1h 39m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5a16a52a54e64cae80fd9ce40b20c3cb : SUCCESS in 1h 34m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5bbf8feaeb584b908ee9577f14060b13 : SUCCESS in 50m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7a5913bbee3a406abbd6c0a3371ca653 : SUCCESS in 41m 09s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c9b0183f82dd4f1bb180a198f8a194a8 : SUCCESS in 42m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1a7e0e9d26d4b33a613e5269ea6e27c : SUCCESS in 32m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3782ddacd2d749b481a3c419dd5b5180 : FAILURE in 51m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb45fd3c3176441c90fcecbbbf85406b : SUCCESS in 3h 00m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82d5eff9d14f47a59d1c54496d274000 : SUCCESS in 1h 47m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/831f594ce47d4ea6ba0cff2b5dce3f19 : SUCCESS in 1h 26m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e914c9946d864dd7a3504767c58f8a80 : SUCCESS in 1h 15m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0aecdec17d339a071814ea51443f9ca3e2483590","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-01 00:42:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b052d47d57ccb6b2b0375fb3d583b4e2a9bbb739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-03-02 08:48:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"efdd1e8fb8c0ab9fc533a7dc953594a6d5500014","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-03-02 08:50:30.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ef8eb27b6b62436ba6cec776b311054b501141be","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-02 09:21:20.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"877b0135d840bb6b2be7416c3200050da3f7bee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 11:31:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a19e9e9dc30474892c2055c1ca3df03\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f91d89218a18494a85306f9e80e83a29 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a34d7b7fd294db8b7a5c5d6747b4c2a : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d1f46fd5c794d429c60de5a9eb201ed : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47efa6dd7c3145b2892b4e73d537daca : SUCCESS in 4m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8c9a9f75bbb4fd398bfc8bf073c73e6 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/addf3d85231c45f095f14061a27d18ea : SUCCESS in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/643e74286d2f4371b2f14d25f3764453 : SUCCESS in 6m 41s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/582ee28af227438ca2f1434bb7635d33 : SUCCESS in 4m 41s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/e95e9b048a764c7799435304e6e891b1 : SUCCESS in 6m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/53e8e418ce124782b9d6732a6781ae16 : SUCCESS in 1m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e4d571b1910487f8a164816d7f6efcb : SUCCESS in 8m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/30e57ef26bac491ba817fdbb453b0407 : SUCCESS in 4m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85f0d4a4226542098e19b8216e5fffb5 : SUCCESS in 1h 16m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/20b470d7b4b54c42bfea94e518527c72 : SUCCESS in 1h 44m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/189507a46ba94d8383aa8517b7841313 : SUCCESS in 1h 12m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/22f41f32d2784e8890afbd0aa8df7554 : SUCCESS in 1h 43m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/26e3f6915ea948c1a095e45371d55607 : TIMED_OUT in 2h 34m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d0d020f41b5d4f43b1988d461750633c : SUCCESS in 44m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9a720281ee9f48019b5368ea60d63e29 : SUCCESS in 48m 08s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/152baf62463448af90d06cfd919020e7 : SUCCESS in 42m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/038e7ace8e3a4bc2982edc8ebd1f00bd : SUCCESS in 31m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9213f736c6ee48aebf6458b20906c258 : SUCCESS in 38m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d06fab8a9d4c44a6ba030d3fa90ad5e7 : SUCCESS in 2h 39m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/919c070378074414b5390bb3d2388847 : SUCCESS in 1h 35m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/2ac07699027a4b538b8bdd2b5aaf040d : SUCCESS in 1h 32m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1f4ea57273524426a9c160054ae7714c : SUCCESS in 1h 02m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dd1cbe9aaf80ce1029b645d5146cc22abd936cdf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-03-02 16:06:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f5441508ee3f1b84741934694882be64e9343a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 19:44:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519e62f6f61640ae8a4bbf8f27bc6b86\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8efa01941c423ebb4c904ae0076054 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102fe4a9272049618f9460736fced860 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd4f02b7e7d84431963cacbf2d098f35 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80569db40d9543629604a65b6c24e063 : SUCCESS in 5m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a593e38e195d4956a439a8cb85167582 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5dd2f8c2164803a3db9f484368f632 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80073a10a12f4139bd5cbb24c47f22cf : SUCCESS in 5m 37s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/cdbe030b687f4bc3aa79a03d913ae506 : SUCCESS in 5m 20s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/eebd9b662c404233b94fb255443f1607 : SUCCESS in 4m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6564e37e54674987a966e783e0fec630 : SUCCESS in 3m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f9ef24963524e6da2b5d7f772b3e195 : SUCCESS in 6m 40s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fea0d249be1d455c9f5a433a9903707f : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a68b78f955c34d0a9409d07ed48cee3c : SUCCESS in 1h 25m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c6f36f3d05ee4d48816322dd46caaf2f : SUCCESS in 1h 44m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/89d7bc2be4db4581a94cdf9faf512b1e : SUCCESS in 1h 25m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9497906f281046d5b6459277696fb073 : SUCCESS in 1h 41m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/48b90140d5364ebb99648397328fc74c : SUCCESS in 1h 14m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed8d453d70e94ee99924acb5619ce55d : SUCCESS in 39m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7a958d798574b5ba61e78d50aef94b3 : SUCCESS in 44m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1165a0a458e4434184fdb2aa0dc386fd : SUCCESS in 50m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67f793f09c6d4416bb653d5b9430a360 : FAILURE in 55m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/719faf759fcb4a888aab33b572dbdcbb : SUCCESS in 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f6ceec1da9a4b0492a7678a90fed795 : TIMED_OUT in 3h 04m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c43d86ca0c02496da583c96f4542169e : SUCCESS in 1h 36m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/7302c049cbd048719c4b17a3ebfe64a6 : SUCCESS in 1h 29m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/27e24b97a24b4121844494c5db47be0f : SUCCESS in 1h 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"43fcaaa0463808e0517aefd6c61cb193cccfbc2a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-02 23:10:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f12d76c23abb2dde58ddea8014632a943d64966c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-03-24 08:48:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4be6eaa8cf4ce363c72999d59df1ac1027b15a37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 08:49:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e7b7d6cd297a7d2b2f68c56535122e88136dc92d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 10:28:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1364266526634f7eb04236d0debc6791\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86e31aad3fdd482896e5cb91f3355731 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb40bafcaf644bb3922951c4b6ff35cb : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51ad29a17e424d5ca18377c7a62a4310 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3778d0679088488d8001911a79f908d8 : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a038747254145048d223837db0ce4bf : SUCCESS in 4m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/665bc68776554283a8df0095a98f7672 : SUCCESS in 3m 51s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e914f3415b08448689f7762050397fe1 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1f6cc4424a842259d96d54b78335cf7 : SUCCESS in 58m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a3ca1ae3e5304a53b9120abd6dc4b79e : CANCELED in 1h 11m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ac727bedfda0465b8026023400ac16fd : FAILURE in 1h 11m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/402523163b2149d0b00e6056711f3ead : SUCCESS in 42m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dec340b9ca7449e2b75ed420637fc113 : SUCCESS in 41m 24s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5374ab3cafe34d0b91fd95547e858a59 : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":3},{"id":"37100b4a8582dab90c7345482da4be8551a23cbd","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-03-24 10:51:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"abe8968d986fa1172c3670b21aa61625559d432a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-24 10:51:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"628ad77d2a21495a6b8c47056a1e877c4d0fe900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 14:02:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aeb2a801751f45159120dc350bcfba83\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dfeb8be06734ee1a2fb880ebf6e6667 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d711da4c4249acb17880705e181a55 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d3deac558a14ce1881ed298273a348f : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68125bda56554c26b526a58328d7c7eb : SUCCESS in 3m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/534e23b4809e4750ba971c0082cf07b4 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb7ccb641ef4f6d855f4c915ab13f2c : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ada214217b2745638162dd42c39ce2e8 : SUCCESS in 5m 23s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/8e41bde8f5c24169b552771ed017d417 : SUCCESS in 3m 44s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/0d80d16df2f84313b7fa230f79ae8440 : SUCCESS in 7m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e8c8a023f0954484840f13f310daade9 : SUCCESS in 3m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d65d92137b694337a520d366efdbfd6c : SUCCESS in 5m 39s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ff6f2d38232041bf81360eae51834ed9 : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40d7180f7ec14f69bc0bff48c448b832 : SUCCESS in 44m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ceb6d480b1f24cada61d2819722cb91b : SUCCESS in 1h 34m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/aad98f5c38d646b8961637e376e9be9f : SUCCESS in 1h 17m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b7f4e826f81742558fab97c4a2d3629b : SUCCESS in 1h 43m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/59834b20744d49b0948a6134a97afc80 : FAILURE in 2h 07m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c607f25ffaa4d8e8bc0485faa5f0edf : SUCCESS in 38m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a6ca2f2cc6d84b51b5e0bf608637d409 : SUCCESS in 48m 17s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b01735f5ab6b4050b53b89dd3804bb7a : SUCCESS in 42m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ddfef185fd44ed0aa5413a5b40d4bee : FAILURE in 31m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/76c1ad8908f34730b6680c3f64a6a372 : FAILURE in 37m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/69ad80bf32874b2ea5be85e10272d3e2 : TIMED_OUT in 3h 03m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/07a59a7f60b3405c9b4e2063f2f78bc0 : POST_FAILURE in 1h 43m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/3d6dfb5e7a3549c3bf484b7b4c9b910e : SUCCESS in 1h 32m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a5efcd4ce4c5416088a454cfa5ae3850 : SUCCESS in 1h 04m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8921316c9e6645faf4ceeb6b1786dc3863fd9a2d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-24 18:23:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e45fa4d74fedd3e36ee2366f831b58e6010e76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 21:30:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3979f4dda93c495f8828a52669039927\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/003acfa42c324edaa727df82383ea070 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290a03c21b804a52a230baf5f40341ed : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a466a815d49048a0983c4ba871245b6f : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c17c8b8310084e658d3ab70e42afb8ac : SUCCESS in 5m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82a5be28337e43329f4073e3045cd9ba : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a168e5746e741678d78b82fe08b556c : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1cf50cca2f34c16bb2f8a6f0f1204d4 : SUCCESS in 5m 04s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/57d3b3a4bbe54163887b93a474e5b63b : SUCCESS in 5m 30s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/639c2a96518d46aebac6c7c122a6fcf4 : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e28a98961c7438ca1d5b9c9453f9494 : SUCCESS in 3m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2895695e4835441d8a2797d1fa6ae078 : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8040098fea5d4f29bd42f42924d77fb8 : SUCCESS in 4m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae2469ea45b843a4b45e49078a04d8d3 : SUCCESS in 1h 47m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c273fd13327a4956beadf5bd4ce53cc7 : SUCCESS in 1h 37m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/941339092db64c389300069078e4925d : SUCCESS in 1h 28m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/22ce4ea2381f4d9e9529e64f52ec7edc : SUCCESS in 1h 51m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/530d30afb0ac411f923ee50090079bb2 : SUCCESS in 1h 22m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9f18af28a9748648a91b917984908b5 : SUCCESS in 48m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10e699dc45be4daaa969b07a73198acd : SUCCESS in 47m 47s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9295f55069ac4fd5a16406cfad12e26d : SUCCESS in 53m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7acb6c89fbb4b3687b30e8bad4532ec : FAILURE in 53m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/318e3ae62c9f40b784e75b8eeb5e1c48 : SUCCESS in 38m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b33a98d594a64d828a4efccf1f4ca6f4 : TIMED_OUT in 3h 03m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a20ac5bfdd1448769a67c6b88bcfb7a9 : SUCCESS in 1h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/09a55ed3e6924f24b9c0d742d853c3ad : SUCCESS in 1h 36m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f52f8278e4384918aae6a08f81953916 : SUCCESS in 1h 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3b0ee83de88c583f58bb47e8e3451db0b6fdf4d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 21:30:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5bc0783fda1b22bf57b6ab3a2ef572b189d901e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 23:07:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67af876551a54ec9b56ce3a3f90842af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6913d2a37206494bb0c243c75e000ca7 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b035e95c93174d95ae14957566bd87dc : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d25267f75bb4a21834be5ca9fbe1c8d : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318f6050b45843f888abbd11b2f61abf : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eaf9b0d941044f8871304f0deb8582b : SUCCESS in 6m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/483a453d3e784403b040e5c6333030a9 : SUCCESS in 3m 09s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/245903f6841e47e7a40787419d291eab : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/badbdcd30c864319ad6e4f5766fbad7a : SUCCESS in 1h 31m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8e176b027569467ca579ce6615974c32 : SUCCESS in 1h 35m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9109569c69774937a3aff33f8780caab : SUCCESS in 1h 27m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b49d098173cd4123bc7d8b0e2d0a951b : SUCCESS in 36m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5bf8696066334d4fbaee89a89de8c06e : SUCCESS in 47m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/15caef097de3433eb39cae6247d8683e : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":3},{"id":"2414a95be810d91874d320ee9755c3eee753bdba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 23:07:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"27b75025208172a02469d12eda31b52bd39a26bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 23:08:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4af1154e75747cd90a620d14492add7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8882e9a1fe184eac8ffd6bd4b13b94bf : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7e786720ae94c868bac9b16affe9f23 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9052b4e7203af8fdd8e2116599f2669c0496ba97","revisions":{"ab327252b674dab5c1c628cbcefc488a48fa3b46":{"kind":"REWORK","_number":1,"created":"2025-03-24 13:55:40.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/945316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/945316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/945316/1"}}},"commit":{"parents":[{"commit":"0b5aca4e2fc644e40eee93fea69c3dc687e0d616","subject":"Bump octavia-lib requirement to 3.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0b5aca4e2fc644e40eee93fea69c3dc687e0d616"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 13:54:49.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 13:54:49.000000000","tz":60},"subject":"Catch SubnetNotFound from network driver","message":"Catch SubnetNotFound from network driver\n\nWe should catch the SubnetNotFound exception from\nthe network driver and raise a ValidationException\nso that we don\u0027t return a HTTP 500 Internal Server\nError to the user.\n\nChange-Id: I3f6b8709a1fff9715961425c85b5f0d4b9359012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab327252b674dab5c1c628cbcefc488a48fa3b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab327252b674dab5c1c628cbcefc488a48fa3b46"}]},"branch":"refs/heads/master"},"77224379e12df4347ca0336c592cf59bd972a9ef":{"kind":"REWORK","_number":2,"created":"2026-02-27 14:30:43.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/945316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/945316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/945316/2"}}},"commit":{"parents":[{"commit":"cacd6b56a69954744887a373c527c4e06a287b74","subject":"Test LB API giving HTTP 500 when get_subnet fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cacd6b56a69954744887a373c527c4e06a287b74"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 13:54:49.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-02-27 14:27:32.000000000","tz":60},"subject":"Catch SubnetNotFound from network driver","message":"Catch SubnetNotFound from network driver\n\nWe should catch the SubnetNotFound exception from\nthe network driver and raise a ValidationException\nso that we don\u0027t return a HTTP 500 Internal Server\nError to the user.\n\nChange-Id: I3f6b8709a1fff9715961425c85b5f0d4b9359012\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77224379e12df4347ca0336c592cf59bd972a9ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77224379e12df4347ca0336c592cf59bd972a9ef"}]},"branch":"refs/heads/master"},"9052b4e7203af8fdd8e2116599f2669c0496ba97":{"kind":"REWORK","_number":3,"created":"2026-03-02 08:48:55.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/945316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/945316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/945316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/945316/3"}}},"commit":{"parents":[{"commit":"cacd6b56a69954744887a373c527c4e06a287b74","subject":"Test LB API giving HTTP 500 when get_subnet fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cacd6b56a69954744887a373c527c4e06a287b74"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-03-24 13:54:49.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-03-02 08:48:24.000000000","tz":60},"subject":"Catch SubnetNotFound from network driver","message":"Catch SubnetNotFound from network driver\n\nWe should catch the SubnetNotFound exception from\nthe network driver and raise a ValidationException\nso that we don\u0027t return a HTTP 500 Internal Server\nError to the user.\n\nChange-Id: I3f6b8709a1fff9715961425c85b5f0d4b9359012\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9052b4e7203af8fdd8e2116599f2669c0496ba97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9052b4e7203af8fdd8e2116599f2669c0496ba97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
