)]}'
{"id":"openstack%2Foctavia~946866","triplet_id":"openstack%2Foctavia~master~I920676dd4672b9bdcba8c771b8d7470fa88fe14d","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-09-10 21:26:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I920676dd4672b9bdcba8c771b8d7470fa88fe14d","subject":"Fixed missing port_id in additional_vips API","status":"MERGED","created":"2025-04-10 06:07:51.000000000","updated":"2025-09-10 21:27:52.000000000","submitted":"2025-09-10 21:26:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946866","meta_rev_id":"93e8adcfa2b3155dd97673539b91c803c9ed56b1","_number":946866,"virtual_id_number":946866,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-10 21:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2025-08-22 05:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-31 15:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-10 19:32:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-10 08:50:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-31 15:35:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-08-22 05:49:41.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"1704b8b7bf31263d7a14e6a0979f920f3b888c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-10 06:07:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d88cdd5e046f221009215e676820f264f38f6d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 08:50:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/419a873323f449d3ba4ec124c7882f6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51a53038a8a441cbb1f1c812cd6dc004 : FAILURE in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28656db49b048d6822fb42b0acd1ee3 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2791dfda9d6b4aa89fb0577667858fe1 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19dd6a90052b41e5b36ebce279772068 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/119355c7b4324d3c97ed699be9a9f969 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa20ba245c1b4315b470f5c913a92841 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7035e5a4e43240fabf1af284348eaecb : SUCCESS in 3m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/aaf44cb387584b388065c623eabd2049 : SUCCESS in 2m 34s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/21ce9e784096441b9a2c139190a631c3 : FAILURE in 3m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/08ecc57d8a0b46aab2eba0e71da50eab : SUCCESS in 2m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6fb6252fd9e44a7a96c52d6f878a0c74 : FAILURE in 3m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b7473a90a2041efa60cadc0e4fc419e : FAILURE in 3m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5fad09ee8ac74aefaaf57264d91afb52 : FAILURE in 3m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99766edf71b941cb9d5494be63e32d71 : SUCCESS in 33m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b3353d82245546489abd867d5fbdd726 : SUCCESS in 1h 22m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1534e8e7e2984bf5bee273c210173103 : SUCCESS in 47m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e2ca208b47b4c14b76f5075a5706c67 : SUCCESS in 1h 59m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aae77b19449407fa3217cac98264946 : SUCCESS in 1h 12m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0bd04bea66c543a08815644f21670247 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8306d761b51a42349d7b3885b5fcf31a : SUCCESS in 1h 08m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d2c356adfbc4b589d35746245951852 : SUCCESS in 22m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db75c04174294cd3935bd1c1ec1c7334 : SUCCESS in 1h 05m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c4eddd9ba2f44440b85415bdfe5a4fbf : SUCCESS in 57m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/013eaa2bed2849268e2833813f75314e : SUCCESS in 36m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ffcf3147db54bd1884039bbc165c861 : TIMED_OUT in 2h 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b19eb22beed84d7083e6f5f8c789c88c : SUCCESS in 2h 25m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c25854da6bf4407cbd3abadaa7488b68 : SUCCESS in 1h 52m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/246a7b0842524b0195b84f0d5efa7e0a : TIMED_OUT in 2h 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e4ace41db5e8806726c4e2dbb8a6ba20d9ddf69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-10 03:19:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"17963bacc4d9cc2983505d96e822d790abbbb915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-10 05:56:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d83fbce07aa4a729c917059cc8f1732\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd5b65e6aad7486187e24c4388186920 : FAILURE in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e9db1a64fb4dc7aea23f7e72ffe082 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bcb92aad8ae49ccb8ef02ad6e51210e : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ecf1e4093894e22b20207402aeea08c : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e79a826a1a054267a124266d47dd481c : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab756f175f54109ab05266b4d96949d : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39a2e6c77e8f4e63b816351c0c86ee82 : SUCCESS in 4m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/17da9ca81405473c9804c51936523901 : SUCCESS in 6m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/38c1a26122cb4e48a4654e05963f27c2 : FAILURE in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d01be9ff2bfb436e87b0b70639826d3b : SUCCESS in 4m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bfe3e1a49c054007aebf4e693388f962 : FAILURE in 7m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d2ff07563e234863a40cfb5b6f4b641f : FAILURE in 5m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d25cfbdbdfd4ac29f9f1839b17309ff : SUCCESS in 1h 25m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a512352666434d3785777773a53865cf : SUCCESS in 2h 23m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a23cf88d1f774bdcb267b8373b1da5a8 : SUCCESS in 54m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c1985c96e6245e9bf80d797a90a5f6d : SUCCESS in 1h 53m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea9aaf6aa9c14461b48acb9a89d7b71f : SUCCESS in 1h 18m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/220e7f1a455644338356eaf1edfa8164 : SUCCESS in 1h 54m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f9a151edf12843d8bc03f1b55ac4fb96 : SUCCESS in 44m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b7aa94e65514c129ca7aaa3475bb5cb : SUCCESS in 42m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/232f4d1f2a5341b79cbc69c4a882cf90 : FAILURE in 21m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d0abb7ecf4747ae96ed239256047896 : SUCCESS in 41m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d00317955f5346d9b7e9540bf9d3caac : SUCCESS in 2h 06m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4f52433445b44edbaf57f9bbe3993675 : SUCCESS in 2h 34m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1cf0d050d748410c8b8cb1934e5c09c1 : SUCCESS in 1h 45m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f735b1d82d48615bc6d3f3bdb345025e9c046e7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-07-10 11:46:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61f1841e21de7027614a459d43a4efdcea29aa3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:45:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00c5c4fc420244af83b4d91ee2aa99fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7669e2ab4ece4237aa890147df7c26b6 : FAILURE in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc4b950ae3204e419ea431e516064ca6 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab772a74771c4068955a98bf8482ef92 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bef66d1ba0be49d186baeb6ba187fd26 : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3bcfd77510a443c85a2fc7fb8a9d91d : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7db22e417254c8a8efdf2c208271875 : SUCCESS in 15m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32b917317c2f4a52ad0689034749cf64 : SUCCESS in 5m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a2569a45af3f4533a2896b64437b32f0 : SUCCESS in 6m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1f1b4a7edf1342b0ae5b97b7b03918e2 : FAILURE in 9m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/edbd065ae64e4e00b7b215ab9c45efb6 : SUCCESS in 5m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/510e6177f050445cb6a3ceef226f6e32 : FAILURE in 10m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e9e70e952e624629803dc7fbba32ca90 : FAILURE in 9m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d727837f0f824093a9e570963f5daff4 : SUCCESS in 1h 28m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5e6d7050a184430e972bc1052d355843 : SUCCESS in 1h 53m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3741438e56544b92b4c48ba70ca5c032 : SUCCESS in 1h 07m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a77225f72604b82adce13a1c0d202d2 : SUCCESS in 1h 49m 34s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b0169b5d1554d1aae4935bec8578cfd : SUCCESS in 1h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/009bde42ee3c4941b4ccb8271581a39c : SUCCESS in 2h 01m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c8e49a5e5d164d29819c5179404ec356 : SUCCESS in 1h 09m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e26f59b1d0d241579851d1ecaef9d59b : SUCCESS in 39m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ac56a7564ab41f7b89cb01ab2adee0f : SUCCESS in 47m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0fc55f13d434047b34bb15ebbade216 : SUCCESS in 39m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/79243b5b044f478f952145a79e6f379f : SUCCESS in 2h 24m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2488a046a10b4d83ad5087ccd70bd4e8 : SUCCESS in 2h 57m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2a8129b860ac45d0957b9984634f81b9 : TIMED_OUT in 2h 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"568e283e96b9b9b6a8e8f8792792068004bdcdcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-07-17 06:44:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d4ace629920b8ea2f544d613e480c5c3253e28fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 09:25:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7237604d975c4243b7eacd02e21142e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9021c3f0de234bc38107d210ab20e04f : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2adabfe7d4ab4002a729cf39d155e794 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1665f01b7f4446896f3cb38f4eef088 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9289aeaa15ec47f8b9a70fe5006203b4 : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e724bc0038d4bff93e2f168f89e11d7 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1031129ec064dbca454b4ac952b50a6 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/458c0f7c7eb14977837e056a00be51a3 : SUCCESS in 2m 43s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/216486b52f994244949f8dc93b7c8122 : SUCCESS in 8m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f587ba0b28274a42a1b10932d9da5505 : SUCCESS in 7m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/37e86c38f30e4d3f85fc25a0ab4a2286 : SUCCESS in 4m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4bf32a5ea2342ea802a1edd7b354d96 : SUCCESS in 5m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6a85dac072134171b8b682849be26d8e : SUCCESS in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/274e1ec13cc74873aae8f427559bec72 : SUCCESS in 59m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fd7d72cea21d42ef9929c13422a9f2b3 : FAILURE in 1h 51m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/06d9449f981f4bf7bb06788926af98a4 : SUCCESS in 1h 02m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e32bb02e75fb4f158d00c2726c1815e8 : SUCCESS in 1h 47m 32s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/850e506268a84f098323518c6e5afd8d : SUCCESS in 58m 16s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c7da7d901b1149d5929b6ea2edc37536 : FAILURE in 1h 55m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7aa5eaf06a4a460fa7c470a19afd6af8 : SUCCESS in 1h 09m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c5d3c8233cd4a8299026c991ae51b22 : SUCCESS in 35m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/436883cdc8b8435ebf47298a95303508 : SUCCESS in 42m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06554fa1eacc436eb70917ea94ee35ad : SUCCESS in 34m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ba116e2af244c628ccc3ea46de0a1c5 : TIMED_OUT in 2h 33m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/99d67b1abb814d33b1aa14c034ccc33d : SUCCESS in 2h 39m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f181df9d30bc474192973990e0af4f88 : SUCCESS in 2h 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0cdb0460f52dc299fec0c78cd4e49e0b1791c6ca","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-07-30 15:46:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"afba10f2ce58ebcfc05119f34835fc354fd8e904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 18:50:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99e47a80fb5546169136f126dfde1f2b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/322724bfea0a471cb64f893523c211cf : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b5f58b065341b8afab3d27986f45e8 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a96d5c9023c747d196af97b66ea74013 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5c22093d8cd49ba99e6473d47631484 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b587bc623089488dbfd562db9ff727f4 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e376c4fa5f84898a23ab220f84be170 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db834eb2f5b24c96baf096b9e2f6607c : SUCCESS in 3m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d55809749b044dc9a4cd3f85cc312cb1 : SUCCESS in 2m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/be3d290d027942c38f64a19b87ca58cd : SUCCESS in 3m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1b9fc8b3295946fa91fa20f492a07d35 : SUCCESS in 2m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8d712fcb0dcc482180449b278b9df158 : SUCCESS in 4m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f5e43613d63c436c8bfb7026bd34e46b : SUCCESS in 3m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ebcff7d14119463e894c45aa1e177926 : SUCCESS in 49m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c1e7b93d1a544a52bfc7332f8d22cd37 : SUCCESS in 1h 58m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d098a219eb8c4653be0b7ab5f1cd543d : SUCCESS in 1h 02m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4316f157d1584bc69ac6c0434abcb5f4 : SUCCESS in 1h 46m 33s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7329114fd15467b849e6099302b5135 : SUCCESS in 1h 11m 12s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/50b8a7dac4f54c29b3562ab7659cb67d : SUCCESS in 2h 02m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f8eeb5ab30ad4765ad0b75203771c873 : SUCCESS in 1h 19m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/699168ff74284ae1a444a2199191eeb6 : SUCCESS in 41m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e77b8e5c67474168990d284ed91c02a5 : SUCCESS in 50m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47fca3caa4694eea8b370148e0cb581b : SUCCESS in 34m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2d106574cf774a649ca8fc550b3cf739 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6833875b39494ac4bc5a8ad6323c6862 : TIMED_OUT in 3h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d26b55a2f094decb5c1a1f00953d389 : TIMED_OUT in 2h 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c633312c12a9e999f93106567cab76ae9198fdde","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-31 15:35:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27f5e48d699990ac08ed55917afa7ffbf9132d1a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2025-08-22 05:49:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2242d006b0676f9c40a3402e6e912745ed22da56","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-09-10 19:32:08.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8ceed00775761c0b75f79da833d703167181ede3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 19:32:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5734ac6baca89b672ea7d170e088d7fd5eb3f684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 21:26:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82754eb48514022878eff8ed4145c44\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d87507f213443c79929c68d42546009 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d576b1f3260049a18c5c71a70ff0195f : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46c8ead18e7e4bf18b71fc4a3d93ee1e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/603aebc023a94c059e053c46c857c6f0 : SUCCESS in 13m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d52a149df37f4094b6b1a79d82600d22 : SUCCESS in 6m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/788f34b0c07e4fca90cfbc22a285ca2d : SUCCESS in 2m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bb25d32ed0944cafa79512c44f0f8508 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19987eba95cb4ad38bdba062dc3b8df1 : SUCCESS in 1h 09m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5e51ca9b68d842ed8c275d2f6ea7033b : SUCCESS in 1h 53m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d974453c7534c8786e08b5377cf07b7 : SUCCESS in 1h 03m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19a11ae2f4c4420aaa0a1eaf373e6727 : SUCCESS in 37m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16418dc668894d39b4b3fc24aa987549 : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":3},{"id":"2dae3cd11b61fb54f4a483170ef9d9967c30b275","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 21:26:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"93e8adcfa2b3155dd97673539b91c803c9ed56b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 21:27:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dd74277c90f416fb01bc396a4ebc412\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1a364c90044cad903f950c9fce8fb4 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e8a572498474ca2bbc604cfb3131995 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bfa1abb2913695ab7ae17f21f8dbcf89f5f94854","revisions":{"4a87d520a35dc6ed1e7e86fffff73d912fe4b576":{"kind":"REWORK","_number":1,"created":"2025-04-10 06:07:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/66/946866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/946866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/946866/1"}}},"commit":{"parents":[{"commit":"3b4144b65b432bd9bc72b192c9813754a8fd9f36","subject":"Merge \"Remove duplicate hacking checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4144b65b432bd9bc72b192c9813754a8fd9f36"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 06:04:55.000000000","tz":120},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a87d520a35dc6ed1e7e86fffff73d912fe4b576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a87d520a35dc6ed1e7e86fffff73d912fe4b576"}]},"branch":"refs/heads/master"},"5672d453b2bd2f21b4465e0ac1fd3d5a7816f36d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-10 03:19:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/66/946866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/946866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/946866/2"}}},"commit":{"parents":[{"commit":"599873b01ffc0adadf94ecbdb791351275ef10f8","subject":"Re-enabling voting for octavia-v2-dsvm-tls-barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599873b01ffc0adadf94ecbdb791351275ef10f8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-05-10 03:19:21.000000000","tz":0},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5672d453b2bd2f21b4465e0ac1fd3d5a7816f36d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5672d453b2bd2f21b4465e0ac1fd3d5a7816f36d"}]},"branch":"refs/heads/master","description":"Rebase"},"bfa1abb2913695ab7ae17f21f8dbcf89f5f94854":{"kind":"REWORK","_number":3,"created":"2025-07-17 06:44:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/66/946866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/946866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/946866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/946866/3"}}},"commit":{"parents":[{"commit":"cd4fa250d14f9bf99e769c89eb4153bd58768316","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd4fa250d14f9bf99e769c89eb4153bd58768316"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-10 05:56:43.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-07-17 06:44:17.000000000","tz":120},"subject":"Fixed missing port_id in additional_vips API","message":"Fixed missing port_id in additional_vips API\n\nthe additional_vip object includes a port_id field but this field is not\nreturned by the API.\n\nChange-Id: I920676dd4672b9bdcba8c771b8d7470fa88fe14d\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfa1abb2913695ab7ae17f21f8dbcf89f5f94854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfa1abb2913695ab7ae17f21f8dbcf89f5f94854"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
