)]}'
{"id":"openstack%2Foctavia~947550","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I8aa5f128011068321c88f66b1fa986f2ead19034","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2025-04-22 08:48:23.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-04-22 07:06:11.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-04-18 17:53:18.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-04-22 08:48:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8aa5f128011068321c88f66b1fa986f2ead19034","subject":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB","status":"MERGED","created":"2025-04-17 04:54:26.000000000","updated":"2025-04-22 08:49:32.000000000","submitted":"2025-04-22 08:48:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947550","cherry_pick_of_change":947549,"cherry_pick_of_patch_set":1,"meta_rev_id":"e6ed5aba0d0016be66e7b47ddb2cd5c7f0b9ba3d","_number":947550,"virtual_id_number":947550,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 08:48:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-04-18 17:53:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-04-22 07:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-04-22 07:06:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-17 04:57:47.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2025-04-17 04:57:47.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-04-17 04:57:47.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-04-17 07:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a08227a2f9ae64fc8f192b922a8580e181eda61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-04-17 04:54:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"6d76730f68009f77f62c310c9bd4aead1ec9ea6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 07:28:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5858c93b63c42778dc8692a694aa657\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/269e694658394931b8e3315ebfccd716 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c135decf5314057976ea45e30856dd4 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb7e074f2e294911956b72da066b54a4 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c19aa03182c244cea9d68e44481da8e7 : SUCCESS in 6m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31179874547546c180a4660a62f2df28 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b6e39de43844fbaad8aca584b888c4f : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6f76a3b70af4aa4b8e41e59f61f99e8 : SUCCESS in 5m 28s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7124b2a91a6242edb1032c7b16fd5ff2 : SUCCESS in 5m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/cf0ee94a49b047bd9f92d6c4093179c2 : SUCCESS in 4m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f096a1985eeb4d53b4439729e2d309cb : SUCCESS in 4m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce7541cb67354d2b87023d71157ada63 : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd9c16bfe7274a77a5fcf94292112be5 : SUCCESS in 5m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/61de20d4e2c0440c9163f1f7840c3fed : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f03951ca99e74411aef0e1d597bc0b62 : SUCCESS in 41m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b74d6b085cd4498787bf225c80f44298 : SUCCESS in 2h 23m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0519b9f7bcd848ada1ae1913714720d1 : SUCCESS in 50m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7791f29ec6d641299d86cc0a9942ab91 : SUCCESS in 1h 25m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4d1ebcc69b454774b32adb71737cdb88 : SUCCESS in 52m 32s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/85069b4558fd4828aed11b2823aab1e8 : SUCCESS in 22m 37s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8aa6f571d74a40fd8bca0c8078afe439 : SUCCESS in 28m 16s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/34e638d8253c476e817a92288bbb9d1c : SUCCESS in 31m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4a65755040434d5fbc74f928ac47276a : SUCCESS in 30m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7370013639da4c1aa340fa695733121c : SUCCESS in 2h 14m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b6db5d65f0a14a35894cefc4a904730f : FAILURE in 41m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7de2b3be1e43460e921a765c8b10f8fa : SUCCESS in 1h 44m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/70d1708789ce4990a2c044f0bbb0985f : TIMED_OUT in 2h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dba7c81b187fbfd43fdf1fdbdb8e20fdee9cf23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-18 17:53:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ecc565fabbf18a3ce2cfb7c032979e2774faee2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-22 07:06:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b918d8574ea4f3601df8b5ca3e76eb8b0951c593","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 07:06:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69d49bf673e19159a0d05168085f58a0663b3a0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:48:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2338694f8014756b50f8002569e9341\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f9b01718ca4f76bc895d84c421e969 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb3feb264f0f4960a9af0bfdbb8b6c83 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/591a54f63bb64bd199f8c54bbc516d25 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e626e6c9adca494a9055aa57e097623b : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1296efc7ba6b426fa9687ef435a0daf7 : SUCCESS in 4m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c44d9101f86496289f127d17b888c2e : SUCCESS in 2m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/547c54ba73d4411eb28e40b019b16e2a : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4296f6b9baa641c1a4c7971b2cfebfb2 : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/00562dfdc62b4dc78a71a2e46cbd9d4a : SUCCESS in 1h 40m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4353b6fe10884f848261f2843775a2c9 : SUCCESS in 1h 18m 47s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/46a3f0e40cf249c9b759fc241451c7e6 : SUCCESS in 32m 14s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5ee488e90ae54f488d793f5b0c4a1f06 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4c87cdd420b6a700eaed330a57552737b3f775cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:48:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6ed5aba0d0016be66e7b47ddb2cd5c7f0b9ba3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:49:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1da5c44192c84fc0a38692c94901276e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa30057f0abb4737a228e699f2402171 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee0794910a2144508ea77c8a1ae0a011 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e02055af0e0c8840076a3b76cbf1c254882ed44","revisions":{"8e02055af0e0c8840076a3b76cbf1c254882ed44":{"kind":"REWORK","_number":1,"created":"2025-04-17 04:54:26.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/50/947550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/50/947550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/947550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/947550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/50/947550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/50/947550/1"}}},"commit":{"parents":[{"commit":"3439d458d3557792ff04b8860cf6b818dd2d2dbe","subject":"Fix amphora image builds to use DIB bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3439d458d3557792ff04b8860cf6b818dd2d2dbe"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-10 12:48:26.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-04-17 04:54:26.000000000","tz":0},"subject":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB","message":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\n\nCloses-Bug: #2101883\n\nChange-Id: I8aa5f128011068321c88f66b1fa986f2ead19034\n(cherry picked from commit 30fbe2119cd540df31b5160b4ff8d8e47e531e3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e02055af0e0c8840076a3b76cbf1c254882ed44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e02055af0e0c8840076a3b76cbf1c254882ed44"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
