)]}'
{"id":"openstack%2Foctavia~947552","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I8aa5f128011068321c88f66b1fa986f2ead19034","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2025-04-22 08:48:28.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-04-22 07:06:36.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-04-18 17:53:57.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-04-22 08:48:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8aa5f128011068321c88f66b1fa986f2ead19034","subject":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB","status":"MERGED","created":"2025-04-17 04:56:06.000000000","updated":"2025-04-22 08:49:49.000000000","submitted":"2025-04-22 08:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947552","cherry_pick_of_change":947551,"cherry_pick_of_patch_set":1,"meta_rev_id":"70aeb74874d2715699895be04b5f9fe6de41f7b1","_number":947552,"virtual_id_number":947552,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 08:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-04-18 17:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-04-22 07:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-04-22 07:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-17 04:59:01.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-04-17 04:59:01.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-04-17 04:59:01.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2025-04-17 07:29:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5c9578fb797d5a581514a0a0c223f1dd0bbb9c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-04-17 04:56:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"6235e279f63f18638aead0e387698cfee5b64085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 07:29:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3898b6f65f9a4d01a8f5e96bcaa5888d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc61cfbf7dc842db8c2f987c1f630f57 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4e7d74daff46ab82914003ad88ea7f : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9677b88574c34122bfa9beba303495bf : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5a3f2b3c0274e529fbd2a78224c947f : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8811fa8f1f1e457e999f0c668c456ba4 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f64028fef88147ea8adc1b03b6dd145e : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9addba0f9f94c67a677c5d0d3a8797c : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c37b80795b15408691b3d185aa054e91 : SUCCESS in 5m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/48fb2049d89648ca82900d00a4ea4620 : SUCCESS in 4m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/731206f934a04027a275763406ea8a1f : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/952f701e67314ebeb8aa15a98abb3ac9 : SUCCESS in 9m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5fe722827aa4be09ca5b0a8f5753233 : SUCCESS in 6m 38s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/93cc730013684a8d9a9fb5cac7d4e834 : SUCCESS in 1h 15m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2349e52731574033bc00db5f30329f75 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/0e1c932069cf4ae597e1c8713e589d5c : SUCCESS in 1h 00m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e813a35ba0684fd5b53e2677487bcf9f : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6f9ec7e2e25c4fcb93e733364277e370 : SUCCESS in 1h 05m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/74bf7b137e0946c18535a39153b30551 : SUCCESS in 25m 17s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2695b2ab4a6c4dfd985ff370d7e711e8 : SUCCESS in 38m 23s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2023-2 https://zuul.opendev.org/t/openstack/build/4e20a32fbe4245f2addd513d7986af90 : FAILURE in 56m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/cb3126dcfa294c07a318b4ef3cda5bf8 : SUCCESS in 22m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2023-2 https://zuul.opendev.org/t/openstack/build/96be42e23731400189a55799219fceff : SUCCESS in 2h 15m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2023-2 https://zuul.opendev.org/t/openstack/build/42dc41dde8784e71be5c2c8eab8cf150 : RETRY_LIMIT in 4m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a2ed01c1c38344a7ab01c164bc6d05c8 : FAILURE in 15m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2023-2 https://zuul.opendev.org/t/openstack/build/0aae7d83664841d7a8f34eacb855aa20 : TIMED_OUT in 2h 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b37e7670d0a60a489d0b920cfa262702464fa683","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-18 17:53:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fca940c872726910576e8bebecd1ac18f7bf293a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-22 07:06:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17e4c3a919832d8504abd2858357bf6787e1c402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 07:07:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fede48ecf150b34449e640e607736d56e0f9d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:48:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ca87cd0be2b4b0da5b9551a92a12500\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78380669b2754af793a616ac013024f4 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c05aaa4576224a38bc7bc55ca42b92fb : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b4f72f821b24658abb6445bc20002a8 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d4404ab8c3846c592adffb968450456 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d414d361863548ddbf1e07aaba4de00c : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fddd4b17097f4333aa4b9bffa2ee662b : SUCCESS in 5m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af0dd88ad50c45f8a36526655402e097 : SUCCESS in 5m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9c7c402ba8f14a2891ae0834e83d5296 : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/044a6a1909314776857a734e5eddc02d : SUCCESS in 57m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ef021a2ec93e41a4aac114aabe5c72fd : SUCCESS in 1h 31m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/0a35ae148cf84b208f90ef27aab0abd3 : SUCCESS in 1h 06m 08s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/62b53d3f8b274492a07e74bdc89e3ba0 : SUCCESS in 21m 41s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e04589cdc8304bfdb085d51ed1af1a31 : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":1},{"id":"102a8d7564238f81cb3368fc79f346e648f50908","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:48:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"70aeb74874d2715699895be04b5f9fe6de41f7b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 08:49:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101b09bba37d497880d56da8e9915737\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87632bec6492409083cc277df3719eb3 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1e05bf2a4264d6ab5952f803d8b2445 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"148abf84d18ddf7cc41818f1db7945f470701d45","revisions":{"148abf84d18ddf7cc41818f1db7945f470701d45":{"kind":"REWORK","_number":1,"created":"2025-04-17 04:56:06.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/52/947552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/947552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/947552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/947552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/947552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/947552/1"}}},"commit":{"parents":[{"commit":"a68edea6e243b1bab88ab73558d5a81581cedac1","subject":"Merge \"Cleanup duplicate DB sessions for quotas\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a68edea6e243b1bab88ab73558d5a81581cedac1"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-03-10 12:48:26.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2025-04-17 04:56:06.000000000","tz":0},"subject":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB","message":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\n\nCloses-Bug: #2101883\n\nChange-Id: I8aa5f128011068321c88f66b1fa986f2ead19034\n(cherry picked from commit 30fbe2119cd540df31b5160b4ff8d8e47e531e3d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/148abf84d18ddf7cc41818f1db7945f470701d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/148abf84d18ddf7cc41818f1db7945f470701d45"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
