)]}'
{"id":"openstack%2Foctavia~948767","triplet_id":"openstack%2Foctavia~master~I8a8f39a00732d7b92098108f3336500cf10dd5cc","project":"openstack/octavia","branch":"master","topic":"psotgresmigrations","attention_set":{"38009":{"account":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"last_update":"2025-05-04 18:10:38.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8a8f39a00732d7b92098108f3336500cf10dd5cc","subject":"Fix alembic migrations to work with postgres databases","status":"NEW","created":"2025-05-04 17:12:16.000000000","updated":"2025-05-04 18:10:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a9fd8fd2669a96dab85c287b7e19d2b4c40adf2c","_number":948767,"virtual_id_number":948767,"owner":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-04 18:10:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-04 18:10:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b11a7737fef18b8245930ba47de51360ae92852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"date":"2025-05-04 17:12:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6b5561eda447202f84964377f6fe34cee3e8258","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"date":"2025-05-04 17:21:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6508866aaee89719546e706f750d343d7fd4b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"date":"2025-05-04 17:29:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f5740cb98277368cfd59c7cf78c68f5bf8d0b2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"date":"2025-05-04 17:34:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a9fd8fd2669a96dab85c287b7e19d2b4c40adf2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 18:10:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/673a596b1d0c45edbfdcf63d1eb9a767\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a37e9dafbb8d492db6b8895c1724d243 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42a4e015fbc34b66bf5450ee774a8e1c : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6400de5714d44ea98c1f835923998aa : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06d68170d8c546789412e460a41c7be9 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b213c1a6f7348f1926658c3e64eda1c : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faafd2fe51d84661aec05eacfe0e0915 : SUCCESS in 8m 45s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3fe5fae0b5444cc09ad41b43c86940b1 : SUCCESS in 3m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f0579c932fe5440287e06906fab99182 : SUCCESS in 5m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a3d7517272914536a5fb1282ba486a9e : SUCCESS in 2m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/73c1b8f23bad4887bca0529ef18f35f2 : FAILURE in 1m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e7248f34d5e541e193afe9900d83ef4d : SUCCESS in 7m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3291898d8c97446db1dabef08513b98f : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/617bda4cdd17463f9b8a1ed9b1005e71 : FAILURE in 11m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cd3e5201f9c54fe4aa8fc398cc8e5673 : FAILURE in 14m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b77c89151c0f43928f3021299e0ee30b : FAILURE in 14m 13s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/921055b82297438a8abda086a4efefa8 : FAILURE in 14m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb30b377f3f2411891a69cd74228eb6e : FAILURE in 13m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/487ffa9cd1724fe8816a2daf7090f864 : FAILURE in 14m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/54cdb91c209d4aef868752a31e85acf4 : FAILURE in 14m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3be6ffe165024793b31537e83af41876 : FAILURE in 14m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc990c7a865f4408b576cd76570632dd : FAILURE in 15m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6a46f32812ff4cde935e5f74fafd0ea8 : FAILURE in 8m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f414f5c614c0420a89d3c25641e520e4 : FAILURE in 24m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fdbc0daeb620446aa780876c5c5639f3 : FAILURE in 25m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/434572cc44c045d58c62b2349e73553c : FAILURE in 35m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2cf3b60ba1794c6d863145ef7211d93e : FAILURE in 24m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d040206e08d44d28b94fcd8789ff805b : FAILURE in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8210b34668c72c147aa52a350c7c2187e49211e5","revisions":{"c87b645d40faddc358cacb542c854292cce78ae0":{"kind":"REWORK","_number":1,"created":"2025-05-04 17:12:16.000000000","uploader":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"ref":"refs/changes/67/948767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/948767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/948767/1"}}},"commit":{"parents":[{"commit":"84f512687a6a1995378bfc5b13753ef4a145601c","subject":"Merge \"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84f512687a6a1995378bfc5b13753ef4a145601c"}]}],"author":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:08:57.000000000","tz":-300},"committer":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:10:51.000000000","tz":-300},"subject":"Fix alembic migrations to work with postgres databases","message":"Fix alembic migrations to work with postgres databases\n\nTask: 23331\nStory: 2003185\nChange-Id: I8a8f39a00732d7b92098108f3336500cf10dd5cc\nSigned-off-by: Ryan Belgrave \u003cryan@belgrave.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c87b645d40faddc358cacb542c854292cce78ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c87b645d40faddc358cacb542c854292cce78ae0"}]},"branch":"refs/heads/master"},"eb112a65b14bf2b049a7b653162b42841d67c93e":{"kind":"REWORK","_number":2,"created":"2025-05-04 17:21:30.000000000","uploader":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"ref":"refs/changes/67/948767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/948767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/948767/2"}}},"commit":{"parents":[{"commit":"84f512687a6a1995378bfc5b13753ef4a145601c","subject":"Merge \"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84f512687a6a1995378bfc5b13753ef4a145601c"}]}],"author":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:08:57.000000000","tz":-300},"committer":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:21:06.000000000","tz":-300},"subject":"Fix alembic migrations to work with postgres databases","message":"Fix alembic migrations to work with postgres databases\n\nTask: 23331\nStory: 2003185\nChange-Id: I8a8f39a00732d7b92098108f3336500cf10dd5cc\nSigned-off-by: Ryan Belgrave \u003cryan@belgrave.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb112a65b14bf2b049a7b653162b42841d67c93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb112a65b14bf2b049a7b653162b42841d67c93e"}]},"branch":"refs/heads/master"},"7bd6970def012a3e7df7968124438c4901f47690":{"kind":"REWORK","_number":3,"created":"2025-05-04 17:29:00.000000000","uploader":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"ref":"refs/changes/67/948767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/948767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/948767/3"}}},"commit":{"parents":[{"commit":"84f512687a6a1995378bfc5b13753ef4a145601c","subject":"Merge \"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84f512687a6a1995378bfc5b13753ef4a145601c"}]}],"author":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:08:57.000000000","tz":-300},"committer":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:28:55.000000000","tz":-300},"subject":"Fix alembic migrations to work with postgres databases","message":"Fix alembic migrations to work with postgres databases\n\nTask: 23331\nStory: 2003185\nChange-Id: I8a8f39a00732d7b92098108f3336500cf10dd5cc\nSigned-off-by: Ryan Belgrave \u003cryan@belgrave.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bd6970def012a3e7df7968124438c4901f47690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7bd6970def012a3e7df7968124438c4901f47690"}]},"branch":"refs/heads/master"},"8210b34668c72c147aa52a350c7c2187e49211e5":{"kind":"REWORK","_number":4,"created":"2025-05-04 17:34:20.000000000","uploader":{"_account_id":38009,"name":"Ryan M Belgrave","display_name":"Ryan M Belgrave","email":"ryan@belgrave.me","username":"rmb938"},"ref":"refs/changes/67/948767/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/948767/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/948767/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/948767/4"}}},"commit":{"parents":[{"commit":"84f512687a6a1995378bfc5b13753ef4a145601c","subject":"Merge \"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84f512687a6a1995378bfc5b13753ef4a145601c"}]}],"author":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:08:57.000000000","tz":-300},"committer":{"name":"Ryan Belgrave","email":"ryan@belgrave.me","date":"2025-05-04 17:34:15.000000000","tz":-300},"subject":"Fix alembic migrations to work with postgres databases","message":"Fix alembic migrations to work with postgres databases\n\nTask: 23331\nStory: 2003185\nChange-Id: I8a8f39a00732d7b92098108f3336500cf10dd5cc\nSigned-off-by: Ryan Belgrave \u003cryan@belgrave.me\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8210b34668c72c147aa52a350c7c2187e49211e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8210b34668c72c147aa52a350c7c2187e49211e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
