)]}'
{"id":"openstack%2Foctavia~948798","triplet_id":"openstack%2Foctavia~stable%2F2025.1~I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb","project":"openstack/octavia","branch":"stable/2025.1","topic":"remove-wsgi_scripts-stable/2025.1","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-05-08 00:47:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb","subject":"Add wsgi module to Octavia and remove wsgi script","status":"ABANDONED","created":"2025-05-05 12:49:59.000000000","updated":"2025-05-13 11:32:36.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":902812,"cherry_pick_of_patch_set":9,"meta_rev_id":"0aa23ede3191fafe70ff1cf4ea52d63638cb1640","_number":948798,"virtual_id_number":948798,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-05 15:54:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-05-05 15:14:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 12:49:59.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"},{"updated":"2025-05-05 15:14:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-05-05 15:54:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-13 11:32:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"5df3ea15f805a6e6a5e727c3423f2e0ec2fe0d98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-05 12:49:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f8a12243e575e9029e5ac938f2a84c9d2384f9e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-05 15:14:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4c6933e660dd6a6f49b470098aa5adfbf6acca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dcec94e44874f859ecfeb9a79f28640\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee17ba9512794857b4629f3bf4b55215 : SUCCESS in 1m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070c9e2d85d34d89a82e406efb2ad33a : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf047c91c2d141bd8cb29fbc67df0bac : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be6ca164c65843a89afcd889f0a9333f : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17d3073743d7495095480bb79e3eac19 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c99bf2e1e864d8ba89f324bf6b805bc : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82adf22e514e469aa6ea18fa736e528b : SUCCESS in 5m 19s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3f20bb6dbf494d92b3a5651224231e49 : SUCCESS in 3m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2ba283de77f642f1a438f30054738880 : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe10ee498ef54030ba6d1763f7433e78 : SUCCESS in 3m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c66ab4ab1274764a147c8e57224115d : SUCCESS in 8m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/718adc26ba9c4ee6beceadd2ffe37ac7 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fda1837baf6422d84680b7c5070aa65 : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fe76d733f8254ae7a7ac2e41161b658c : SUCCESS in 1h 46m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/91a613084cc54126ae4ef1494df3aa71 : SUCCESS in 54m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f90c1102998f42b0b87157c4807d4559 : SUCCESS in 1h 43m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2282465d92d4ea38f3c82bffad9a113 : SUCCESS in 59m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d6ccbc62a28d425195f62672f32e27f4 : SUCCESS in 1h 50m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cf97c9239a7f41629f8812d8a95cf7ee : SUCCESS in 1h 03m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/96d30072fef845edab42c7114b186e7b : FAILURE in 26m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/84a34941144e48a7b64c18738d196d7a : SUCCESS in 40m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cbe1a9018c84d3da27696610706c52d : SUCCESS in 33m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e000d9a5ad74581b9e47c76477cf2b7 : SUCCESS in 2h 31m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bfd30b256b6642aa8c99bf585ff38cba : TIMED_OUT in 3h 02m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7387801b3b094493a06c603cd4289ca1 : SUCCESS in 2h 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ef6e8452addba4bff60616e478a44761e64d1a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-08 00:47:14.000000000","message":"Abandoned\n\nThanks for the backport, but we found a backward compatible way to solve this problem that won\u0027t break deployers on the stable branches.","accounts_in_message":[],"_revision_number":1},{"id":"0aa23ede3191fafe70ff1cf4ea52d63638cb1640","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:32:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"169999a8e47c43ee2eb1fb66737dc697c3ad7667","revisions":{"169999a8e47c43ee2eb1fb66737dc697c3ad7667":{"kind":"REWORK","_number":1,"created":"2025-05-05 12:49:59.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/98/948798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/948798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/948798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/948798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/948798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/948798/1"}}},"commit":{"parents":[{"commit":"3d5e3d470d96af4923e656699df2fbddd9ec061d","subject":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d5e3d470d96af4923e656699df2fbddd9ec061d"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-04-30 07:01:33.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-05-05 12:49:59.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\n\nCloses-Bug: #2109665\n\nIn order to make the CI work, this patch was merged with:\n\nzuul: Drop centos9/py39 and make grenade and barbican non-voting\n\nThe requirements repo has dropped upper-constraints for python3.9, so\nthe centos9 and py39 based jobs are no longer working on master, let\u0027s\ndrop them.\n\nThe grenade jobs are broken due to the #2109665, make them\ntemporarily non-voting, so that we can merge the fix[0] on master first,\nbackport it and then re-enable them.\n\nFinally, octavia-v2-dsvm-tls-barbican fails due to #2109584, so a fix\nsimilar to [0] will be needed for barbican.\n\n[0] https://review.opendev.org/c/openstack/octavia/+/902812\n\nRelated-Bug: #2109665\nRelated-Bug: #2109584\n\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n(cherry picked from commit 8ebf8d27c4209932ea1e5dfb3ca397af6c92e14e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/169999a8e47c43ee2eb1fb66737dc697c3ad7667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/169999a8e47c43ee2eb1fb66737dc697c3ad7667"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
