)]}'
{"id":"openstack%2Foctavia~948859","triplet_id":"openstack%2Foctavia~master~I17d3725fabdfce4b0d1bafe301e435992e1d5e13","project":"openstack/octavia","branch":"master","topic":"ssh_disable","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-05-07 13:33:14.000000000","reason":"\u003cGERRIT_ACCOUNT_37981\u003e mentioned you in a comment","reason_account":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"}}},"removed_from_attention_set":{"37981":{"account":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"},"last_update":"2025-05-08 06:11:25.000000000","reason":"\u003cGERRIT_ACCOUNT_37981\u003e replied on the change","reason_account":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"}}},"hashtags":[],"change_id":"I17d3725fabdfce4b0d1bafe301e435992e1d5e13","subject":"Fix for ssh disallow not working in amphora image","status":"NEW","created":"2025-05-06 10:14:41.000000000","updated":"2025-05-08 06:11:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"ff987eb306aaa25ee6c50fd8fdc99ffc0285efe6","_number":948859,"virtual_id_number":948859,"owner":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-06 13:12:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2025-05-08 00:44:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-06 13:12:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 08:47:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-05-08 00:44:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-05-08 00:44:50.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5c1ec585d5100354a9a9a9b5d3941dcf00094193","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"},"date":"2025-05-06 10:14:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dff29f74e07c367457ff217bb4c8ef095f6150e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 13:12:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63bc3e55ca724e919d45a7a7ebde33cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fbb56a6cd8f485788e25304a1a84101 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892a82ad92d34c1e92f7c6382c9b1b59 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19f30c208d794a1a8f45e8e9db50ecbc : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44a74d7c9c784741a27e44c4e18a559c : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6e077466ad84e45b4e5bc54fd266d75 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a159a8b4034e668c79b198ed410274 : SUCCESS in 12m 38s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e208ce4761cc43078adf8c9082ee450a : SUCCESS in 4m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/17ed631cf39c4fefbaa1968d5d142992 : SUCCESS in 6m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0d6dd58aabc743cf8e053b691f9a50d3 : SUCCESS in 3m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d93c1891098464eadda3cf1756ee56d : SUCCESS in 5m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7e77111afc844f1282a9d6465fc971e0 : SUCCESS in 6m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1bb1e7328d24ca8b9a9a9cf8fead75e : SUCCESS in 43m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cb9a8856e5b246daad8b039d246a5104 : SUCCESS in 1h 45m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d6eec553011f4aa7b2209c02e8920dea : SUCCESS in 56m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39f65d51df6448289ef29a400094fd1a : SUCCESS in 1h 44m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c7a597b918e425d9dde495e1bed5a40 : SUCCESS in 51m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32b56bfced734230a2dabe3d71ee4ba9 : SUCCESS in 1h 48m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/65737db5708c435a884beb54df6ece33 : SUCCESS in 1h 02m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab8009b060904eb5968ce89e3eb440f5 : FAILURE in 31m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0dd32089d922488fb9bb13daa43c567d : FAILURE in 32m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/418f38608aec4a2480ee03251202f2eb : SUCCESS in 39m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/42d6ab1f47554f7987985b4754991d0a : TIMED_OUT in 2h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/580d35339ca94b62bff47bdfcb521688 : SUCCESS in 2h 55m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e7182dd22ba045b98335b9fdbf2c95ed : SUCCESS in 2h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad5562326c45fa4fa7d4b4e763ae93f8bc026134","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-05-07 08:47:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"120d3c3014d06cc407e9e75fe02d141fe1717adb","author":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"},"date":"2025-05-07 13:33:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e48ed483d63784bfb26a6a0b22838784b2b36f3a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-08 00:44:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62202d3d66c471a44962a587e1567db0e762c720","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-08 00:44:50.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"ff987eb306aaa25ee6c50fd8fdc99ffc0285efe6","author":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"},"date":"2025-05-08 06:11:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"985843b2b2f9976bfb9a5587162ec7e10adb7adc","revisions":{"985843b2b2f9976bfb9a5587162ec7e10adb7adc":{"kind":"REWORK","_number":1,"created":"2025-05-06 10:14:41.000000000","uploader":{"_account_id":37981,"name":"Manoja Kumar Sahu","display_name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","username":"manojacloud"},"ref":"refs/changes/59/948859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/948859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/948859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/948859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/948859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/948859/1"}}},"commit":{"parents":[{"commit":"8ebf8d27c4209932ea1e5dfb3ca397af6c92e14e","subject":"Add wsgi module to Octavia and remove wsgi script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ebf8d27c4209932ea1e5dfb3ca397af6c92e14e"}]}],"author":{"name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","date":"2025-05-06 10:05:40.000000000","tz":0},"committer":{"name":"Manoja Kumar Sahu","email":"manoja.sahu@rackspace.com","date":"2025-05-06 10:05:40.000000000","tz":0},"subject":"Fix for ssh disallow not working in amphora image","message":"Fix for ssh disallow not working in amphora image\n\nCurrently openssh-server package is not getting uninstalled even after passing \"-n\" flag as per the process while generating amphora image. So the generated image still allows ssh. Removing package-installs.yaml file from remove-sshd element as openssh-server is unable to uninstall using this file as this package is getting installed by amphora-agent element https://opendev.org/openstack/octavia/src/branch/master/elements/amphora-agent/package-installs.yaml#L76. So moving the openssh-server package removal part to post installation section of remove-sshd element.\n\nChange-Id: I17d3725fabdfce4b0d1bafe301e435992e1d5e13\nSigned-off-by: Manoja Kumar Sahu \u003cmanoja.sahu@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/985843b2b2f9976bfb9a5587162ec7e10adb7adc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/985843b2b2f9976bfb9a5587162ec7e10adb7adc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
