)]}'
{"id":"openstack%2Foctavia~949280","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I7e0b17dbf9284db27e183a9a27e716266e26a237","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-08-18 15:06:32.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I7e0b17dbf9284db27e183a9a27e716266e26a237","subject":"Add an octavia-wsgi compatibility script","status":"MERGED","created":"2025-05-08 21:39:18.000000000","updated":"2025-08-18 16:59:28.000000000","submitted":"2025-08-18 16:58:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949280","meta_rev_id":"8e6df68c68496d5da2904a67a10eefa0d130791d","_number":949280,"virtual_id_number":949280,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 16:58:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2025-06-25 14:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2025-08-18 15:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-08-18 15:06:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-09 00:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-13 11:39:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2025-06-25 14:05:21.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"76a6b0a19429272c0f2225bb98f6e509856254eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-08 21:39:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"929d062e3782a65a3c036e18359bfa86ee2ce9dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 00:15:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61fee1c536d04972b622c70aeb57a01b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b2b0f991fa2c4b9dbbac62eb83162302 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1147e025210944369e83342db1fe6a79 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a96428298e4146a1942556bdd7baf4 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cc405f250c84ce29b918cace4a3f971 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a526ad04371c412682492f5e7a3a65d0 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/125cc00be6aa43a4a85783f3961afd38 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68636296184943df92ff8de05fda9304 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cc41bb0f0804ab094bd8aa96125488d : SUCCESS in 6m 23s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5e197aea39b94cfdaa774227ca828391 : SUCCESS in 3m 44s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fc6fabd99dca49828adde06e76f3d4d8 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f9318d8b1504c6a9e40eaccfa32d9ee : SUCCESS in 2m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/54b63ba269d74dc08f849c13941e2eb7 : SUCCESS in 7m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97bf930470f14b618f2e0e361aa933f7 : SUCCESS in 7m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3ebbb3a4da3c4a718b62932f3a36125d : SUCCESS in 3m 45s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5bc59705ecc14d5b8692c6cc5faa1d39 : SUCCESS in 1h 00m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c9052fb7c7a646c8a1796232c9a431b7 : SUCCESS in 1h 50m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b1766059a6b245f19e406e9ade199316 : SUCCESS in 1h 25m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/eec97172ade54786ad8574d404f9e9b0 : SUCCESS in 1h 25m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/db28114101f94ef096150cbcff350375 : SUCCESS in 58m 03s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5c7d9c80e1944f07a2153a273b90f733 : SUCCESS in 51m 47s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/850d9ee9e5f14b469cc94b124a114cda : SUCCESS in 44m 03s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f00ff01b8dcf4d41baf04d688ad5ee7b : FAILURE in 52m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/28cc09f03a404ac6b30ed3c760fb4bde : SUCCESS in 21m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/544208de2dfc42f68ecfa782e610e7f7 : TIMED_OUT in 2h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/51faa21b4e03445ebbdaf1a29600358f : FAILURE in 38m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b58107c746c24efbb979d1649e8a0cce : SUCCESS in 1h 52m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/7672008d465343fe945b2f4bade007a5 : FAILURE in 1h 48m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e4c68d15c6b58f0954dbdd4cb1667e08365eeb3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:39:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9cb2338eac709ce10250614356c80eb071601ed","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-25 14:05:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44dccb3f2d38f3dfee859c01a7434bff2457ade2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-08-18 15:06:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"754eb8dc77bcb650e79c72f674fc94978e99f88a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:07:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"664091420783a83b0775331119e19c7f9575a968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:58:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b74d96838d476181d1c2d5f23bd02c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d1e0d3a205a246d58f7fd78d65d70ba1 : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3ee5d013f44c7bb1ae59f97a7c9d7b : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c59fb3053cbc4afe9750bff149e6d2f7 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4db066b30a71400eb638698cfa432c6d : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44efc82474547c59796174430a95f97 : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/392eab6a199649288d9dc352c82477f9 : SUCCESS in 4m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/803430c65aae4384ac148ece41bb52ed : SUCCESS in 2m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/38bc775a11214828a95bac1ac5b0a711 : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/4554b6dbd0044e77afe629530572d221 : SUCCESS in 1h 03m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0c338fb74c24432eb03fec31fea24cd3 : SUCCESS in 1h 44m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/0efbdc0c851142f8bfa56251cbd551a3 : SUCCESS in 59m 17s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fb01bdadbc774ceeb28eabbea3001df8 : SUCCESS in 32m 22s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b1b9b44220ce4bf3b855615ada56147f : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d07f105235c6cf7debc952b865e91cf175367d9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:58:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8e6df68c68496d5da2904a67a10eefa0d130791d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 16:59:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a11487173624deb82fd9252d4761504\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61e50f4cd7214f168150425f2eb07e70 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56d86639069d4e978ff9f9659ba1328c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8249ae9b837acb14a60e8d235cd0e8e1b4dde1c1","revisions":{"8249ae9b837acb14a60e8d235cd0e8e1b4dde1c1":{"kind":"REWORK","_number":1,"created":"2025-05-08 21:39:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/949280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/949280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/949280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/949280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/949280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/949280/1"}}},"commit":{"parents":[{"commit":"8e02055af0e0c8840076a3b76cbf1c254882ed44","subject":"Ignore load_balancer graph in task results to avoid break write data in jobboard DB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e02055af0e0c8840076a3b76cbf1c254882ed44"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-06 15:14:27.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-08 21:39:17.000000000","tz":0},"subject":"Add an octavia-wsgi compatibility script","message":"Add an octavia-wsgi compatibility script\n\nThis patch adds an octavia-wsgi script that gets installed with Octavia\nfor backward compatibility on the stable branches.\nThe pbr wsgi_scripts function no longer generates this script\nautomatically due to recent changes in setuptools.\n\nThis is a stable branch only patch.\n\nChange-Id: I7e0b17dbf9284db27e183a9a27e716266e26a237\n(cherry picked from commit b8824ede82c42b81ca35248bd6ab6fb8bf51f3cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8249ae9b837acb14a60e8d235cd0e8e1b4dde1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8249ae9b837acb14a60e8d235cd0e8e1b4dde1c1"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
