)]}'
{"id":"openstack%2Foctavia~949548","triplet_id":"openstack%2Foctavia~master~I9ee1913e57a1addd623c993d9c9083fad9f2bb33","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-06-26 05:00:15.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-06-05 07:14:47.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I9ee1913e57a1addd623c993d9c9083fad9f2bb33","subject":"Fix duplicate VIP address error reporting","status":"MERGED","created":"2025-05-13 00:58:51.000000000","updated":"2025-07-02 17:40:42.000000000","submitted":"2025-07-02 17:39:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"949548","meta_rev_id":"9f475395514ace4453b9fa0f74d483122c47bb49","_number":949548,"virtual_id_number":949548,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 17:39:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2025-06-05 16:31:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-02 16:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-06-05 07:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-06-20 19:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-02 16:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 01:01:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-04 08:27:31.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-06-05 07:14:47.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-06-05 16:31:48.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2025-06-20 19:54:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"b662b8375c0e47b319c17cb8822fec4829bd5f7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-13 00:58:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7287ff92db061861d1b87881921e307ac522bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 01:01:24.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/949546 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"6d35d0f8d7db5e8ffc7695833a4362572cd66e68","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-13 01:01:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc1607e9bd40c70208a4023e2e1e0e2e5606894d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-13 01:03:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be41909dc82ba9934777aa911d1a7e4c83be0d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 03:38:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6114a92e680b4ef2a525fb0ec951b148\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcc97c60ca44453faf0d130c1a1707ef : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7065318410448eb6882b66775300a4 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2498253a5c049c28ec6e66575ed83f3 : FAILURE in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f9416756ffc420d9267c5eb0a72b77a : FAILURE in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4fb6f7f099e4b38b1ceb37c09fd1733 : FAILURE in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bd44a1c5bcf436d836dd7365500a551 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e78e4a227b5e4a9184f2f7a64f44530b : SUCCESS in 4m 40s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e6cd1002e545465883c6d02c560e39a0 : SUCCESS in 4m 03s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/bf3216728df0466c8d93377f4414afe5 : SUCCESS in 6m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5e8003f31bd54889b40d58ef1323f873 : SUCCESS in 1m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dcc8bea0d5a54b45842296ab0fed1443 : FAILURE in 5m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/91a59bf3afbf4cbfa9b87645620d431b : FAILURE in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/62645242aef644d7ab4875b6cd4d42be : SUCCESS in 1h 30m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/81331e0cb4c0486f81c9e66f349e9719 : SUCCESS in 1h 25m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/052be6855740499ea4552793c1f2d11e : SUCCESS in 45m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c428d47be21f48e49f075f77f5ef2dd1 : SUCCESS in 1h 52m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4db7a7514bd4f93b0b7f8a58456ad12 : SUCCESS in 1h 01m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7f1a1a580f58454eb17c086f3e28af57 : SUCCESS in 1h 27m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/85185538688047a8ae942355993359a2 : SUCCESS in 1h 00m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f322e359f2ac4383b908f9572d497c01 : SUCCESS in 24m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/516eff2e3515474aaf887e273ce39339 : FAILURE in 39m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb0c28bc2e7f46bc8c2927046cad8d94 : SUCCESS in 23m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9af96e1b04146428635ef9bb1ce2def : SUCCESS in 2h 09m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/91823e491bdb4d239822f11c21157f70 : SUCCESS in 2h 09m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f38ccf0eee2148cf9ed7444e72c2495b : TIMED_OUT in 2h 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a805a8e0c4f81492c339e4c0bbb4c19c66ce5db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-13 20:23:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7796fe0f18bdcab4389b68ed7f98fd38ecbdf1fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 23:27:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9df295d6e13647b3a9ef9e92bbc3c26e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/682d765ef96348ac890bd499ca91da4c : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6885e5ccec4b4192adab07ef7992c258 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89ac9dc5999840d3ad76695c27cb22f9 : FAILURE in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32b316e476ee489392f53bbfb89dc780 : FAILURE in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ceecd9acd5cb447f9221ce69792beb6e : FAILURE in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f908460f2d48829139a9d764aa54ca : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/289ecd1ab4534daeaa1bde211a4059f4 : SUCCESS in 4m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9b5a4e9f293146a4935ab81e5822522a : SUCCESS in 5m 19s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/10e1d292ca20436c8d15abace934d876 : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/84d86f096eaf47788b6f58019e9f8d0a : SUCCESS in 3m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca78c1f81deb4a40bd4fffa790877864 : FAILURE in 6m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7b577566dcbf41c9a6157a0b44f17975 : FAILURE in 5m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c587e843bbfb41a8b4ab7f314bccf971 : SUCCESS in 1h 04m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2305e9d4b7fe4d08bd0f4fabf11897a2 : SUCCESS in 1h 52m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/81d28108da2244a39ce895c1a7319090 : SUCCESS in 1h 34m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fd9e3cf3e9848a19bcc9d9010d59e95 : SUCCESS in 2h 02m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82755a611be041bd8fe7328ed51590e6 : SUCCESS in 1h 28m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7ec1741ce85f47bf8f7b369813222efe : SUCCESS in 1h 57m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5989e4e0d0a648ec95e35d277fdfe6f5 : SUCCESS in 1h 27m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2420ab7b89a245c78febe3517981e8d9 : SUCCESS in 57m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9908c29f42ed4c45bb9382b46a6678e4 : FAILURE in 43m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3779e4b6723e44d1ac765de45b393c1e : SUCCESS in 25m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7730f71fbdc741568e9b35ccaa5b0bac : TIMED_OUT in 2h 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6790617cdb3244e2871a22dee898b273 : TIMED_OUT in 3h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0bcf1c54cc724240afb8d23e1b929164 : TIMED_OUT in 2h 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db1ca6bf3e37c75844441a7152175b5f6eb45855","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-04 02:35:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"879ceb80fb19cea6de2b4dc3f6fe96e1f33e23f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 06:01:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27230d799b7c4d7ebe6c23e54ecef588\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d92024b77d304dddad517b5a000c47bd : FAILURE in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85271d3651184072ada7262b20606728 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cd76281717840088f84e1617ce6be87 : FAILURE in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44838cf8c1a14f1f838e0ec3011d47ab : FAILURE in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1fcec2f04af48129445257dc1978f10 : FAILURE in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46e3481c3f654687b1e3705b14589519 : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/284f3b9b2b0a46eb9a6ae5be6325b7db : SUCCESS in 5m 46s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/70b96efa0efb4d55917a6f031eed9722 : SUCCESS in 4m 47s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2941703b4c4c4110829d404c9aeb60ad : SUCCESS in 8m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a3375a3364054d8d87bec3ddc3afc497 : SUCCESS in 4m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2b76df55373f4ab09d85e5dc545e71e2 : FAILURE in 8m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/538a6880d08345a988020c105f131153 : FAILURE in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c63e80be3e06475994d7b55bf67b7449 : SUCCESS in 1h 14m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/db8fd1a5f41144748b1f048817efb9ad : SUCCESS in 1h 47m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0ff82f8f9d7b4549869ae9fea4501f55 : SUCCESS in 1h 03m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9b7cf3837ef4e41888605d247431915 : SUCCESS in 1h 44m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bfdd843ef0440ec873d34d76efa8dd6 : SUCCESS in 54m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1272347e27fc49b9b3d8227e801ce1ad : SUCCESS in 2h 19m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/519717a026624eb8b68a48914ab973a3 : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60de82cb8ed849429faa0141e06721e6 : SUCCESS in 36m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/236de4c593154d2fb6b605c68342dacf : FAILURE in 37m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67ca2ae3f5c24d19888868a12b3d4778 : SUCCESS in 32m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e0e17202b8147288a8692270e5d39f3 : SUCCESS in 2h 13m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d87ddc1eec0c494a9ccff01fd1618d00 : TIMED_OUT in 3h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b333698d546440d594a03a242200aceb : SUCCESS in 2h 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01f51a20ca0e1b8f3ec8db4b7415a9aa31028ff3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-04 08:27:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bda39ebc50491903871ae367b6a8171facbebae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-04 20:51:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6afaac8b8bcbcc84334f42150e69815b3eb8a3cf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-05 07:14:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2a4040e155441ba3512b0e64ac5982e90b4bd0c3","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2025-06-05 16:31:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45d4b4551ac93126813e8fc542c1c24fc216c86f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-20 19:54:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2e4738815ca403b9192b20b6e789b7d2c9d2ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 23:07:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c443a33935642d8b3ed2ac7bbf556dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2580b0479090479eb109ea27ea620fd3 : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f2df72a0e3427aabc83eba60587656 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6667d0d49c9244bcab85a199ed0985fe : FAILURE in 7m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73087d0574a54f9789740d08737cdf57 : FAILURE in 6m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1410a6452b54f7789b24aade6bbb061 : FAILURE in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b895330956742bb975ce9137d217616 : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a061bd3588304ea9bd181ffaa041bced : SUCCESS in 4m 24s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bfe62734947c495a80f946d9aefb9768 : SUCCESS in 4m 10s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0e3841edc89649799c853ff1a782eea7 : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7ee4a8be6bbd498bb554b08f00b18bc2 : SUCCESS in 3m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99ce5a54c9ae40abbed68693e5512351 : FAILURE in 6m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9643ddff4e374c4296a65de4dc30a894 : FAILURE in 5m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee27b8a80a69459d9c4bffe642cf5b12 : SUCCESS in 1h 30m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e3791892aeba4c6995c1d1964d2cd004 : SUCCESS in 1h 59m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/15589b17e8ca420f9c7a26fbe8bf5f68 : SUCCESS in 1h 03m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a4513efd4fa4f38a9c6b23fa05cccc4 : SUCCESS in 1h 58m 50s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3747cb82a1e14484a13b934194c78a25 : SUCCESS in 1h 15m 39s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/741e0c7ad9b9416fa16a8feef8c19db9 : SUCCESS in 2h 06m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1d7fe20611c84b7b88ca30686fb41b64 : SUCCESS in 1h 11m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33e3e9a5ee614213be7ecc976d1fee5f : SUCCESS in 39m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d56fac52c1eb47e68030c55a13296e37 : SUCCESS in 45m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27d26f8cb36443cb9fd5e20d875dae93 : SUCCESS in 54m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/24a4409a9adc4423809a22e5737b6b2d : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e090cd5c027d440b9adfd78164f1c2ed : TIMED_OUT in 3h 04m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/472413d690ec41ee9a8d0fc021fa36d4 : TIMED_OUT in 2h 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e086810fa9123fb18538ad59592ee2449c5817e7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-20 23:15:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5fb4be1363ec638a5a2f17cc67304b12890cde2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-26 05:00:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c29dab90ba3da59a2243c9dacc7088bcf3f5f21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 08:05:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2be215e8c614f36871ba0ba5d7dac7a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0feff7943d164d9da8ceb85d501a8086 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1a04eb2f9240afb238ec111bb019e5 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90e86438d7334dedaf6baf400ddc0fba : SUCCESS in 6m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a01d16dec45413e8af53de57c88e6b4 : SUCCESS in 6m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dff0ea48283f46ada292d4540eab02aa : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7327b967185e4f22b03f1a5f8ba16c49 : SUCCESS in 15m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14a003f582b547be8f9e80c20b2727b2 : SUCCESS in 4m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9bd0c96fb5154b748fab55bf577245b7 : SUCCESS in 8m 21s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b8a96505f6444e7eacfc2ba0c715b65b : SUCCESS in 6m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1cbb8b41c34d476e83a47bbaa1a421fb : SUCCESS in 5m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13050e66c6fe491b91304f0c461c783d : SUCCESS in 6m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2c328475919843b4acbc1f68ea36ecf1 : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1b5ebc5dc0114550913a821e86fdb57e : SUCCESS in 1h 24m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/04e38e7b53c44b8f937d0a62872ed863 : SUCCESS in 2h 11m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/60d04db1629f4aeca0197c03a23cd534 : SUCCESS in 1h 35m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0eb8cbc5656d45acb1a3080a2a88219c : SUCCESS in 2h 12m 24s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ffd0a886a444d19bf73ed190d049d72 : SUCCESS in 1h 27m 51s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a169542c984d4aaeb37b4e34fd0ab7e6 : SUCCESS in 2h 15m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32a1cdf8e25b4bbf9412c4db0fa30b3c : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/73ceafa17b4244b58fe5f955306708f2 : SUCCESS in 39m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1eaf109c27a942c7aa100bfb8c0658b7 : SUCCESS in 1h 06m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0b34cdb9cb9462dba7b80e41cc69cf6 : SUCCESS in 43m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8ada191f206416298e64fdb3e5a6e28 : TIMED_OUT in 2h 34m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f0be559184a24320b48387b581c9b979 : TIMED_OUT in 3h 03m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9015c88d5b014f17a93dd2d63587e5d3 : TIMED_OUT in 2h 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4714b9a85423b7fd7b4d0816f7fad22bc4f4a997","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-02 16:03:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae6a4697eece6d9c82d4e55769fbec80f5ff3241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 16:03:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3161474823aa6a9c057d2bf0cc2b8a559f5571e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 17:39:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72e1383297204dceb3629580fbffe2b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e931e52e54544a0d837e378dfa7af4c7 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab674b8eacdf4b109e8ed43cf40e672e : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6feff86c33fd4f9880ac8ff30d61efff : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69fc89ea98d5476d9cfcf285c4e87652 : SUCCESS in 13m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97bf3e2c650f4299a413467198ee5846 : SUCCESS in 5m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/16205e34672c497b8c97c3711cc8bbe1 : SUCCESS in 4m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf36699fb5a8472bb105c590261188d0 : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/909107c193a8417baeac208b1aba7ca0 : SUCCESS in 1h 14m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e3ba8f57d52747d1a3eef0902ec63f0b : SUCCESS in 1h 35m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7d1cd47a45604d85b21d0b6ce7afc7d3 : SUCCESS in 52m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a91481041a0c427bb7b9ce2a68355daa : SUCCESS in 27m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/212d2d1cbf6d462c84aebbe6ea29a401 : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":2},{"id":"71af7ae009b3925d997b28829d624f066c7c34be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 17:39:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9f475395514ace4453b9fa0f74d483122c47bb49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 17:40:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/498bcd0494214393bbdf4c43fc97f0e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1dd39074fa74a25a3d866e1cda47f41 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fa97b2db18045719922490630da20e9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e6fb72231235d737c09327d7235dbea794a13337","revisions":{"427affb172b9d31c5bc09a3ede1a7906d1a9afeb":{"kind":"REWORK","_number":1,"created":"2025-05-13 00:58:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/48/949548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/949548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/949548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/949548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/949548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/949548/1"}}},"commit":{"parents":[{"commit":"4db66dd06d0f0cb9fae427b6b53d92ae2851c118","subject":"Fixed publish-openstack-octavia-amphora-image-noble job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4db66dd06d0f0cb9fae427b6b53d92ae2851c118"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-13 00:46:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-13 00:58:49.000000000","tz":0},"subject":"Fix duplicate VIP address error reporting","message":"Fix duplicate VIP address error reporting\n\nWith the transition to supporting provider drivers, the error reporting\nfor cases where the user requests a VIP address that is already in use\nwas broken.\n\nThis patch corrects that such that the user will get a clear error\nmessage.\n\nHere is an example output from the CLI:\nThe VIP IP address is already in use: IP address 172.24.4.25 already allocated in subnet fb3df045-07f9-4bd5-a7fb-12b016e5f873 (HTTP 409) (Request-ID: req-cece42f5-6dc2-49f8-a551-c11fd1abeb81)\n\nCloses-Bug: #2052682\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/949546\nChange-Id: I9ee1913e57a1addd623c993d9c9083fad9f2bb33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/427affb172b9d31c5bc09a3ede1a7906d1a9afeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/427affb172b9d31c5bc09a3ede1a7906d1a9afeb"}]},"branch":"refs/heads/master"},"e6fb72231235d737c09327d7235dbea794a13337":{"kind":"REWORK","_number":2,"created":"2025-05-13 20:23:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/48/949548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/949548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/949548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/949548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/949548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/949548/2"}}},"commit":{"parents":[{"commit":"4db66dd06d0f0cb9fae427b6b53d92ae2851c118","subject":"Fixed publish-openstack-octavia-amphora-image-noble job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4db66dd06d0f0cb9fae427b6b53d92ae2851c118"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-13 00:46:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-13 20:23:19.000000000","tz":0},"subject":"Fix duplicate VIP address error reporting","message":"Fix duplicate VIP address error reporting\n\nWith the transition to supporting provider drivers, the error reporting\nfor cases where the user requests a VIP address that is already in use\nwas broken.\n\nThis patch corrects that such that the user will get a clear error\nmessage.\n\nHere is an example output from the CLI:\nThe VIP IP address is already in use: IP address 172.24.4.25 already allocated in subnet fb3df045-07f9-4bd5-a7fb-12b016e5f873 (HTTP 409) (Request-ID: req-cece42f5-6dc2-49f8-a551-c11fd1abeb81)\n\nCloses-Bug: #2052682\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/949546\nChange-Id: I9ee1913e57a1addd623c993d9c9083fad9f2bb33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6fb72231235d737c09327d7235dbea794a13337"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6fb72231235d737c09327d7235dbea794a13337"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
