)]}'
{"id":"openstack%2Foctavia~949581","triplet_id":"openstack%2Foctavia~master~Ic40ace51297e72ae56a955f41c6bb3a22dd5b46f","project":"openstack/octavia","branch":"master","topic":"add-octavia-quota-sync","attention_set":{"35873":{"account":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"last_update":"2025-05-13 12:16:36.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic40ace51297e72ae56a955f41c6bb3a22dd5b46f","subject":"Add task in housekeeping to fix wrongly updated quota","status":"NEW","created":"2025-05-13 09:23:50.000000000","updated":"2026-05-13 00:36:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5a601d7211dd25b4783095788581ab2c0a904c03","_number":949581,"virtual_id_number":949581,"owner":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-13 00:36:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 12:16:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"931b80f0f3f3276802cf3fd51231f0e1979a08f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"date":"2025-05-13 09:23:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33c5dd778ce4ca10f524be44ea6a57305ad08e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 12:16:36.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b59d210dac4a42d7b779964c2e0d8205\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2e551e6a3e4462adc8233985d589be : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c120d5a7d2754c1aa2b852f52895fc83 : FAILURE in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0385a50ec32e42fab55a1cc8dbdb3ae2 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/188e6f66cf6b4628b5e00580d884dd39 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c9b0249b84024fd5bbf465d074f3cb57 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5d87f06ea741698b2a0923316cb9c7 : SUCCESS in 13m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e27587bdead643ae869267605b287d25 : SUCCESS in 5m 16s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/70d626902f6d447c9b6b71cb2c4488c9 : SUCCESS in 9m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1993421dc3c749f492b3937bf2ff6799 : SUCCESS in 3m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e50905a8ee642118f1c3c09cb6fb2b1 : SUCCESS in 4m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1e9fe459e41d4f92a7866f64e43f7adb : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1209fbc39b58473c9fbd8a2cb7ac0510 : SUCCESS in 1h 32m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0a7cb935e91f4a57abf181a365570afd : SUCCESS in 1h 31m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7e9631bbfa294defadbb467578bfebe0 : SUCCESS in 56m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9d33bd103c445fca71144ddca3d3248 : SUCCESS in 1h 32m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e32e3bc907c408dad19bf4a81418532 : SUCCESS in 1h 20m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/409f67039a184d84a67a1372ca9df206 : SUCCESS in 2h 32m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9641961c98644698a41c048ad15c065f : SUCCESS in 47m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/017bbb4fb8f544e8bdf1469781d24ed8 : SUCCESS in 56m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60d7ddef815f4828a612e1e70126c7d7 : FAILURE in 35m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27ee873223cd478babace7dd80d918ab : SUCCESS in 37m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/efd3707ab19941cd9f859f897cb9f435 : TIMED_OUT in 2h 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/10fecfe804554441965e629c19397597 : SUCCESS in 2h 50m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2e8d14f9b8d34cf398c70b6fe16184b5 : SUCCESS in 2h 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3931234945c71f6c4db0a6e44c3c9a2023734fec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"date":"2025-05-13 12:50:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8f556fc9afe87f2954b2f4a9c7f58d067c8b1e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:49:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca319cc62494ca2bd5faf1d349399d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1f6de19e25747c4a7799096568cbd6c : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/610510fe752a4391b25910d3173b6888 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e000961054214a13a099ba87445bb34d : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1801b8f55085445c83e43225a9da2a3d : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4516fde0814441d8bffee34e963e9d41 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d060a08eee8549aaa1865e9f1630c8fb : SUCCESS in 13m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/89aa2f946ad945378588975d690fa51f : SUCCESS in 3m 23s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/adf7cd3ba29540298c4d2e08ba28fb40 : SUCCESS in 4m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5c649fbef9554eed974b62730be4e2c6 : SUCCESS in 4m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f438c957089e457681a090945731acc8 : SUCCESS in 9m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8c06ab8061b249b183a379e2d28506dc : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a127c78a9384a178cae57bc496b2162 : SUCCESS in 1h 06m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7c3ec39dbb574dce9097e7f685ad51a6 : SUCCESS in 1h 47m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b50c5b82714b4ad4b71bb0295ccdc1ae : SUCCESS in 1h 08m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/463ef26d01a840c884aaeeab97adb064 : SUCCESS in 1h 45m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/154fd91cfb05484f9147d9d47bcf233b : SUCCESS in 1h 04m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d1abdbe992d14a489f960ebdb0ba1a74 : SUCCESS in 1h 52m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1bb68881082842488d47b63a160a64e1 : SUCCESS in 1h 08m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/277e8487af9c4f99a3fad44653481954 : SUCCESS in 41m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60ea7e6ab5d743589d4931e29d8156cb : FAILURE in 42m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/560e1e199f50495dab6f84579fa01744 : SUCCESS in 39m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d3929b2c3f4a4c908ddbcefa47c9d29c : TIMED_OUT in 2h 32m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37da09272c6243118b7f143480192d76 : SUCCESS in 2h 52m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e14c34fb9c4d4b6b83e6e35918242aba : SUCCESS in 2h 23m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"08e8054cbb4e97c59aa0156357e0db7a737a4a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"date":"2025-05-23 13:44:00.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf4a238012dd10f958388eea6855c0a6664868a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 16:29:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ad4194b56944dd390ff255615514daa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a0c85d526c54436aa18f14a32d45d1c : FAILURE in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47758c3ce1c4c278a9c3a58d79110a9 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79fd49b1219b4cf78bd92c62ca997c68 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a517c8b90dd4207adf0ab2d58945fc7 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2091eb3b17f4cc0bbe07486fb4ad36f : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed1e9991b904183aba02d67387806fb : SUCCESS in 13m 00s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a5ac40d21cba4706b33c6b903eab32c1 : SUCCESS in 4m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6367f299c7ac433fb6ba86bbddf6aaab : FAILURE in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/688ea6130dcb4d409a8b982ef23e5565 : SUCCESS in 3m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4d9219888674a52b2719ebf98ce8457 : FAILURE in 7m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/86e7bd31856043f783f61956e86fe972 : FAILURE in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c3292d523d95453282d5ecd53af1d540 : SUCCESS in 1h 21m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c07eaca7d45e46dda00bf0deaa51f1a4 : SUCCESS in 1h 50m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0abde565a4594ba28dcfaae61827aa33 : SUCCESS in 1h 09m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e7a05e050ad4435b43cc49aad8747dd : SUCCESS in 1h 47m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eda7412c243f47f795fbe95716c3d447 : SUCCESS in 1h 09m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f61ea3853cfd4d639b32610b8fb4524d : FAILURE in 18m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9d33f973d05543ecb20ab05c6ca587f1 : FAILURE in 16m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d5ff97af3c2843a7b0bb5339c0181bb9 : SUCCESS in 49m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8303ad8381ea4cadbfb8012f8086fd6b : FAILURE in 40m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68b4a0d6ff77438d8f69913d76436c5b : SUCCESS in 45m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10806053065b45c982891fa2df310232 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/192109ccfc87435f87ebf9ce097c1781 : SUCCESS in 2h 44m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9f378ed9be714aa9a59c5af009da4052 : TIMED_OUT in 2h 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b1ef66f26f264eb420ce90d0237ae060bb6f5ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"date":"2026-05-12 22:10:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5a601d7211dd25b4783095788581ab2c0a904c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 00:36:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a0d14a725444cad8c3f174f28292e5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59bc2bf4c28f486998a8d8fa4960fbb7 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7930d3ed5ef4ba2a8af7d3a6d976796 : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c45fb3f1a59e457facaba35e7b976c83 : SUCCESS in 6m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b570007708f442dd933d2bab1b414b8c : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d30e7d7c2424056a67f03a22163a1e2 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3171e1e9c3b24d18956a0d16c61b6916 : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b438ea9f08344379db63dfcb443fd2f : SUCCESS in 9m 59s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/8460393e8f194f52b332317f86ab0c8d : SUCCESS in 4m 56s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/1e58dc15c5a8443b9013df9aa321e886 : SUCCESS in 5m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0a591520d6b4667af561243e28b1634 : SUCCESS in 3m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0154fcf18d146b4a0985db62a555230 : SUCCESS in 6m 22s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/05de750988ca40a18f38c5df7748aa59 : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03c3a65e0c0a4c2fb5ec66989cb399fa : SUCCESS in 1h 28m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f7d622b73d2c45828100869313a610c5 : SUCCESS in 1h 30m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3f7010d412e0466e92e46f0880721a47 : SUCCESS in 52m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7a66be40affb431785fa41406b183346 : SUCCESS in 1h 36m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4d38a9e3ada24deebac8c74f1c1416b0 : SUCCESS in 1h 07m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b529e499b96413b997041acdc1f7365 : SUCCESS in 33m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5243e6b7641d4299ac87eccc570d8996 : SUCCESS in 53m 06s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b93cfdd79013446893f6528606049be9 : SUCCESS in 47m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d7a237a589f42639157f19ce852eefe : SUCCESS in 37m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d78ea4bdb0a14cef88b4783dc1a4cda6 : SUCCESS in 39m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/43f202b3b6c2440d9a7c3269901c882b : SUCCESS in 2h 24m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5855873072374c75a37c5cbcc54803f3 : SUCCESS in 1h 30m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/05f01e0270c749b8ab685fbf24594fc8 : SUCCESS in 1h 20m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/675ba4b4b1794dae8924e3d5b9b95502 : SUCCESS in 1h 08m 02s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6bb7f6e74930aa7f75e52df20e0e8894f9a7d8ff","revisions":{"20e49164428e45ee89653192085ee6375433ba8a":{"kind":"REWORK","_number":1,"created":"2025-05-13 09:23:50.000000000","uploader":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"ref":"refs/changes/81/949581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/949581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/949581/1"}}},"commit":{"parents":[{"commit":"599873b01ffc0adadf94ecbdb791351275ef10f8","subject":"Re-enabling voting for octavia-v2-dsvm-tls-barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599873b01ffc0adadf94ecbdb791351275ef10f8"}]}],"author":{"name":"Daniel Failing","email":"daniel.failing@cern.ch","date":"2025-05-13 08:28:58.000000000","tz":120},"committer":{"name":"Daniel Failing","email":"daniel.failing@cern.ch","date":"2025-05-13 09:19:13.000000000","tz":120},"subject":"Add task in housekeeping to fix wrongly updated quota","message":"Add task in housekeeping to fix wrongly updated quota\n\nChange-Id: Ic40ace51297e72ae56a955f41c6bb3a22dd5b46f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20e49164428e45ee89653192085ee6375433ba8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20e49164428e45ee89653192085ee6375433ba8a"}]},"branch":"refs/heads/master"},"21b68ea6b891451ab2260fd62ef47ef6fe9e0126":{"kind":"REWORK","_number":2,"created":"2025-05-13 12:50:56.000000000","uploader":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"ref":"refs/changes/81/949581/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/949581/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/949581/2"}}},"commit":{"parents":[{"commit":"599873b01ffc0adadf94ecbdb791351275ef10f8","subject":"Re-enabling voting for octavia-v2-dsvm-tls-barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599873b01ffc0adadf94ecbdb791351275ef10f8"}]}],"author":{"name":"Daniel Failing","email":"daniel.failing@cern.ch","date":"2025-05-13 08:28:58.000000000","tz":120},"committer":{"name":"Daniel Failing","email":"daniel.failing@cern.ch","date":"2025-05-13 12:46:38.000000000","tz":120},"subject":"Add task in housekeeping to fix wrongly updated quota","message":"Add task in housekeeping to fix wrongly updated quota\n\nChange-Id: Ic40ace51297e72ae56a955f41c6bb3a22dd5b46f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21b68ea6b891451ab2260fd62ef47ef6fe9e0126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21b68ea6b891451ab2260fd62ef47ef6fe9e0126"}]},"branch":"refs/heads/master"},"d3f43a136ad31be34e6ab6d8f744f528fd59bc21":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-23 13:44:00.000000000","uploader":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"ref":"refs/changes/81/949581/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/949581/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/949581/3"}}},"commit":{"parents":[{"commit":"599873b01ffc0adadf94ecbdb791351275ef10f8","subject":"Re-enabling voting for octavia-v2-dsvm-tls-barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599873b01ffc0adadf94ecbdb791351275ef10f8"}]}],"author":{"name":"Daniel Failing","email":"daniel.failing@cern.ch","date":"2025-05-13 08:28:58.000000000","tz":120},"committer":{"name":"Daniel Failing","email":"daniel.failing@cern.ch","date":"2025-05-23 13:44:00.000000000","tz":0},"subject":"Add task in housekeeping to fix wrongly updated quota","message":"Add task in housekeeping to fix wrongly updated quota\n\nChange-Id: Ic40ace51297e72ae56a955f41c6bb3a22dd5b46f\nCloses-Bug: #2111589\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3f43a136ad31be34e6ab6d8f744f528fd59bc21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3f43a136ad31be34e6ab6d8f744f528fd59bc21"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6bb7f6e74930aa7f75e52df20e0e8894f9a7d8ff":{"kind":"REWORK","_number":4,"created":"2026-05-12 22:10:20.000000000","uploader":{"_account_id":35873,"name":"Daniel Failing","display_name":"Daniel Failing","email":"dev@danfai.de","username":"danfai"},"ref":"refs/changes/81/949581/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/949581/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/949581/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/949581/4"}}},"commit":{"parents":[{"commit":"d16fa5e723701c95cf97a3e9ea57b3856e09aaa7","subject":"Merge \"Add check for Octavia Healthmonitor network before deleting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d16fa5e723701c95cf97a3e9ea57b3856e09aaa7"}]}],"author":{"name":"Daniel Failing","email":"dev@danfai.de","date":"2026-05-12 20:25:05.000000000","tz":120},"committer":{"name":"Daniel Failing","email":"dev@danfai.de","date":"2026-05-12 22:10:02.000000000","tz":120},"subject":"Add task in housekeeping to fix wrongly updated quota","message":"Add task in housekeeping to fix wrongly updated quota\n\nSigned-Off-By: dev@danfai.de\nChange-Id: Ic40ace51297e72ae56a955f41c6bb3a22dd5b46f\nCloses-Bug: #2111589\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bb7f6e74930aa7f75e52df20e0e8894f9a7d8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6bb7f6e74930aa7f75e52df20e0e8894f9a7d8ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
