)]}'
{"id":"openstack%2Foctavia~949600","triplet_id":"openstack%2Foctavia~master~Icbfd8a97c99c3c5b5ac6e3ec27d41627e15dc458","project":"openstack/octavia","branch":"master","topic":"remove-wsgi_scripts","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-06-25 21:51:35.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-06-25 14:08:16.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Icbfd8a97c99c3c5b5ac6e3ec27d41627e15dc458","subject":"Add pyproject.toml to support pip 23.1","status":"MERGED","created":"2025-05-13 11:19:09.000000000","updated":"2025-06-25 21:52:36.000000000","submitted":"2025-06-25 21:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949600-remove-wsgi_scripts","meta_rev_id":"db2547f625b1d183e5c7420100f47526fe2909f2","_number":949600,"virtual_id_number":949600,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-25 21:51:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-06-18 14:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-06-25 14:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-20 19:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-06-25 14:08:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 14:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-18 14:57:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-20 19:56:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-20 19:56:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"a446907686f52ccbb0edcb17c99609be2ceb434d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:19:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74139e50fd9522bc37bebf741d8cb39bbfc2ac0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 14:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce9f4a717ca94bcfafecca65f36d242a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18fbb8535943425aae174c022e39b9e4 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6dba13e9c674625907cc6ee59732e73 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de4979f0d87b4526b0dd2e65d3aed78e : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/997e7469a1d74292aed7c86e1b9b5415 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80a61de655b743d1a3851829b75d73c9 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c669eb363d470193f3b92923ee8409 : SUCCESS in 12m 33s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cbb2d6f8d9824db1aa95ee9e8dd985c8 : SUCCESS in 5m 05s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/83d99c7288324a758ddadf1b40c45461 : SUCCESS in 7m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/be16ffec4a924502ad953aae81878757 : SUCCESS in 4m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2a36ddfcfdd49c29a8d27665e8c2342 : SUCCESS in 7m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ce04b43ef9c04702ae5714ea763344e2 : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59285512705e43ca9afb5158d5dda2c7 : SUCCESS in 1h 25m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e56e731ce21643978cb0669d4600a982 : SUCCESS in 1h 45m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/022f5c76c59b42c29e82175f923c2f13 : SUCCESS in 1h 08m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8be14cac223f4d33b8a936886f05690d : SUCCESS in 1h 42m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16eb60589d104bc6acf28fd775f42026 : SUCCESS in 1h 00m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/602848e2d4aa48a6a2ed0399d5be3f5f : SUCCESS in 1h 54m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/84852e2b24444f48b9772ee1d2f948e8 : SUCCESS in 1h 02m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/987faf1d75f348fab8076c95ae3d5828 : SUCCESS in 40m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/efbcf02dcdac4a7c82f017c1fed29a8b : FAILURE in 36m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04ba82e331eb41538cd43a740ad52b4f : SUCCESS in 38m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0410e1420a7944aa9621f33bb099d251 : SUCCESS in 2h 32m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2320df40e1540589313463792534dc6 : SUCCESS in 2h 44m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ddc4453d1e0a4c3488c4887a6635a7f1 : TIMED_OUT in 2h 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c88aa74412418bcf7288bd408737b2930e1033f6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 11:46:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e992161ec9633d5be84ee30904e17e1eed89f1f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-18 14:57:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4320ebc698f25c83b07066de21d781025532379","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-20 19:56:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e2cc62ffe5515a8b8f92893ea9651ab248e9e262","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-25 14:08:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58df0b17dfc35d7372e7c0d66d3e235f55840110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:08:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f416675e56ca08846a4e5aabac66758755dd4a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:37:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ab83a382b4a4a39836eb9dfc5093704\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ebc8ff62c84db9a692c829894dea8f : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ecd84e0c5e945a3973e53f05f62dd1c : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b362b19b6c24400b9e94a2978ea6457 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df854bbb1a2b4835897731f6146fda21 : SUCCESS in 6m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ee316f0c10243d68ba520b45c99ad37 : SUCCESS in 3m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b8b5a92506554698a61e506dbdeecff1 : SUCCESS in 4m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ca3e8b5d5a44632a11210581ec2542a : CANCELED in 9m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7bf2cc46f1b946c69dcc79eafa89da57 : CANCELED in 9m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/14506fad91a34d83b4a28eb10225a200 : CANCELED in 8m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/776140372ada41d1830c89835975754e : CANCELED in 8m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ec369c65242d46e59b227fd74fe710c2 : FAILURE in 9m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f738f593b55b49190f093d9e043538bf6760403a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-25 16:38:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00e2fc6b69a26e0aa6132ee201548e0bdfdd9817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 19:14:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/385105fde689419b8796c571f5d908e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdbb70ab9fca42a481dd55ffd8f7c2a3 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf2ce558af349b6b5c624a99306343d : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37a0b5469d3746df85f3845b3955ab0c : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4995eccd949b49899e67f5ad1537e3db : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3214cffe852a4f51953d2e1f6a52a5c1 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed6310c20c54d65af86994464f1260d : SUCCESS in 11m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/376edd06acea4bd3bbc8f863a20ce3c5 : SUCCESS in 5m 26s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5e0ae92395a6465f90ce88a86c9b12f2 : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4d16d7a230934ca6882966464cd9445e : SUCCESS in 3m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5b1b4d6744b548fa93463d27bdb74240 : SUCCESS in 8m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1839710c006f489a9c9f2adbf064f78a : SUCCESS in 3m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/658ca2edee494ebab579bf4234a589be : SUCCESS in 44m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8c8a32a9619148e39547003d448ab9e1 : SUCCESS in 1h 57m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/02af893d10f54ef7a14261dec5c9beac : SUCCESS in 1h 11m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/681623b1076145739d55847997565c8d : SUCCESS in 1h 53m 11s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/396855bcc82a411a8a7d1790a9bb4f8f : SUCCESS in 1h 07m 59s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b7a937c073114be8a8e977d71eaac93f : SUCCESS in 2h 02m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2dd29136637540ad9238e2d8a72fddea : SUCCESS in 1h 08m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa72191e579b4a4896ce47fb2dad275f : SUCCESS in 43m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/752ea6f8f5d14633babb248662c881d3 : SUCCESS in 50m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f83a9a1a537401f9ae06d1e898bb34e : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ea1af651f6be4964a6e50cd521a3b837 : SUCCESS in 2h 34m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a5833eeda1604f3984b8699cd9b3fb9f : SUCCESS in 2h 32m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0ad1940619ae4b0cb6cee44f0c9f4398 : TIMED_OUT in 2h 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5f98aca8ebf44baaeebcaa888ae0d2ad190e4be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 19:15:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aaf7c7def66a3485d7a6ec308741b15f8eafa56f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 21:51:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2f0bef648bc4d44b9989e5638d696cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d89dcea836694c5e98c6cd59061b6b59 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fad418cf6410484aace073ed06dfb67a : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd15c53e5f5b4740bcef0e6b96270f98 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c090532f4d45e795f3df139f7626d2 : SUCCESS in 11m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a4fd728dbc5a455e8793ea5bc6d6b637 : SUCCESS in 3m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/91e539491543406baec33012b5d0e247 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a101e4f37e644719cd26d2ada2058a7 : SUCCESS in 1h 42m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9990f34b001e437ca81c7cf6ec7dae63 : SUCCESS in 2h 30m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/273cb78942364d3c8b939c39d616d2a0 : SUCCESS in 1h 07m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a7c0ac9116784c0ca9846ea0681bd768 : SUCCESS in 57m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/71c9679b7e7a486d8daa141ac564c3a5 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"17a8eb33d0bec6f6300862d844f48c0ec2aa2350","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 21:51:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db2547f625b1d183e5c7420100f47526fe2909f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 21:52:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97be26d98ac9466b8d1d48edbeecf61b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203cbfd9752b483c855c46b63a182706 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac5b939013f84fab41ef356aaf8587e012c38eef","revisions":{"ac5b939013f84fab41ef356aaf8587e012c38eef":{"kind":"REWORK","_number":1,"created":"2025-05-13 11:19:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/00/949600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/949600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/949600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/949600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/949600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/949600/1"}}},"commit":{"parents":[{"commit":"599873b01ffc0adadf94ecbdb791351275ef10f8","subject":"Re-enabling voting for octavia-v2-dsvm-tls-barbican","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/599873b01ffc0adadf94ecbdb791351275ef10f8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:18:53.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:18:53.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip. pip 25.3 will go a step further and enforce presence of this\nfile.\n\nTo address that, this change adds the minimal pyproject.toml to enable\npbr to be properly used to build editable wheels.\n\nSee [1] and [2] for more details on the removal of the fallback support.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/8368\n\nChange-Id: Icbfd8a97c99c3c5b5ac6e3ec27d41627e15dc458\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac5b939013f84fab41ef356aaf8587e012c38eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac5b939013f84fab41ef356aaf8587e012c38eef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
