)]}'
{"id":"openstack%2Foctavia~951402","triplet_id":"openstack%2Foctavia~master~I5dd0754757cbd720ece557dbe59650c7535a5286","project":"openstack/octavia","branch":"master","topic":"remove_py39","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-06-20 19:44:29.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-06-20 17:39:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-06-20 19:44:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5dd0754757cbd720ece557dbe59650c7535a5286","subject":"Remove python3.9 support","status":"MERGED","created":"2025-05-30 12:47:48.000000000","updated":"2025-06-20 19:45:32.000000000","submitted":"2025-06-20 19:44:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951402-remove_py39","meta_rev_id":"7ab6c314a52f93a9ae1100f69ab83f671dec8d32","_number":951402,"virtual_id_number":951402,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-20 19:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-06-18 15:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-06-04 15:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2025-06-20 17:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-06-20 17:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 16:17:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 22:02:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-06-04 15:47:01.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-06-18 15:46:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-06-18 15:46:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-18 15:46:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"defcea2c2de8d6c65c4f118b9ef6ec56e298f3be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 12:47:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54e1e20745ce765d9bb7c61c2f43c27a85b3ce34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 16:17:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb8699e7bd584f74b4c67ae892c64107\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04c63d1bc8ad47c397707623e6244f1f : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64483d70d1f946aba4a9465706344949 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/648abfd3a2a641e7b796e827112a1cdd : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b96d1859c437456ca347acfaa6a63da6 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8388e203659145ecad65c5a71bdceb8a : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6839ed2970c549d09d37cc9b0ac714fd : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f036aa991ac443b2a8bd088b9fb5fa27 : SUCCESS in 12m 27s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/343ba9f95b2143d6bcb6353da6de3445 : SUCCESS in 5m 05s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/54193d71864c4f31b31553a41def8eb1 : SUCCESS in 9m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/32ecca4822d244fead7d0e571398d85c : SUCCESS in 3m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0e116abd5f844cc1a12090ed893cd992 : SUCCESS in 8m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0e9356ee89784b6a8bae71352385b317 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0880b46fc78e4fca9d540efc27203842 : SUCCESS in 1h 28m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/34837542f54942f9a446fc7b247332c8 : SUCCESS in 2h 09m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1560f02a586c4f7ab767a6b75d729170 : SUCCESS in 1h 14m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f2e876094694b58b45d7657b6768f60 : SUCCESS in 2h 15m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69982a48043d44fd997436bcdb61808c : SUCCESS in 1h 18m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/898e8c8425d24e9789b101410f221482 : FAILURE in 14m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/15c3c702744b460a92a6ff3fcc938dcf : FAILURE in 14m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fca54c9d7ed645c7b381640304190d77 : SUCCESS in 35m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d14fbf6111d41c790885de9efecdde4 : SUCCESS in 34m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7cb6c5a894564817944a1667374b18c4 : TIMED_OUT in 2h 32m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f54cf6785be645c688dc6ed8400a6659 : TIMED_OUT in 3h 03m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e20ad956bb4740a79f58564c6220921d : FAILURE in 2h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e18f3bb1cb67bd8b0565c6b27252184f8f5350bf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-03 22:02:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad253659c8ef2eb4cd27774a52d7fd7f41ffae0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 01:07:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4c46d933f84aaa9e69ed66e2fb31d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2575c26258ad4a0b9c41699ad010e2e1 : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fea577e96ad7426f8556f136cdf6018d : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2e4ee43e8234c438a454b8940e809e5 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06997d759f624c748d2c159c64ec7e9b : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c2e9a71f7bd4c3f9306031558581f29 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/631b340d013d4de28bb58902ba420bb2 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4798f6730414b4f94104e1b1c1a0625 : SUCCESS in 13m 15s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c18193d5488e4be9b260bf0a01e7667a : SUCCESS in 4m 37s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ff65f426db8448cbac9751514123a1ea : SUCCESS in 6m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/455e6347158049d79fcb6b87c5a719f4 : SUCCESS in 3m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1b7d7d0b6af46508a9cc3d84f7a3391 : SUCCESS in 7m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/effcae0c8c104ab3a23eb404804b7a67 : SUCCESS in 5m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c22eebe637ae4fedb90edb3206ba8e0e : SUCCESS in 1h 34m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2d4891b7fee347229b46faa6d61de8dc : SUCCESS in 2h 03m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/317acf1f64b948cc8b4078d52738a829 : SUCCESS in 1h 08m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/579dcae8b38349abbf9d4df31545dab5 : SUCCESS in 2h 03m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc69ddf4a2740519f370e1a8383219f : SUCCESS in 1h 12m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a88cf28faa27434cbd14db6939a53a7c : SUCCESS in 2h 07m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eb57bbbe7a7046fb93f1b831a71eeb1b : SUCCESS in 1h 22m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/284f01fea3834537b335c0694565634e : SUCCESS in 56m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/097bedf47d0245f5bc2b556f37614998 : SUCCESS in 34m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a21a0c80da749aa8b38486d02a626a5 : TIMED_OUT in 2h 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b3e9e88bd71848c18ee6aae9cdf676a4 : TIMED_OUT in 3h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ffc3bc65e58047b7a3c0d900c9ceffef : TIMED_OUT in 2h 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8df62b781bccf58c1ea96022efe631c7d0e776ae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-04 15:47:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b97be073d25c6d03d61ff4b069d2e0b99603184c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 15:46:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c96d57e77cd1459ad727b491f69269bccb4ece9e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-20 17:39:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9751f7653599ad7432d8bc538c020e27ca5706e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 17:39:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0fa0b3a60f732570295fcc4c6421a3bb23ffda4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 19:44:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559c291b030d449aa48a8c43f45071a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e896f4c3e0b8453898878d02ec881371 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580d601ee50543dda27de401e68551c4 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faccba2211864047b45be3d5faf87450 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65778e00a61b495cb51dd13394f13f9d : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f4d897998164b2e9d409617a433279d : SUCCESS in 12m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/37202f52f9c34292b813552bbdb54e35 : SUCCESS in 3m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/635f78558a054117b3dad58cb600a93c : SUCCESS in 8m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/960c6736a5594dd4ada53c3fad470c8f : SUCCESS in 1h 29m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ab6ecb1b5737461ca3fb2514fb7d5d74 : SUCCESS in 2h 03m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bc43a1123857437bb88684ff1377e8e7 : SUCCESS in 1h 07m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a769f6234cd46a382c306089b0528e9 : SUCCESS in 41m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f06808ad8d445afb592bb36a2c87738 : SUCCESS in 1h 08m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0aaf6cb03014562f6282b18a8b0a80398bb77ea4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 19:44:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ab6c314a52f93a9ae1100f69ab83f671dec8d32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 19:45:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c39af8e5548c4d0ab7d142fe42dc8a55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c71d95cc3e548e4be495c47ee5d0622 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7456719c889d52f322eb4471d9f951d7e09b07b9","revisions":{"7456719c889d52f322eb4471d9f951d7e09b07b9":{"kind":"REWORK","_number":1,"created":"2025-05-30 12:47:48.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/02/951402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/951402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/951402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/951402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/951402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/951402/1"}}},"commit":{"parents":[{"commit":"794fc75d19bd8fde09029a6c4bd497a55ecd2da7","subject":"Fix functional tests due to oslo.context change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/794fc75d19bd8fde09029a6c4bd497a55ecd2da7"}]}],"author":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-05-30 12:45:51.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-05-30 12:45:51.000000000","tz":120},"subject":"Remove python3.9 support","message":"Remove python3.9 support\n\nBased on [0] python 3.9 will go EOL before 2025.2 release.\n\n[0] https://review.opendev.org/c/openstack/governance/+/941141\n\nChange-Id: I5dd0754757cbd720ece557dbe59650c7535a5286\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7456719c889d52f322eb4471d9f951d7e09b07b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7456719c889d52f322eb4471d9f951d7e09b07b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
