)]}'
{"id":"openstack%2Foctavia~951487","triplet_id":"openstack%2Foctavia~master~I3ab110ab4fab8b0e2240325e6f4e29a572cfdb78","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-18 15:41:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I3ab110ab4fab8b0e2240325e6f4e29a572cfdb78","subject":"Remove note about old pip\u0027s behavior","status":"MERGED","created":"2025-06-02 05:35:44.000000000","updated":"2025-06-18 20:58:05.000000000","submitted":"2025-06-18 20:56:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951487","meta_rev_id":"cc02cf9d669b5e1386ef65ef9e5992fb9f232eac","_number":951487,"virtual_id_number":951487,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-18 20:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-06-03 22:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-06 22:11:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-06-06 18:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-02 08:16:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 22:05:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-06 22:11:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"b42a141b93b83e925b19a4026b0f706ca0177aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-02 05:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29365f3ef5db4f89b4998acd51bf3dde013fc586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 08:16:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/944b962bc0914e77914bb97e65340881\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/53a4656c941840a884248d6821ef69cb : SUCCESS in 4m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/633f4c0137bf40469e53eba50eb4dd92 : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5977a5eaeeef4626bd05d331eb128de4 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482e258e423549dfb345d43723f3e4e5 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37af0014cc44401a967a403c3112a362 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/865cbb73a4b4463f87c7dd6fe19fe863 : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f29f03d760a45deb03a55c8a595a294 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f589449ceb0543709661e6e85288eb3e : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46989af75076428cbc3ae4f2087634e1 : SUCCESS in 6m 04s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/aaa184f4dbd84f578e8f54584e1397e2 : SUCCESS in 5m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ec44b003a63a43a7a27e1104a1e7dc30 : SUCCESS in 7m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6d0a67e8c6c246629bc68af9e6ae1f5d : SUCCESS in 5m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaeba02815fa40789c4bfc9c6022b961 : SUCCESS in 9m 09s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8a09f61c45f94b29b784e103c42d9486 : SUCCESS in 9m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cf776c754e0b40e4aa0211363090dfad : SUCCESS in 1h 17m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/14c28a69fa124d5bbe6314143f0687e8 : SUCCESS in 2h 21m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/faaf4e735f264a73bb423772a2e31f02 : SUCCESS in 57m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e90715b67a2438186e8b73be4859608 : SUCCESS in 2h 26m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c206b3292b342ee913fabb9d45d09af : SUCCESS in 1h 18m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/481e586aef044420a02f805175a735c4 : SUCCESS in 1h 51m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a90337249cb54dc8b596176198642d5a : SUCCESS in 1h 24m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b003bb2ed9d342c48e7ef46e2d115f58 : SUCCESS in 32m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/790ca095aa454abfb0611b053d589e02 : SUCCESS in 34m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27b70327a0ac4c1d9bc1adc9bfbeaa5f : TIMED_OUT in 2h 33m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5cc4ff43b6384a3094fe8d9d91997d55 : SUCCESS in 2h 35m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ba6b9d241c434d48b79eb2472f8f1d49 : TIMED_OUT in 2h 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffeb39c2071cc249e92f620381b33f5e77048007","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-03 22:05:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c7dc1b79ad55ef66b9963f09277480f7b383f13","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-06 18:59:24.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03dccc8e8c076b6ec604ea3bce7751736c84099c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 19:00:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4e8ac26a0f037bf9d8e24b7585971bd2b2ebd58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 20:09:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/818e571d65fc4d3fa681d27a5854bbfd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d8200f6d83f44e9814c13088a9d2b28 : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fd2cb2cb56ef43249a4b1157788bf999 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50ced1b6833d400eafdd5bdda1322051 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7a493e3f1a84e18b50f699dee7bcf0a : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29d3d559b2004b67a47fe4d07073cb97 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcfb00f2cf67471cb7ebcbfba1a9a317 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b94920eefb7f46f59ef87457f6ba8b94 : SUCCESS in 4m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aba1abca76704a9cba6e4893cd7b9304 : SUCCESS in 3m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a1381b5c987f48c39f98e75336587a4f : SUCCESS in 5m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40569aac67964ac7932c8836f332e5b8 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/05134b18645a4886844dbac1b58c6b42 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/60821165ee8b4c549a30397d7bd227c0 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a58d7038791443f794a97381efcdf3d9 : FAILURE in 35m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/faf121d2d2a64e0293e86dd23a247cfb : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b11b4c4ec85248f89f4b854ee3e86f75 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/441a438550a34b62a6d3069a67d4757b : CANCELED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d81268036e76e9287b64a0a59cbf05bfaa6600a1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-06 22:11:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83cd8ce7c5dd2534e4635ea55c74adc4ba195ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 00:44:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c247052715e4e13bbf4397157143869\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8b3c3a99bdb04f3aafb41ac166e5da63 : SUCCESS in 7m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8920cc18646f4e2a829757fdaf3b03ff : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79b2c426774548438f87504075b7b5c0 : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4ceb8f364b4423893c2296910c95db : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1acce0ea897b43e9bf5b08d6c31b8e34 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25b13454820f42688e621185397d6fe4 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43b7ef6ba9b941798bee05261aaf8662 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a742ba07ec410cbcfd04f8aa5c993a : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bb5e401365b4c039e76b24cfa49d19e : SUCCESS in 3m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0f29c455305c4934916122f0abcd1920 : SUCCESS in 4m 59s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0611733af7824d22ac5692945149f3e3 : SUCCESS in 7m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bf08c387350741a494df5e675368190e : SUCCESS in 1m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/996a4d534ab54379ac1a168bfc0ab58a : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dcb1358c7a6e4df08dbca2db541ac218 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3eb602c5ccc41588323c57705a41591 : SUCCESS in 1h 15m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f0270779c40745c1aaa7d86de08c1518 : FAILURE in 35m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/080329b541974b6198d8e40c0cb10c5d : FAILURE in 1h 41m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87521920b7c34d9d93eb1158119c6b84 : FAILURE in 35m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12a7f718fb8f436ea0ef3251beb4602a : FAILURE in 1h 48m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6b2773c100394e1dab711a96a697fe49 : FAILURE in 35m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/89f2b5dd49274730aed1bccf3bba0b82 : FAILURE in 2h 23m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29fd84b78260481a82189037269e6bfa : FAILURE in 52m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eeb911c60e044c0388c75fe677d30b11 : FAILURE in 41m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce04ee3b9af147f780c26bcfe73fe3b2 : FAILURE in 1h 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4673cab18e24d3f8ff66a2be00d9421 : FAILURE in 2h 31m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/23a9f9b05b3f45a08b17db969c7d9575 : FAILURE in 1h 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d7a782479c2fbc1ef52e9b3b13bef871889fc62","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-09 15:51:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c89eb49ceb9cb1371011c5579d94eb88b4e09ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 18:37:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/47964371452a4aeb93a813658bfa7d3b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a621e88dadb54634a2e26211efb950a5 : SUCCESS in 6m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9fd3920c537a4311aeb005e9a7d304b5 : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da5cfce9826345f0b34315b003618728 : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a38d97e8b8d947c99d73aaa3cddfcc1c : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35b8566819cd4843b8cedd0246b0c714 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b18fdf12ef7149a790e220f270a3acc9 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d58d3b3c0b546f899847e29db85e8ec : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3b2fd31ad044aaa855469e71e195f0 : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c463b6468ff43f99f6b4398749971e6 : SUCCESS in 2m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/972f5f6eadea42da933c3f272cef6434 : SUCCESS in 6m 07s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/75330911b1364bc4893b12df881dd90f : SUCCESS in 4m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c7fbcf38c13f4b10b473a5622eded26b : SUCCESS in 4m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8dc0701729e6496ab7497dc4841b749d : SUCCESS in 4m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/05c8d57a7a5340cf878847a44a4bb12c : SUCCESS in 8m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3231e57f735e4253a34c9824207d2a31 : SUCCESS in 35m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/854bcd01e9ad4c718e0cd6c20792b383 : FAILURE in 45m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/175c9d6f0b54497ca630afd549f807c0 : FAILURE in 2h 20m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bce3478f8254ca2a1f3347321768cb6 : FAILURE in 43m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cb0aa3ba2bf44c29cc20c6fcef39e0a : TIMED_OUT in 2h 32m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a716ea57b0914bc7a7544dc8b1fe99d7 : FAILURE in 46m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/78f0fddfc053484aaae4d26004e96873 : TIMED_OUT in 2h 32m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a85a8091753141c89c69e9814b9b3449 : FAILURE in 47m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fb92d03e1784d3c97b0cdc2145f0c28 : FAILURE in 58m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fa60d038520a487ba49f41e9c4139955 : FAILURE in 1h 42m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1171801bea0c4acabdc81df318e5a1c3 : FAILURE in 2h 38m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9c9355d187974dc4a8ad16bfffff0e9d : FAILURE in 1h 56m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6ee69b80d05ea69e5cb8da4468c1e2b8fe34fc6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-17 16:14:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f09ca1313b617fe754af9383f50a16f27ed9a6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 19:21:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b508eeadcb34bc1a0b17ede88c32fa9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5807dfa0e1ef41dc9f95f2ed1c64b2cf : SUCCESS in 6m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b4db053f0814e57a26eb5133c798663 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/032661e646a342959ee3c09b229c3239 : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de04faa347948d9860cf22a7bbc2a52 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e671000fd8f49e09db0ab42c2a084dd : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b15d819dd2414bb6ba31fe6868d3b78b : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/989ccdae8add4094a668f9aeb040edcf : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad5a956c15e45949f31b64a86f5cff5 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/229f312c8b3b4dc3ba30416d90944f26 : SUCCESS in 4m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c6006b0b6ce14cfe8f175eb53fec5027 : SUCCESS in 5m 28s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b8468c5dcdf848cf95e59860e8234ddf : SUCCESS in 6m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0a7f5ac5bf654b0d8da019910db1ff09 : SUCCESS in 3m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e7764d1dbf7c466ba803f54fc296e652 : SUCCESS in 10m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16e8a0b773134b49ae65c594b1c7dbc1 : SUCCESS in 9m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29b81c9615bc410ab539b1b44dec94e9 : SUCCESS in 1h 27m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/05b63193558d4ffe8ddf865d043a26ac : SUCCESS in 2h 24m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/de3bfc7456744cf58ae11dcb93324144 : SUCCESS in 1h 34m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d75c68c67374e7fbf55569ad607c28c : SUCCESS in 2h 27m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56c17181e1eb44f59d698cc4024cee4c : SUCCESS in 1h 05m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9c6807cd9fdf456482667e6880fe9a89 : SUCCESS in 2h 29m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/86345b4ec27e4187a0fce24daa5e2700 : SUCCESS in 1h 23m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60f34d2eee724a11b210eddb4f446355 : SUCCESS in 37m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e27489fe7424f01a34bc86f9aeffe7d : SUCCESS in 52m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5607fd4945d045e59f52a7e51aaae03b : TIMED_OUT in 2h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e23ab89d446490c94ad89bce0ec010a : TIMED_OUT in 3h 04m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1fffda55fe384d738be49519306782de : TIMED_OUT in 2h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d14fb94196ec7eca39c5e60f5a739feedfadda9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 19:22:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbf1719d47c1b6e4235654924cdacb5281a9e83e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 20:36:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b77eab761ae4cfeb4387682ee7d1661\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d3524be2ce248aab61c68620f97b428 : CANCELED\n- requirements-check https://zuul.opendev.org/t/openstack/build/2d659eee61f94378bdf19473d3733eee : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/741c934e16044a5b9af5ce98ab19bea7 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8b7f6c2979a4a6fbf24246d623ce4aa : CANCELED\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7cc982fb6364519a7f7c55db2eae270 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffb8e7f184d44c281dd985a60a1369e : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c71c84332c846858870b42a816b8602 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/200ec20ddcd04c0b9dbf1a43623bee48 : CANCELED\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/606203f6d1404aaba3041238c44de2a1 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e67a93fd8a74559800d94a0ca98a402 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/963da1f2250c4dadb374215f04549f8b : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/724f3711f7544c7daf2cabf895c76849 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c40b75d17f8d48d7bbd4285e3dfdbbe6 : POST_FAILURE in 8m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a694d1c14d54beb9660dfafb7b3396c : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/733030bba07c48dbbbc6689cf264bc35 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e61f28b7e6054d9a943d673aab3e158c : CANCELED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"687a586374e7ee4d52da36186e87ed497f486639","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-17 21:28:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fe7f945bcf665ea98e3a1a2432a22d785040bec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 00:27:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14e41847aa844a2290f539ee7eeb5e58\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de8fcad3ee2d40bf86b5dbf1f5545023 : POST_FAILURE in 5m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8695d4850fa441119ea90a995e94a37e : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/189bb8a514f8465c8f037c749cacbb25 : SUCCESS in 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8695b08db3e408db6178c2d7dc9f3bd : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13777d123af94e7da9fb892914450c12 : POST_FAILURE in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d849c3b5cc74c4b88ba92c773c101c3 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef3bd34d80f14fce90c95ac929a64d68 : POST_FAILURE in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678cf671dcb0479b978098728444cfd9 : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b89ad5f38f545079a06f87f8195794a : SUCCESS in 21m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a6aae5d36bf64a8da2b3014ce2da5a13 : SUCCESS in 3m 59s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a45709022f1141168f337bf7d73b273a : POST_FAILURE in 9m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cbb8d7fcca34d8388bbaaabe34b971c : SUCCESS in 3m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ccc0aee6d23c4dbeb79ba99770c39f56 : POST_FAILURE in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d1400b2127d34b20b6c93b9c4e039b77 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f73bcd4ac514455b9e8611ace5880761 : SUCCESS in 1h 36m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/33aa455c69c340d0972b4e64996aca5d : SUCCESS in 1h 55m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/29af29eafdb842a6a12af1370d8585d5 : SUCCESS in 59m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86ba35010e284d5e83beee92b43bf5a3 : SUCCESS in 1h 46m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c71f80d6fe0442cb4c757edddae63f0 : SUCCESS in 1h 03m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6a3ff49b181b4a36be2a84c91677492c : SUCCESS in 1h 53m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aa000939ce104b9282d65bda30931981 : SUCCESS in 47m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/938f547305044aac90592dc607b96c01 : SUCCESS in 34m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8daf69eeb7704ff2bfc34e9c53db69e0 : SUCCESS in 35m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bb7e156a23e0432b83cece928783bf2e : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ffd91878a4cd40d499f62cf24d647ff8 : SUCCESS in 2h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cc67330e220a4efbb4a34d9626e2f674 : SUCCESS in 2h 09m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab1418544687f6d63cfba678529c2aed38504e84","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 01:13:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02792829c7d0cd636c52c8b5a49b3c9b60a6d6b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 04:26:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b05240afaf4a4da1a63bc0c018bc218c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7dc307dda78549af9c45fb726de70a75 : SUCCESS in 6m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/68ea6187795a405daf6df5e3431ccd08 : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ae0d65cc2674d10aeb3802e9c8e51c1 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d433c5359047a6985f38da79d13b1c : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/747cdca9c69c4938a5e1946239b2b53e : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/244ff393530c4969a8f0095cadcb5dfc : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2bbde379b8a94dc78e40c4a3b36790ca : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5726580fc8304e90ab00ea7aadb93da5 : SUCCESS in 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82878cde3092428d9976f225ff59c2f2 : SUCCESS in 5m 31s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4dc4636d0fd945248207f30d9cd762cf : SUCCESS in 7m 36s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/73979f115fcb48df8cbab491fe8c1134 : SUCCESS in 7m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/20b694bcfd294943a8c2e163b8e161b6 : SUCCESS in 3m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ecad736a6b904ddbae94b02eaa92b5ff : SUCCESS in 7m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9c14d853e2ca49f6a4ea8c70db881f72 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2ec95aa3e696400091a69b5bd67e040a : SUCCESS in 1h 44m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/27b3f354accb4dc6bf3064e0ba0d0ee0 : SUCCESS in 2h 28m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1a91b60fd342406fbd6eda6f9b11e91d : SUCCESS in 1h 24m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c98f500b8f8d4ca4bdb3675f653c73eb : SUCCESS in 1h 46m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4424740ec5d74517926a00dd7be25d86 : SUCCESS in 1h 03m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/00d52fb6c385478483c952f3264b46f4 : TIMED_OUT in 2h 34m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/281c0c546d0d4bdabfe841f42e771a29 : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c2638efb52b483fb5a6babe1f4e93e8 : SUCCESS in 43m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b94df6840d6647d4a5da44a8b5a65521 : SUCCESS in 42m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73d07ea1759443939ba48c29088ec267 : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/adb82e1dd88f4577a2dbd2f05f8d0c07 : TIMED_OUT in 3h 04m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3d8eea93bdcd4729b89137464f03cfd2 : TIMED_OUT in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8747dd4df51247ba9ce0f72723b8f30f43acefd6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 15:41:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c084978317e700d82e1b640ed712236ba04faf24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 18:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b38d888609504806b75e63717fe647fd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6be287a098b647b4877f93ddae19fb97 : SUCCESS in 7m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1383a9a34434844b93e9a78e4bfb9ff : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1bb047a2131456dba717abdf9c20dde : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/341a0aa451cb4d248dd94dbc97d22573 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d78ab6c88cb49eeb1e32e34bf692671 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6431282f182402ca60fcb29be3c15c9 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45961072ee4a4045927fdf16a9ea77e9 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9428bf0bd44e6d9e66dbb07631c5d7 : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd5c9710c17d4079a32e4d6e840e140a : SUCCESS in 4m 26s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d2098239be144425b692e32f15241c18 : SUCCESS in 4m 59s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5e74acf9e3a64877a24d7a41fd1f7a16 : SUCCESS in 7m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/89c5f50fca3440de887995e22e5a8637 : SUCCESS in 3m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/16df113d44484f1e86e17b6e161938f2 : SUCCESS in 7m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8ef4ab41605a47739616bb440cb67559 : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed8976345d384c94b6531417759713f4 : SUCCESS in 1h 24m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d0dd532a57654c15b1b5093fc66c2958 : SUCCESS in 1h 51m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6c35e1483fc046d985f1cf67aa64ff3f : SUCCESS in 1h 02m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58b93cc90c754f508cfe853a542ab073 : SUCCESS in 1h 42m 15s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2830a9e39cda48208ab5ecfd3e7f3a21 : SUCCESS in 1h 00m 25s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac86d37186f04b05a612ea3c00891a58 : SUCCESS in 2h 17m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/70e169e4e08144eba309dec822542e5f : SUCCESS in 1h 32m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3fc6cc6b31b64060b8dd029f4730086f : SUCCESS in 43m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b56cb46091364b398c47dd7c7226d51e : SUCCESS in 43m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7eecedba02554bbb8fd2619cc8218067 : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a41407b2c0844c7ea3a6ec53db2df2ff : SUCCESS in 3h 02m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/006d6dfe97a644c78d50bd78a0d30c57 : TIMED_OUT in 2h 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b96238872d65e6c3ef468a6410396ad316118449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 18:49:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"629ce962ad7bf68b893bc7b45a90cc6f1c1bd1cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 20:56:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b72779ae45244ebb905c30f0e062dd3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a13424c83a4649b68498ea6e4d73a0e4 : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/21ad8df840dd4ca8837f633ddbe342cb : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95cf314184584bc18affbe7d002ce183 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1459dbbace3d4c15a1fec04b73ad0b66 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a42b17f5841f4071a3d0e3a873b0db68 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72ef8456e7ea4ef3a854c8e40b2fcce5 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0662700cde2240df8b61b3afd2945426 : SUCCESS in 4m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8777d4e6503745aab22b8d0c9d9985f4 : SUCCESS in 3m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/efc1a001b42d4f6ea20c7d02cce26fed : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/33cf2147aa68443ab863ae74e956af83 : SUCCESS in 1h 19m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/50b2378333444de5867188dc7c9ec470 : SUCCESS in 2h 07m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd2f60779abc4301a2c5494398346843 : SUCCESS in 1h 13m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7deb8880618f426b8b55845daf6a41bb : SUCCESS in 41m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb59f0575c454d42a79b4ae77f049b8e : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7cc97ee92444099603e3bbe3ccc5c25e10df953c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 20:56:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc02cf9d669b5e1386ef65ef9e5992fb9f232eac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 20:58:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713c53edcfca4526a4f286795d31d7f8\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c46fa1872a4b4140a4ee23e2cc371ac7 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7121c73b6a4b4206a4b45d16ef66a571 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd6671f2614b4894864eebf7f3da0c30 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c79e502528470a1567e541a293827109ecedb2a4","revisions":{"c79e502528470a1567e541a293827109ecedb2a4":{"kind":"REWORK","_number":1,"created":"2025-06-02 05:35:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/951487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/951487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/951487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/951487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/951487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/951487/1"}}},"commit":{"parents":[{"commit":"84f512687a6a1995378bfc5b13753ef4a145601c","subject":"Merge \"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84f512687a6a1995378bfc5b13753ef4a145601c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-02 05:35:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-02 05:35:32.000000000","tz":540},"subject":"Remove note about old pip\u0027s behavior","message":"Remove note about old pip\u0027s behavior\n\nRecent pip can resolve dependencies regardless of the package order.\n\nChange-Id: I3ab110ab4fab8b0e2240325e6f4e29a572cfdb78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c79e502528470a1567e541a293827109ecedb2a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c79e502528470a1567e541a293827109ecedb2a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
