)]}'
{"id":"openstack%2Foctavia~951732","triplet_id":"openstack%2Foctavia~master~I9b4860de49c9b54ec0403ac1e682f269846e1711","project":"openstack/octavia","branch":"master","topic":"normalize-wsgi","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-25 04:19:11.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-11-25 02:37:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I9b4860de49c9b54ec0403ac1e682f269846e1711","subject":"Refactor uwsgi config management","status":"MERGED","created":"2025-06-04 02:17:18.000000000","updated":"2025-11-25 04:20:05.000000000","submitted":"2025-11-25 04:19:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951732-normalize-wsgi","meta_rev_id":"b87292ba253d2817d6d4de3f5c456412505fa340","_number":951732,"virtual_id_number":951732,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 04:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-08-26 13:04:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-11-25 02:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-09-18 13:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-24 13:49:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2025-11-25 02:37:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-04 02:31:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-04 06:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 13:04:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-09-18 13:53:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-11-24 13:49:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"516a147494a03359791e3a8ea4a124d88aa0a099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-04 02:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"998b63bfe059f69d0275c8b47b42eb66a6007b0c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-04 02:31:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90d706f6f10c4eb8c82ae23117972fc442d42c62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-04 03:03:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"943833f353573b825a7eeb71974049d9c6e9ce26","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-04 03:04:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97d7043808456131c6fa98fb31ff576441aeb350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-04 03:05:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d171453969320bee7db07d448314a24e21f386f4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-04 03:07:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b16b28e3e88b15a4b2a13974b96a2a7dd64dbe43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 06:13:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21dafd1a9b8c46ad85e88b42de876561\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a482c4a54dbf44f9b2568679a95a571f : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03d5253106e0441da30a1556fed2231d : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9984c4f7af754ce38d8d9ca6ef43664e : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c301b37c457949868824efa8f14e8ad1 : SUCCESS in 8m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fbb62a026ca4cdfba6aa9f397d0c1db : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb5bbbbfd124e5c8c973288a1627252 : SUCCESS in 13m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cb277994278446a0bbf482b367e30c45 : SUCCESS in 5m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ad1efe46378f46ae98c1811d733356ac : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/65d39baf11ab4db69469380091c58be3 : SUCCESS in 3m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/396b3fb425ea4b3d9f40c341235e7016 : SUCCESS in 7m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0ca045651efe44ed93b6048f04d58637 : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/857dd0b55bb6460ca803bf437ab05c79 : SUCCESS in 1h 27m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d4231bc995c14e0b983f31263257e4aa : SUCCESS in 2h 06m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bb1aa4e1cb9d4cf999a250b165784fe5 : SUCCESS in 1h 25m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/815511eadabf4ca882a1a3dd8f354b9a : SUCCESS in 1h 45m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/649109f1ff9b4bb2bfbc3f2a35aed853 : SUCCESS in 1h 04m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4bd8c0a5c97f487894f1bb07e871113a : SUCCESS in 1h 50m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/26084183256e407fb319d008946a7ffc : SUCCESS in 1h 05m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6a20d39cf5749d1af63664363b3aaab : SUCCESS in 39m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/06c84b02cbdd42478761ba3ff385fbb9 : FAILURE in 36m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/681ad450274c4f09960370d919401eb4 : SUCCESS in 34m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fc0379046a5f44d28b8a1aa176f1c85d : SUCCESS in 2h 13m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c80f631646b246cc977434f69a94faaa : SUCCESS in 2h 45m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d758ff7d6e9341819db04639ae69d13c : SUCCESS in 2h 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e4f37ce2072d0b9f13ee2f4821f7c089b15aa47","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-26 13:04:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5c3fe238ef3c82eccc511667045a65153b9ed366","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 13:53:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2d65e137ce0da24dce6a2ad3892397662b18b864","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-01 00:10:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36cccf684eb6ec378fd0d229e7acaa2e0174167d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 03:21:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcca4641e7a3412da6c0a93d47a34a57\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8defee8b3804a19894d3ee2149acff0 : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a318bae4b24366bdcf05875321038f : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6649cd4ba3f47f58373a29e39472d8c : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2129783be516427c8f616ca63154bbf3 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15efab1f71234537b6c2ee5e43e2e3d0 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98fd05603344abbafb4a6c987bd0efb : SUCCESS in 13m 32s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3cf299aea4314695a7656e468f94afff : SUCCESS in 4m 52s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3c082e3371c24379b3ca58c1e36f916e : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/34cb24f7ed324ecc91393185eede2d05 : SUCCESS in 2m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/683ba92e137b4363ab3879c8ddedcf11 : SUCCESS in 10m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f33d2e478f544ed5ab20fa75e4bcf1fc : SUCCESS in 10m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8722d10e486e405d99414029637ca8d9 : SUCCESS in 1h 29m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9f8eb2451d454b01a9f9c0f8798989db : SUCCESS in 1h 53m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6de80d7455e845699c0688f2ff34f048 : SUCCESS in 46m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b133d1bd3f374d918b35481b38bd3d36 : SUCCESS in 1h 31m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9276dc8475aa4de38f28b173066fc1d9 : SUCCESS in 1h 16m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8cd84850c334465a96f93efd1f2ca96a : SUCCESS in 36m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4376d276bcd2439f89ea7a06777e7360 : SUCCESS in 40m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a9f0b6c7bf248ccbefe6094688c89d1 : SUCCESS in 32m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/69b56399c77f4c60a6738c35529e5dbb : TIMED_OUT in 2h 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc2e862c32084ff9a52343dd18d87886 : SUCCESS in 3h 01m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0c06242c165f4a9fb655caa1f6171ddf : SUCCESS in 1h 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eaed5a8d6464ce3209a6d2a26006cb391e7710c8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-24 13:49:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"76e994bdb79f5012030c4b2061f92ebaa251cb73","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-11-25 02:37:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"412c7fa8f9d9b0145291a726988412ff86aec093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 02:37:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fdb33e4216736e2f2a4792334c93377809c4c056","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 04:19:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16a91bacbfe468089ccd7c1e9396e6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f48951289164b65a2ff9c96880915f6 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6e34519a8cf48d783bf323af35dbdd9 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cd6dbc08c4242cba3a77b30ce02ece0 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f8a37095a14977a786bea0db9cf9f7 : SUCCESS in 13m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f2679b9eb4b148da8ce5a64b67bab04c : SUCCESS in 3m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ce19819db2e84ab086f2409703ea267c : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8928ea5337fc41ae94dfcae03d29d63c : SUCCESS in 1h 21m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/60f179aa81da4767bce906f915492b4e : SUCCESS in 1h 38m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f7d1ee98723b4b4aac75e38ff9ed53b7 : SUCCESS in 1h 02m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ff516d078cd448eaad9b66a6c620985 : SUCCESS in 39m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/07cb66cbb2a1410c80c84464c951f890 : SUCCESS in 46m 50s","accounts_in_message":[],"_revision_number":3},{"id":"2a1c1e01c8824994987033e112891dea55470590","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 04:19:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b87292ba253d2817d6d4de3f5c456412505fa340","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 04:20:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2247170f6e04b7fa1ac479bf3cdbfc1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897a925299e649bea733f6629cc6905f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"21eba86a3ac60f5c0846715e9b4d145eadd9e925","revisions":{"cb03cfe18024addc4bf917a0b110861103ad8481":{"kind":"REWORK","_number":1,"created":"2025-06-04 02:17:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/951732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/951732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/951732/1"}}},"commit":{"parents":[{"commit":"a5fc4f3892588be41e11366c9fb442873d5ad091","subject":"devstack: Remove OCTAVIA_USE_MOD_WSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5fc4f3892588be41e11366c9fb442873d5ad091"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-04 02:15:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-04 02:15:08.000000000","tz":540},"subject":"Refactor uwsgi config management","message":"Refactor uwsgi config management\n\napache service is restarted by write/remove_uwsgi_config when needed.\n\nAlso get rid of the _apache_(u)wsgi functions because now apache is\nnot used (unless tls-proxy is used) so the current naming is confusing.\n\nChange-Id: I9b4860de49c9b54ec0403ac1e682f269846e1711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb03cfe18024addc4bf917a0b110861103ad8481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb03cfe18024addc4bf917a0b110861103ad8481"}]},"branch":"refs/heads/master"},"1018c8a8226ac454b168314e95a06317abaf9f79":{"kind":"REWORK","_number":2,"created":"2025-06-04 03:03:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/951732/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/951732/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/951732/2"}}},"commit":{"parents":[{"commit":"a5fc4f3892588be41e11366c9fb442873d5ad091","subject":"devstack: Remove OCTAVIA_USE_MOD_WSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5fc4f3892588be41e11366c9fb442873d5ad091"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-04 02:15:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-04 03:02:53.000000000","tz":540},"subject":"Refactor uwsgi config management","message":"Refactor uwsgi config management\n\napache service is restarted by write/remove_uwsgi_config when needed.\n\nAlso get rid of the _apache_(u)wsgi functions because now these are\nvery thin and maintains only single function calls.\n\nThe remaining template file for mod_wsgi is also removed.\n\nChange-Id: I9b4860de49c9b54ec0403ac1e682f269846e1711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1018c8a8226ac454b168314e95a06317abaf9f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1018c8a8226ac454b168314e95a06317abaf9f79"}]},"branch":"refs/heads/master"},"21eba86a3ac60f5c0846715e9b4d145eadd9e925":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-06-04 03:05:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/951732/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/951732/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/951732/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/951732/3"}}},"commit":{"parents":[{"commit":"a5fc4f3892588be41e11366c9fb442873d5ad091","subject":"devstack: Remove OCTAVIA_USE_MOD_WSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a5fc4f3892588be41e11366c9fb442873d5ad091"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-04 02:15:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-04 03:04:49.000000000","tz":540},"subject":"Refactor uwsgi config management","message":"Refactor uwsgi config management\n\napache service is restarted by write/remove_uwsgi_config when needed.\nThe _apache_(u)wsgi functions are removed because now these are very\nthin and maintain only single function calls.\n\nAlso remove the remaining template file for mod_wsgi .\n\nChange-Id: I9b4860de49c9b54ec0403ac1e682f269846e1711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21eba86a3ac60f5c0846715e9b4d145eadd9e925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/21eba86a3ac60f5c0846715e9b4d145eadd9e925"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
