)]}'
{"id":"openstack%2Foctavia~952287","triplet_id":"openstack%2Foctavia~master~Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2025-06-18 05:43:39.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-06-18 00:08:33.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6","subject":"Update diskimage-create.sh to use noble as default","status":"MERGED","created":"2025-06-10 16:58:56.000000000","updated":"2025-06-18 05:44:56.000000000","submitted":"2025-06-18 05:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952287","meta_rev_id":"d5a8bdb59f5b6f91d14e7687dca2962f98758348","_number":952287,"virtual_id_number":952287,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-18 05:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2025-06-17 03:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-17 12:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-17 12:05:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 00:05:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-11 07:49:27.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"CC"},{"updated":"2025-06-12 12:50:02.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-06-12 12:58:20.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-06-17 12:05:40.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"41c128c8e3500d3b73495f3d2ffc7a54f0da92be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-10 16:58:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"284f68c84902c3f28081ce3d760a6b7410806470","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-10 17:00:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c94c4478b836c9f92e5d6825788586df43bf71d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-10 19:18:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cbc565a08c4f88683d36a3adfc47a85ebf4ccc62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-10 20:31:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"22e7417d190403ab816a3f3078ff1c4ae57e984f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-10 20:32:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"950bb867dc7a3609ecc96d139f4724a54460cb96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 00:05:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d445537896147e189958afb28f0f58f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91e937fcbe4347c694cde480b41a9b03 : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1f893279ca4697bf3afe33e9462ed8 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fd638296aa1468cab928ea4fb718aac : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5500a17c68704858b64fc0c535964fea : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62be33ac71964fec886a0f1c362574f4 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f04d7b84ca41c9b297f0d354f00059 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e6fc60db05b4747937477c84bdc95ec : SUCCESS in 4m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4e93ddc90e4d4716bbd042da6877b318 : SUCCESS in 4m 53s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/796b2694613c4ff0a87de824b9fdd150 : SUCCESS in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce2da5d96f7148b3911b4f67d67474d3 : SUCCESS in 3m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa0212625fc9463eb2cdd40f7eb9977e : SUCCESS in 5m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a8c2a5d672ca45cf8227ed4f3065ef90 : SUCCESS in 5m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8f9bf6fae0148e892bf35f59b47c1ca : SUCCESS in 1h 20m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/38537826bf8345f7bcda6fec591fccf9 : FAILURE in 2h 23m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/68d0163df4c84d0da956e54519cb3e0c : SUCCESS in 49m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6efb9767d33d4dfaa1ab32d761adddf0 : FAILURE in 50m 09s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a2316427c3c4229b387503166193a57 : FAILURE in 1h 34m 34s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ddb49ff01611453c8b2d308c93e3b41e : FAILURE in 2h 34m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d02c4a76fbcc4f1bb72cf6e9150a2910 : SUCCESS in 1h 21m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56b006a0534a464fa40d79effea4b8df : SUCCESS in 55m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/563f024f308743f993fa910b0f9b5467 : TIMED_OUT in 3h 03m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/823b67d7027b4f9d8283157474e4a75a : SUCCESS in 42m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/353eb9d01247453c8610e3f671a75911 : FAILURE in 2h 11m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87b3c9a7dfdb4440b44cd5afe2a003a9 : TIMED_OUT in 3h 02m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf4082693dd04c8da7b72cf8c045bd41 : TIMED_OUT in 2h 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a4b209cc669c6c856328e1626987263fa0ec525","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2025-06-11 07:49:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a4aa76e820789b506f779115dedcd1c2e333998b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 10:56:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dee734a65dc480095e77cf227ac5463\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fabc7cd1a7e459cb5892e235d7d5385 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfb528080fd4fefa2aa52ee4104863c : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a11256a5397429f996ae249c1a5bb9d : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56d41c4bfa1f43349656435c23400e6b : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66041774d74e4048af4a8cda37955a28 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70b8d6dfe804086945a62bf244986d2 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/401c65710d694705aec38778a1e33b52 : SUCCESS in 4m 00s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ced7b21a4f294075982261055a696f14 : SUCCESS in 2m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/11b06aac61c6478ca7fd9accfd648568 : SUCCESS in 6m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f574a9b0cef49868dc816bd24e44aff : SUCCESS in 1m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6dd1cd7d403b4b6289218d8b4697cc2e : SUCCESS in 8m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ba5827b9a3f14ed7967ec8dfc4a23452 : SUCCESS in 3m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63c0e895319c4c5099a8dd62b665e244 : SUCCESS in 46m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/621a2df35e244f4b99ec6c9e048ff982 : FAILURE in 2h 08m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a8bca95b29464ac789744ce1f8808d9e : SUCCESS in 1h 13m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04e634efdbc546e383d28ffaf3987d01 : FAILURE in 41m 22s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5e67703e0af4666aaa86b53441efeba : FAILURE in 1h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6e84872ef5ba4583bc48ce8efb4e7189 : FAILURE in 1h 41m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9a36f4d9ebd34ff08a94810dfc3a2eab : SUCCESS in 41m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/839d1e8a01f44b97a0c606cfe7d86f5a : SUCCESS in 45m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1efc1ba728474ddfbb8c49f24e5ce6fc : TIMED_OUT in 3h 02m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cbbe6dda281490d9988761c923dd49c : SUCCESS in 43m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/40b91a251616459692f2afe03fdb45d3 : FAILURE in 2h 26m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5858c1883af4ca0a5e0954425ce6de8 : TIMED_OUT in 3h 05m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/43eca29d15c04f7ba7a0c24049e81ac0 : SUCCESS in 2h 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"afa1721cf7ea493067b4ff4187ae90560421054d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-11 22:28:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3cccf7a918de4de73068cac752fd6c97ca252eb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 01:32:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4400964027fc45279df56b84d63a8a56\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a82b24097a7347d1b3717ffd36152350 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b775778514a49ce95338e97e01ae794 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e6c33fa87b446c98998e2e03c5c727d : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ad30ead120a4207970a9194a91fde41 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72cae5125335449a8ef33483bc10d7fa : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19dc17fb274b40be8d4d51833881a57e : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2af94fc60d8f4de18834ed818bf91e45 : SUCCESS in 4m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/dc6f46612d7c478f9169a1135695d958 : SUCCESS in 4m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/239d386d30c7471fb54e235fa3d6acea : SUCCESS in 6m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/18e57bdb627b435c8f13170b26bb2a17 : SUCCESS in 3m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be4ae6f47738489186f37ab473f26bf2 : SUCCESS in 5m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d0c239165deb4139b269af7fed6e9571 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3a2e790962d4a3c9b3dad8450ac8a52 : SUCCESS in 1h 11m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d006ac74fa7f469bba68e0fdada10e47 : FAILURE in 1h 36m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/18aa3c5b419f490d9795c4fe03ad4481 : SUCCESS in 44m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c55391d60f404277b616fe91afc9d2d6 : FAILURE in 47m 17s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1c43853b9ef4d82ba043f26756d4194 : FAILURE in 1h 34m 08s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cac27c65abcc4a5f95cbf76899271bea : FAILURE in 1h 39m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9ed66e63c6764221957f05e49a96c14b : SUCCESS in 1h 07m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60df586774a1440798bf6dd0553591aa : SUCCESS in 36m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0efe79046da4487293321ee6d738d29c : TIMED_OUT in 3h 02m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d117c038deb74a7bba6a4d52b3cddb84 : SUCCESS in 35m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b93f758d22214e36bfeff17acb77362d : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c192719e14c7473fbe384ea131243d6f : SUCCESS in 2h 21m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/72b08324b98b4a72a649a9ed17c312e2 : SUCCESS in 2h 28m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"04af1ad4a6d8500a915b69ffc33703a12a20d392","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-06-12 12:50:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee0fea7fbb8530b7af2c2c4fc7f5027562fea3aa","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-06-12 12:58:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3a8dfd889444e7cc3fab53d015f624bb9067f6d5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-06-12 14:34:32.000000000","message":"Patch Set 4: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"63539be2c1f56d8c23347d710491ab54a00efc8c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-12 14:46:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f0895f7740437ae6cff036dcf789ebb630314122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:54:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af627ac2488d4b65b9c3535480a52150\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2409632ee8d4ab5ab8c758cbd17ebb2 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb712efd52d54ad89df30eb5595997e4 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af4f38514036419ba9c920f9d639cc64 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04383516326b40fd80333e5a0008eadb : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8120c7d477e142e8bf5b74b033cde661 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/222e9db0666448378c0455a8fddb4bc8 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe5d1f2d27944d1595d33f2e157d5d90 : SUCCESS in 4m 33s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d7cffba1d7af40e0b303a1e1ba92a773 : SUCCESS in 2m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1930dcc0c3f04c3886ea2305ac1ab962 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/df214852dff54176ba729b7e8498d0ef : SUCCESS in 3m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae2d170cbed04e409c21cff3cffe0f20 : SUCCESS in 7m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7b2fe8730466431ca872da4a1f9bc830 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5b556ebf6c541c18b449d332df1efb7 : SUCCESS in 1h 16m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b54ccffecc4141cc95445205d06434f3 : FAILURE in 1h 23m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b65d39bf41954fda8894a6fa451640d1 : SUCCESS in 1h 25m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec4b69f71e304e26a456c10d667ded2f : SUCCESS in 1h 23m 28s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ce6507c1cd9412eb360fd28df17989b : SUCCESS in 1h 06m 39s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/415bb1c6afd54268a294836ca7f7b016 : FAILURE in 1h 55m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6aec070e1df24ca49207841067c38d81 : SUCCESS in 1h 12m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/549ecf1ed54f46338b35789509e6205c : SUCCESS in 35m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0e56b4e49034069b3f08b9a65b964c4 : FAILURE in 47m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cf79925e6964b6e927d40fb1f538595 : SUCCESS in 54m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a5f71e9deed44e0fa853f1e76b04aa62 : TIMED_OUT in 2h 33m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/672be6a1c64a4b868ebb25c86a66cadb : TIMED_OUT in 3h 02m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0d43d4f38f04b448dcc1f53a01bd67d : FAILURE in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af5fdaec40205349bfb82bd4f58723e9704ac9ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-16 20:23:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e0da1e17b8b8a7b75e46c7832463ec115757654a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 21:34:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dc3d7a8cf9f432c9006bbe59a1d6ce8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a1b3746a37b4ba28cc34cc2943a04d4 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/099a90a8c4694263b543d783b384681f : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3dadececa12b4111946c95dd23d157e8 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b86a39235e1e42e0bb90745ccb639969 : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97ba29d1504b40a0bd0303b11e3682d4 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614e02c06bcc4bf588dc52a451d2b51a : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fa842f882944e80aaba5bf665af48ae : SUCCESS in 4m 04s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/46f23f850f34423cb67f39ac074b6739 : SUCCESS in 4m 11s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ea6e6e102b584769ae30235cd162c147 : SUCCESS in 6m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1a1d74bfbe59483db7e55c86e111f682 : SUCCESS in 3m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4aa3df7501c04562a8516ef90b35dd2c : SUCCESS in 6m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f6820fb1b9f84b3ca8f7b070a285b3d6 : SUCCESS in 5m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ce88071ae404cc4b9fc8a2bf2f92a52 : SUCCESS in 1h 09m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f6615ec1dae44fd787d3da0e0dab5a07 : FAILURE in 25m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cf88923e6ac44d86bec164b74bfda165 : FAILURE in 18m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a5a242068ca4830a51b2531acde47b5 : FAILURE in 24m 51s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1988062670941fcb797ad74ab13e6b1 : FAILURE in 16m 15s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d13e3e98df114bada50e229a86869458 : FAILURE in 19m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/05f94536d33b4f1d8eb47e8a551099f1 : FAILURE in 18m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf851f030ae3485aa32f26a523c751bd : FAILURE in 26m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e7516e09a2b46f1bbb4461f0b28f77a : FAILURE in 49m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/326ba761d2b8446cbef22e8469f6341a : FAILURE in 16m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/27d0feda1f594263a2d6d31e5405501d : FAILURE in 16m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b11fe7e8c752472f8b6547ef5bc57842 : FAILURE in 20m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d9953b83576b4e3497ad2d5e49765033 : FAILURE in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"91f8331fd451dce797c75a260d8971e73c322167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-16 21:39:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d0a478d0f4c3ec16bb5a09c7c6af10566f4687d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 00:43:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/555fc2aab22f46bc8ef68b06be70f233\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d500024d12b40edb91ca094c259d252 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f98505b71b14099b2cd6b8d50550b5a : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0968c261302e42ae9b2fad9462ffeea3 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/29a470c3b3db47d194252f455bf5adf6 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97d9734a466d422683c1006a1480b535 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0695584698f843df845504f3a112be05 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/993e45849ebc48aa9e4962980177121b : SUCCESS in 4m 20s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ebc680dacd444a38af0ddeccd239af02 : SUCCESS in 5m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f8d2a1439ce44a359071ec0f209dfbf7 : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0dee0d5c17cb4be08e52fef3dc668864 : SUCCESS in 2m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78b039f28eb74967a9a99ea4bffd1aac : SUCCESS in 6m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5d28e3aaef764f9eb37f9e5a8afd6462 : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d92e56cffea8452a83c57eb51ea91f9a : SUCCESS in 1h 18m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/980af466d9574986ac24b7426d7f5dce : SUCCESS in 2h 25m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/12ed873610c648828164208190d2bd78 : SUCCESS in 1h 07m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fba3186c1a9e44d6b4b2b2388a895e1d : SUCCESS in 2h 04m 20s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d1bab4ce7884ed49136976b92192340 : SUCCESS in 1h 16m 37s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aaf1d620705d42c9beb7f32195a7eb5f : SUCCESS in 2h 32m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ea3b07dfc5174cd297acc6f0218d3a3c : SUCCESS in 1h 14m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2deb1a210c1c4d86b96fbe5e134c5125 : SUCCESS in 48m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce57d221ecb14918b2f7ea7ce54fe2dd : FAILURE in 22m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e068e6954903486aabbcbab31a58fa38 : SUCCESS in 42m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/85ce8b5f26ae4fd8bf7acf15968da94f : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2f625a2c78740a98372b778593e00ef : SUCCESS in 2h 56m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/77c4dba9b92b457aa6e1812dee43872e : TIMED_OUT in 2h 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"28902241fe0892ed5942feeb8db3e60a1ce7e6cb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-17 00:49:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dbd4dfcaf0a8a847aa662ccf457b381bb218272a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-17 03:38:10.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cd9003ff47ea035909508f23824a4e6cf155b088","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-17 12:05:40.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0490623bce389d634e512c4b220688404fbeaf3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 12:06:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"990b63d69fab74c4eea81e9c3402bc3a4cf6bbb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 13:47:35.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/723758ed301845ccacf1897fcebfdc1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa6ff7982014d5c8f787d33b5478ad8 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b4795ebf47040d7a4f1a6b4f3cb6531 : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50f743f9c8c44fe5b7c4942614cfbab3 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81c4a23ee25345fc877dc7369a0132c5 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cab557b607bd4476a53ac742affc3b3c : SUCCESS in 3m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c04473f283544dbbaaa82b3fc35081c : SUCCESS in 3m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1a9d089a8c1c44a295bdfcc1a5f05538 : SUCCESS in 9m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4fe1a9ad29142cba488279f2f32778d : FAILURE in 1h 36m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/09b3b133c2704ce19365dd351e4d8c1d : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ca3b6842b33541f09b8c4b4a5cbc46dc : SUCCESS in 1h 15m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e3dd5e8f2524c3dba1c5dbd30df0c05 : SUCCESS in 39m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f01d2417b4d47fdb7ccbc8f5e3e6f0d : FAILURE in 37m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f5b40be22bfe0a0e7a717efe00a7c1122bd1b2d4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-17 15:25:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc19be233abd493d201e62fde94e94089d0ff88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 18:36:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da81d0429a84666be635cd8323f3de6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/222f5bdddd29415ba78dccf2517698e7 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffa294424ef49a39e6777026cc12aab : SUCCESS in 7m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75be3882dde546d3aea3e54f0a397afd : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aab17e42e783407fae5d6c9dab3ed2c0 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96cb46762e5b4fcca7a2ce3fa88a3e15 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f2f46a41784cd2b823fdb0a909d7ca : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5047cce16ab849f6bf20eb1b0554ebcb : SUCCESS in 5m 26s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/eea8d7a380494ceab514e68ea4646fa3 : SUCCESS in 4m 45s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/de7e53150ed249ac94986ba6da225b31 : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f72e32b1dee4b549c6efa1af9bb9195 : SUCCESS in 3m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c2b00e85d894514a995681e2e4b4370 : SUCCESS in 7m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/06567207cc904081ab80560fee7fac5a : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60dc8f70b29e40cbbb4480ae5c3a43b8 : SUCCESS in 1h 26m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cb118deba68f4e73a0ec857152242eba : SUCCESS in 1h 51m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dd1fd1aed0f64468a03730b7b62c66f0 : SUCCESS in 57m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8394828d5d174bb7a43c38283584b31d : SUCCESS in 1h 46m 22s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6afcf06f0784f1a99094482df29af02 : SUCCESS in 57m 13s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d99f6fdb64da41288cc790e7d068603c : SUCCESS in 2h 09m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3744430053494d6392fec8e35355949e : SUCCESS in 1h 43m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/643ad0244989491b9a0dfb3bcf3eb703 : SUCCESS in 44m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b07895c1c90454180d6d0bfd7bcd709 : FAILURE in 1h 00m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31943c217b4049f381031183a2400e11 : SUCCESS in 54m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/640fc0165df440de909d0b9016297785 : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d56215a56d94485aaf11ae8461cc77b9 : TIMED_OUT in 3h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d75947d7d1d847eb97e4bd4ef13443ee : TIMED_OUT in 2h 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"099e93959b4cff11b72a12b89a2e8ab0e7f94e17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 18:37:08.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c76d2688abce3a5a2b96e71f53a6a811c064ea49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 20:26:51.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6188132ed876410abe5d99a4ca93a9f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f15fb9559294d518c9e68918135ed99 : CANCELED\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a50d0f164486451c89bede74605a2e84 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fbb5d95e9ae4e0dac0393984828cf06 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41069823ebe43f7a9e9a4d00b6ac683 : POST_FAILURE in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01624deac57747369a0b76abccc40012 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c37bf5026ab4a6b968d6b97c6c8d875 : SUCCESS in 2m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6af5d85f4f7d421ea509b37aba3c4103 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5eda306814164ccfa86e7a79ab60b0e2 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4493b8b8824646e9b6e7429431cda372 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cb5050537c1845c7ad8ea072f8a98162 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1023976f313c47c7a64cda4e76618f4f : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/93e02416d0734480b00856941986f95e : CANCELED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1dca7bc334b72a5926fe834b457ca7e777e162a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-18 00:08:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e154b50e91b0288c052a62a0124296e8667346a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 03:16:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b15f64c0f530462583d6c11962490609\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84eab42a2d3142569eb380d181d565a9 : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b9cde94241743cab4e70e498acd01cb : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44d458668b5248fdad4e4d234f1292d0 : SUCCESS in 7m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07c5d709b9d14890a4dcb5ebe5b647d3 : SUCCESS in 6m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47bf5cd4095e495cbb94c1bd5a07e752 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/799895c5b04c401eb5f71e2b52888ed0 : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c660740236e7476c9948e30fc5aba757 : SUCCESS in 5m 12s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fc404dd4cef040dab62fca48ee581687 : SUCCESS in 7m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2b0ec3114118483eb99a732399a94bb4 : SUCCESS in 5m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3c343389f8d430c992fa4b0e0a9178f : SUCCESS in 4m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad613899676d4bf1add2bb7d962df03b : SUCCESS in 7m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4eeac60c952146628cb94432b8001fb4 : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fbab05326e94c5e86e5e1ad2bc6c64a : SUCCESS in 1h 05m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7ecc598821a849d7a7a260b110b7fe66 : SUCCESS in 1h 54m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b5dda393268e461ab702c0ac00b9368f : SUCCESS in 1h 03m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8884d676c6ac4713bb154dc6cdb0f9dd : SUCCESS in 1h 48m 47s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f991feb2c9924e2da4706246e49cecc5 : SUCCESS in 55m 42s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3f4daa58027041e9bcc9005c25e49869 : SUCCESS in 1h 56m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f55ea4e8d5f7465987b136ae51d0b43e : SUCCESS in 1h 08m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d290f09c9c624c02b1158661d9c6def7 : SUCCESS in 36m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/acd5dc593afd49aaa0abec1d268d1564 : FAILURE in 35m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b648da50358b4433859a5f65cc7e5e0b : SUCCESS in 36m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4794bcd381834b509ffbd733d2fcaf48 : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/79c65f21d1be407bafc5b152128bdee4 : TIMED_OUT in 3h 02m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c3e98d256874254becee0873b8cb653 : TIMED_OUT in 2h 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c8c97d3d0c27262a96b976fd16a71632731133d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 03:17:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c70fd0fad13f0372b1df9aa21f8ad1eac538e5c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 05:43:39.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26cdb3de884c4ffbb24ad3f2fe06f0f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee3afc16598465694964fa39d7b1201 : SUCCESS in 8m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b89dbefbd004c19a3cc3020ad912aef : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10fcf2c353cc4fdfa622b346350a6bf3 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/731164e26d6147caa542134fd5a1741f : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c021def9300f4b8eadc0bf0f3a5abca2 : SUCCESS in 5m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a35ea042f331414e8ff5976f9273565e : SUCCESS in 5m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/473928fd0aa2421288f0ea55c8d59b6c : SUCCESS in 9m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e37bd5d6e954e149fe1a20c3faa793c : SUCCESS in 1h 18m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7dc8212d43b5441590ec78d6e799ddae : SUCCESS in 1h 53m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/06dc77821a72405c919153b9f4be848f : SUCCESS in 1h 12m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da45d102f9d143e6a3b1c54d79cbd671 : SUCCESS in 51m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9baf125eeed84df2bb3435f44d44ae5d : FAILURE in 43m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"88b3bbb64f8de330325969b8c383326e0f117622","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 05:43:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d5a8bdb59f5b6f91d14e7687dca2962f98758348","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 05:44:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eeb94372cf64844b1522f7f9e558ead\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b67c4bf2a6644afbcd0d66b9ac10228 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3860c1cc33e74a48b004a9e05c143018 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"57eb9b49dbc6b87f46e615f07220f048be66e31b","revisions":{"2f1c4cfc26973262f58153e712367a4eec74feef":{"kind":"REWORK","_number":1,"created":"2025-06-10 16:58:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/952287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/952287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/952287/1"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 16:57:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 16:58:55.000000000","tz":0},"subject":"Update diskimage-create.sh to use noble as default","message":"Update diskimage-create.sh to use noble as default\n\nChange-Id: Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f1c4cfc26973262f58153e712367a4eec74feef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2f1c4cfc26973262f58153e712367a4eec74feef"}]},"branch":"refs/heads/master"},"171a025098c75b893774701105a236473b47b79c":{"kind":"REWORK","_number":2,"created":"2025-06-10 19:18:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/952287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/952287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/952287/2"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 16:57:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 19:16:57.000000000","tz":0},"subject":"Update diskimage-create.sh to use noble as default","message":"Update diskimage-create.sh to use noble as default\n\nThis patch updates the image build process to use Ubuntu noble as the\ndefault distribution.\nIt also removes the http-use-htx option that is now on by default and no\nlonger a valid configuration option.\n\nChange-Id: Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/171a025098c75b893774701105a236473b47b79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/171a025098c75b893774701105a236473b47b79c"}]},"branch":"refs/heads/master"},"9ccdc14bb0088c7c07e85eb795bfa5660380110a":{"kind":"REWORK","_number":3,"created":"2025-06-10 20:31:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/952287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/952287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/952287/3"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 16:57:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 20:31:46.000000000","tz":0},"subject":"Update diskimage-create.sh to use noble as default","message":"Update diskimage-create.sh to use noble as default\n\nThis patch updates the image build process to use Ubuntu noble as the\ndefault distribution.\nIt also removes the http-use-htx option that is now on by default and no\nlonger a valid configuration option.\n\nChange-Id: Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ccdc14bb0088c7c07e85eb795bfa5660380110a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9ccdc14bb0088c7c07e85eb795bfa5660380110a"}]},"branch":"refs/heads/master"},"1b0e8e67f0ce33f829c6e9434e0637acafb89720":{"kind":"REWORK","_number":4,"created":"2025-06-11 22:28:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/952287/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/952287/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/952287/4"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 16:57:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-11 22:28:34.000000000","tz":0},"subject":"Update diskimage-create.sh to use noble as default","message":"Update diskimage-create.sh to use noble as default\n\nThis patch updates the image build process to use Ubuntu noble as the\ndefault distribution.\nIt also removes the http-use-htx option that is now on by default and no\nlonger a valid configuration option.\n\nChange-Id: Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b0e8e67f0ce33f829c6e9434e0637acafb89720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b0e8e67f0ce33f829c6e9434e0637acafb89720"}]},"branch":"refs/heads/master"},"9bc173cfc1fdace61ff408cd9c2b4333906a8def":{"kind":"REWORK","_number":5,"created":"2025-06-16 20:23:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/952287/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/952287/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/952287/5"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 16:57:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-16 20:23:09.000000000","tz":0},"subject":"Update diskimage-create.sh to use noble as default","message":"Update diskimage-create.sh to use noble as default\n\nThis patch updates the image build process to use Ubuntu noble as the\ndefault distribution.\nIt also removes the http-use-htx option that is now on by default and no\nlonger a valid configuration option.\n\nChange-Id: Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bc173cfc1fdace61ff408cd9c2b4333906a8def"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bc173cfc1fdace61ff408cd9c2b4333906a8def"}]},"branch":"refs/heads/master"},"57eb9b49dbc6b87f46e615f07220f048be66e31b":{"kind":"REWORK","_number":6,"created":"2025-06-16 21:39:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/87/952287/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/952287/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/952287/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/952287/6"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-10 16:57:31.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-16 21:39:09.000000000","tz":0},"subject":"Update diskimage-create.sh to use noble as default","message":"Update diskimage-create.sh to use noble as default\n\nThis patch updates the image build process to use Ubuntu noble as the\ndefault distribution.\nIt also removes the http-use-htx option that is now on by default and no\nlonger a valid configuration option.\n\nChange-Id: Ic494a426a998a7db91e61001acd9c0ba8c1fd9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57eb9b49dbc6b87f46e615f07220f048be66e31b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57eb9b49dbc6b87f46e615f07220f048be66e31b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
