)]}'
{"id":"openstack%2Foctavia~952397","triplet_id":"openstack%2Foctavia~master~I69e9492937c2b8c2850fed5e89ea7ff9100cd450","project":"openstack/octavia","branch":"master","attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-08-13 15:27:12.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-08-13 15:27:12.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I69e9492937c2b8c2850fed5e89ea7ff9100cd450","subject":"Fix UEFI/Labeling in Amphora images","status":"NEW","created":"2025-06-11 19:22:04.000000000","updated":"2025-08-13 15:27:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"42dab1b303984b848efd63c198f7a7b0c742f3ca","_number":952397,"virtual_id_number":952397,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-26 06:25:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2025-07-08 21:32:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 22:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 05:03:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-07-08 21:32:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-07-10 15:18:47.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"9b5dcaad1595381c2bb9b71037fe6cfec67b863f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-06-11 19:22:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"543f59114de5bdb76bbf3a01bfe676f91b9cef7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 22:25:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8bd9086584e44005ba80143e6b6f31c9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b8d47024742424882428f23c24daa95 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0987803e41542f9bcde9e52f2843517 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4ba18c3853b4ec9b845582ef7529f8a : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/792babb20c9d43b19a4523779e13cba2 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f755c09a84bf40aea7260082b835a5bf : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e74a65d3a1b45fdb203d40f9b0dcf7d : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e70a9ba5b7b40e897d9aca776efd192 : SUCCESS in 3m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9cd7b0394f8344c19f8c23d5df8516f4 : SUCCESS in 3m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/d756ccbc46894e54b9e87cf4447adce4 : SUCCESS in 6m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29cc0149a657428683fa3a84da954f57 : SUCCESS in 3m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83534b88b6064fed97cf6d35e7aff756 : SUCCESS in 3m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d044d0745dd043e7bdd41fb814dd6cf4 : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a11e25a00724aa48608f48114102b54 : SUCCESS in 43m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4e9ad21649bb4cfbb8978afe11b517ae : FAILURE in 29m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1f14db6aa83349c58c17f3d000a5cfdc : FAILURE in 55m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bcab8f76abb403fa5dfc8f4b59db27a : FAILURE in 15m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0139059cac254d38b7755cd84ee88603 : FAILURE in 55m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d2fa3647cf084382b849e357cf8181fa : FAILURE in 15m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1af1cd3c610f450c8a69ac9d96621970 : FAILURE in 1h 20m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d6f7b9fb5264c8181a80962876c74b9 : FAILURE in 16m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/895f28e0824f40c6906e42f8c024cb1e : TIMED_OUT in 3h 02m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73d6e6365ce54089bf05f4cbc043926b : FAILURE in 28m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6a3077a64fd74c5991497dbdf2006e62 : FAILURE in 34m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a50a0f69ed44f57a9319c8ba50e9fe7 : FAILURE in 1h 20m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9b95148bad0c4ba58891ead4d6cc67da : FAILURE in 1h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0b3a234aeca6450daf6a164167ee53ba790052a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-20 18:02:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_11655\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}],"_revision_number":2},{"id":"0314deb69f6a4e6c0b1e54f70a523cdb68b48df1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-20 19:04:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28d5da97c030498ca59357b353e4d702\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61cb64d436c644eaad78659f1cb53b2b : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d165b9bbe66427da3b154d6cc371a0c : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/76f0b92b78464296aac41a7c80c9a0c1 : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1f81f6882424ef5a21f7751453b70d0 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f684242717794da8bf87d258a01b4a33 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9fb3530df784203b56026051cc805ed : SUCCESS in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a6dfe28e59c4e12a38373865a649de5 : SUCCESS in 4m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bb05cec032574464986eae9c1ed7615a : SUCCESS in 4m 08s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2403d455fc3d427d94beb99d67a6555a : SUCCESS in 6m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c1944bfa53fe484187884095274a9e8c : SUCCESS in 3m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df9bf8872b15418b9fe99d89d7269f60 : SUCCESS in 7m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4a1b9f8c66474ccbaf5310fe4d7d4e4f : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fece34816c274258a635df89edb750e6 : SUCCESS in 1h 01m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/125a3b2b2c1541978b6967500ba89aa7 : FAILURE in 34m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8b2f57caeeb64959a8b3a8b394a9c457 : FAILURE in 39m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb1242f7917e4322aabe55a7d4073b91 : FAILURE in 29m 32s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74ef426daf7d45caa742f875267900f5 : FAILURE in 36m 45s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7e8bba159ee4477ea3a73add6c2f40dd : FAILURE in 34m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f755d92a746f472285e4d047aa34af72 : FAILURE in 30m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b5f7f130c7c4e3694231fecab59141a : FAILURE in 35m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b69fcdf171444c5bbc983afb7ed24770 : SUCCESS in 44m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e72447bd776446ee934642d70109f172 : FAILURE in 28m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/16918feb25e1465680b7677aebce227a : FAILURE in 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bbce6d509b974f58a6a55dff11da23c0 : FAILURE in 45m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b9b66bb2a3f8456db4362fbf938b07f2 : FAILURE in 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"affc17ef41b9e9ecc8163523ff71b51f347fc1de","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-26 05:03:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"410a1aa0077e85210186b402bd243bc2772f4427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-26 05:03:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_11655\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}],"_revision_number":3},{"id":"daca94b40d55e7f1bab916a0361c721462b01a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 06:25:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95210af1985043dea95ba57d413d64f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/585dd52a7f75428db102ef3a3fb80c49 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1190e7e8954bb58d27c79298ce838d : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fccd955edf6142d3bfda80f11ff92214 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e4a3cefbe534450b085bce0e10f9b72 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4608bf0484294a5ba9fef5e1beb01027 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d56a2f70cfb499b9ca6d8afb367a9c7 : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c280f1eb46bc4637a067732e498c6ece : SUCCESS in 6m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c22b5808ae484a9a9562942ee8488018 : SUCCESS in 7m 00s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fea9af24fff84d84af903026243fb26b : SUCCESS in 9m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/61a381ce12034162b045a0d83188b533 : SUCCESS in 4m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/38aac8034a7842f49a1fb4648df1e2d0 : SUCCESS in 7m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/acb1d90cd8bc46f1bade1b26de2dd830 : SUCCESS in 9m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4256475b838a48fc9b5683f4530e70ae : SUCCESS in 1h 21m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f6c8314912cb4d9794493c80f4f095bb : FAILURE in 28m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fee108f2bf0c4fe0ac7e7b5e57b31af8 : FAILURE in 35m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1eeb06f89664f82a5264272ae43f6f2 : FAILURE in 25m 43s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4df1fafe9e7d4fba8471030bd0c3a237 : FAILURE in 36m 45s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3eae0686b38f4e2f853ebad481655133 : FAILURE in 27m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d5e0c43792ed49ce9fc9b69d5fd59197 : FAILURE in 37m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d8d53a40cae94a649fed95d4cefb2b79 : FAILURE in 29m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cfd7754540344144866753accc472337 : SUCCESS in 53m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c6e313840794ab7b3ac504c5bf20fa8 : FAILURE in 26m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/06d57b59f07c466b8715acc12ed249a1 : FAILURE in 30m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ede82d1a3afd4d3883256e235de5aa66 : FAILURE in 44m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac7902e71fe249f89cecb016f4e2bd4d : FAILURE in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa47b08c55e05d00a875fb201c107fd6ac5180bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-08 21:32:08.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"577081dd09a03c1834891f2e364e2460f72a8629","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-08-04 14:10:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"42dab1b303984b848efd63c198f7a7b0c742f3ca","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-08-13 15:27:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"64c19455991c2d506aa4cb63d6e6ff8c75128e1d","revisions":{"a76ef247f3ac81e87cd15b5d32c5627f3ff06dc6":{"kind":"REWORK","_number":1,"created":"2025-06-11 19:22:04.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/97/952397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/952397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/952397/1"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-06-11 19:14:16.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-06-11 19:21:06.000000000","tz":-420},"subject":"Fix UEFI/Labeling in Amphora images","message":"Fix UEFI/Labeling in Amphora images\n\nWhen investigating an issue where a compute hypervisor was rebooted\nand an amphora instance booted instead, it was identified that the\nroot issue was found to be that the default disk filesystem label\nfrom diskimage-builder was used.\n\nWhile investigating this, it was observed that the Amphora image VMs\nlack UEFI boot support and are structurally future incompatible as\nthey rely upon the defaults of BIOS boot logic from the early days\nof diskimage-builder. This logic has been updated to invoke the\nblock-device-efi element, which generates a UEFI/BIOS hybrid disk\nimage allowing a UEFI loader to trigger the bootloader as opposed\nto the BIOS boot process.\n\nChange-Id: I69e9492937c2b8c2850fed5e89ea7ff9100cd450\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a76ef247f3ac81e87cd15b5d32c5627f3ff06dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a76ef247f3ac81e87cd15b5d32c5627f3ff06dc6"}]},"branch":"refs/heads/master"},"4ec18c731557ac63b2a289652c183f45b3d4f339":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-20 18:02:14.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/952397/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/952397/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/952397/2"}}},"commit":{"parents":[{"commit":"e981d3e172b8e4471f97251afc491141f06ec867","subject":"Merge \"Remove note about old pip\u0027s behavior\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e981d3e172b8e4471f97251afc491141f06ec867"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-06-11 19:14:16.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-06-20 18:02:14.000000000","tz":0},"subject":"Fix UEFI/Labeling in Amphora images","message":"Fix UEFI/Labeling in Amphora images\n\nWhen investigating an issue where a compute hypervisor was rebooted\nand an amphora instance booted instead, it was identified that the\nroot issue was found to be that the default disk filesystem label\nfrom diskimage-builder was used.\n\nWhile investigating this, it was observed that the Amphora image VMs\nlack UEFI boot support and are structurally future incompatible as\nthey rely upon the defaults of BIOS boot logic from the early days\nof diskimage-builder. This logic has been updated to invoke the\nblock-device-efi element, which generates a UEFI/BIOS hybrid disk\nimage allowing a UEFI loader to trigger the bootloader as opposed\nto the BIOS boot process.\n\nChange-Id: I69e9492937c2b8c2850fed5e89ea7ff9100cd450\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ec18c731557ac63b2a289652c183f45b3d4f339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ec18c731557ac63b2a289652c183f45b3d4f339"}]},"branch":"refs/heads/master","description":"Rebase"},"64c19455991c2d506aa4cb63d6e6ff8c75128e1d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-06-26 05:03:15.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/952397/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/952397/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/952397/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/952397/3"}}},"commit":{"parents":[{"commit":"3af83d8c12fd727410569ec01c9175b16fa818aa","subject":"Merge \"Remove zake from optional runtime dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3af83d8c12fd727410569ec01c9175b16fa818aa"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-06-11 19:14:16.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2025-06-26 05:03:15.000000000","tz":0},"subject":"Fix UEFI/Labeling in Amphora images","message":"Fix UEFI/Labeling in Amphora images\n\nWhen investigating an issue where a compute hypervisor was rebooted\nand an amphora instance booted instead, it was identified that the\nroot issue was found to be that the default disk filesystem label\nfrom diskimage-builder was used.\n\nWhile investigating this, it was observed that the Amphora image VMs\nlack UEFI boot support and are structurally future incompatible as\nthey rely upon the defaults of BIOS boot logic from the early days\nof diskimage-builder. This logic has been updated to invoke the\nblock-device-efi element, which generates a UEFI/BIOS hybrid disk\nimage allowing a UEFI loader to trigger the bootloader as opposed\nto the BIOS boot process.\n\nChange-Id: I69e9492937c2b8c2850fed5e89ea7ff9100cd450\nSigned-off-by: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64c19455991c2d506aa4cb63d6e6ff8c75128e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64c19455991c2d506aa4cb63d6e6ff8c75128e1d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
