)]}'
{"id":"openstack%2Foctavia~952843","triplet_id":"openstack%2Foctavia~master~I1e5d216cabcdf2db9913d90aca492dd4702be7e8","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-25 16:29:04.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-06-25 14:06:43.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I1e5d216cabcdf2db9913d90aca492dd4702be7e8","subject":"Fix missing stestr","status":"MERGED","created":"2025-06-18 15:30:15.000000000","updated":"2025-06-25 16:30:10.000000000","submitted":"2025-06-25 16:29:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952843","meta_rev_id":"c882f0f0fa21253d89258962a4cd079ab5e96269","_number":952843,"virtual_id_number":952843,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-25 16:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-20 16:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-06-25 14:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-06-20 19:55:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-06-25 14:06:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-18 18:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 16:58:45.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-20 19:55:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-20 19:55:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"fa5c4840bd7dd6d8c24d77d11fa3002941742a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 15:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a510f9e8c6e5ad39c46dc400c02a1ef61af9d35f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 18:36:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ab46b65142a48fea3198c35a4218a17\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/04ce93a609464978a22301bf3f6824b4 : SUCCESS in 5m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aacdb4035c22404387c79f0ade221297 : SUCCESS in 2m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65595069215f472f819991a2940e8fa9 : SUCCESS in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff73ce4a03674c57aaa50bc94918f1e7 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ced5410538a743179da3fd0c32514438 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3ffbca178ef4845945806cd5f92f94b : SUCCESS in 6m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/825c9ac354544afcbd67feef0a77dfe8 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb1d32926d441c7a4571480358be720 : SUCCESS in 14m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/489aedfe62bd4138b7cba686f422d0df : SUCCESS in 4m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/198c2b21b2ed4ac392bfe1448a5ec6de : SUCCESS in 4m 41s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6be7c7433746488b8900517078b8a58c : SUCCESS in 9m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/88af2866341d42b28e9d03736980b716 : SUCCESS in 3m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf682e59c82c4372b054c81fe4d92a5c : SUCCESS in 7m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ff7f9a26721942dba767c34a8ebb7de2 : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da99269ddbdb465fa221a7dbf98fa4bb : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a097ca87080e4df9a61b6117b4e3e2cc : SUCCESS in 1h 56m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/15a741aad4334a1c820e282777fb41c6 : SUCCESS in 1h 06m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b496ccefc2884933bdda445eab51cb49 : SUCCESS in 1h 44m 36s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e1a182e2bc5448295572d6b898a676f : SUCCESS in 1h 32m 09s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c105c05e640b467e9dab0a243bf06b22 : TIMED_OUT in 2h 36m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eb66ddd9686a40f49e8b335fb8b84086 : SUCCESS in 1h 25m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7649d397e08b4d43968ad5ead327fc5e : SUCCESS in 42m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a9863ce131644cf9bc880b6779d5c79 : FAILURE in 55m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aadf3310ea154890bf801b6c8beeef73 : SUCCESS in 41m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/62cb5d38b3b04dc9a0a3d70714229e1c : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cc5af346b8e41429f47a0b001b6b656 : TIMED_OUT in 3h 02m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c99264594a84079bff9500ea6576865 : TIMED_OUT in 2h 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3d68dc7953799067562aa99a61ab7ead0d78163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-19 01:24:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"30a1694bbf4524e56f27e4fa2b08b174ff01eae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-19 01:25:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"309e22494b7c0b56f494827032c24c0f99c9266c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-19 03:06:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db783a4b9a6a676d06ec3ff52ad30af9aec316f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-19 04:35:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7727de44b4704bf69f6eaf69ff44e76b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2f84f7b5901c4fa08e8c3e40a90f750b : SUCCESS in 5m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51f932d9dbbc41db921dfc9278f90035 : SUCCESS in 1m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b5fbf5c103340e2add1e18b6ede9469 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11aa5dc76d6f416cb328de507ab4189b : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c85b4907baf448ef97ef7fc873397f62 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f94f5be543254b17873c6ff4d6bc084e : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d15c85be387b41d3b3b3baf34153e931 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65299ffcc894876a12e7ef9da55219b : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e05a17fadba641b5a7823189ffe00639 : SUCCESS in 4m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/168af8c721ec42e4b92c77cb91b488a5 : SUCCESS in 7m 22s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/97929a8eae164664a699344f9bf7317e : SUCCESS in 6m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eab3f7d802d549e88f39b7f86ca583e1 : SUCCESS in 4m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fec07deb1ab546d18db1a0ddbeabc5f0 : SUCCESS in 5m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/38d39143f28147098ccc388a3eb5fe23 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2aab9022d9c4e31bba284e9f6c8e49a : SUCCESS in 1h 39m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/02f54cf6266e46ce9fe43bea446727a9 : SUCCESS in 1h 56m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/95575648bd86450f8ce17e9fb3b499a4 : SUCCESS in 1h 11m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dffd66941ad4449fab9733d84907f62a : SUCCESS in 2h 14m 12s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38900b585e92400485ea040a7a5ac1d2 : SUCCESS in 1h 15m 56s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b4801fa357214f928fe4703d7c88e359 : SUCCESS in 2h 20m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4fd29815b02e48b1a31968addea9dbfb : SUCCESS in 1h 18m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a211884381dc456eb574c9addd8e9655 : SUCCESS in 42m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/865be6ba7d3546e2ac1eca38d584927c : SUCCESS in 56m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5963418a01d4698bd40c3339ba96d3f : SUCCESS in 42m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d843e4d8406b4defb9584c98c0561ae4 : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d772b305470b4895b5f0a4d85463f24a : TIMED_OUT in 3h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/241501454f4d4cf5a05472f911370423 : TIMED_OUT in 2h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"42b7b952594a6073bcaf7537b81beeb1aabd76ef","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-20 16:58:45.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4f3691c184a50e050b4060a6a8968015a2668376","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-20 19:55:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2457b3002697b84bd16a9a9369ea2b4771e45553","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-25 14:06:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f0430b400896378f379019e5e8a01852c90bd27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:07:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4bc852c3c150e92a924037aa682e2c44102e5762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:29:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bee89703b1604745acbf3cd34d067b75\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/02012f3c012a4147bdea56c96c49fac9 : SUCCESS in 3m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b74bf202f35041fea5bc752c735989a4 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d72874237165418aa8c00e2036a26214 : SUCCESS in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f45c5b4914cc4891893be361c5a3393f : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d554ccb9888464982b42241bc8540ed : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c0323f4711403195526c32f39eb412 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61d6cc1161e94be480d27718096dd956 : SUCCESS in 4m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/defae4f86da14358b66b67112cbef6fd : SUCCESS in 4m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8fbfb752c68b41c997cf7757dde33374 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dad06ef7ff184cb49d3667c220264756 : SUCCESS in 1h 27m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/76b5a9aa4b7a4bab9698c7bb22eefa76 : SUCCESS in 2h 21m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/67b5b6e3a2924bf48126f64c603806a3 : SUCCESS in 1h 24m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/217a77ff258d41a7b465edc0bfc3a80f : SUCCESS in 48m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4cdf9e6ab16e4406b5f1e3ab3ecfc64b : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":3},{"id":"3dcbe1ccaadf28a0c6c54d984751da47c886acf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:29:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c882f0f0fa21253d89258962a4cd079ab5e96269","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:30:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c17c4e78bb274f259e8c2284c5117154\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/13c447e7f2ec4df1992fe39764145b55 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81eadd63839460e95d37739780fc0d9 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1779cf1ea5a741c999bb56d750928fa6 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d86418da372df81c3cf1e3272f9ae6333207540","revisions":{"aaac1a2b17787b3639e9306100d558ac1a297fa3":{"kind":"REWORK","_number":1,"created":"2025-06-18 15:30:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/952843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/952843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/952843/1"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:28:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:29:41.000000000","tz":540},"subject":"Fix missing stestr","message":"Fix missing stestr\n\nThe stestr library has been used to run unit tests since [1], so it\nshould be part of the test requirements.\n\nAlso python-subunit is not directly used but is installed as\na dependency of stestr, so can be removed from the list.\n\n[1] 7f3506b2956c4f6edc24703ea8099ce783715897\n\nChange-Id: I1e5d216cabcdf2db9913d90aca492dd4702be7e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aaac1a2b17787b3639e9306100d558ac1a297fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aaac1a2b17787b3639e9306100d558ac1a297fa3"}]},"branch":"refs/heads/master"},"34de113c1d8651e1aac62f4791ff31db782939ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-19 01:24:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/952843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/952843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/952843/2"}}},"commit":{"parents":[{"commit":"e981d3e172b8e4471f97251afc491141f06ec867","subject":"Merge \"Remove note about old pip\u0027s behavior\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e981d3e172b8e4471f97251afc491141f06ec867"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:28:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-19 01:24:02.000000000","tz":0},"subject":"Fix missing stestr","message":"Fix missing stestr\n\nThe stestr library has been used to run unit tests since [1], so it\nshould be part of the test requirements.\n\nAlso python-subunit is not directly used but is installed as\na dependency of stestr, so can be removed from the list.\n\n[1] 7f3506b2956c4f6edc24703ea8099ce783715897\n\nChange-Id: I1e5d216cabcdf2db9913d90aca492dd4702be7e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34de113c1d8651e1aac62f4791ff31db782939ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/34de113c1d8651e1aac62f4791ff31db782939ea"}]},"branch":"refs/heads/master","description":"Rebase"},"4d86418da372df81c3cf1e3272f9ae6333207540":{"kind":"REWORK","_number":3,"created":"2025-06-19 01:25:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/952843/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/952843/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/952843/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/952843/3"}}},"commit":{"parents":[{"commit":"e981d3e172b8e4471f97251afc491141f06ec867","subject":"Merge \"Remove note about old pip\u0027s behavior\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e981d3e172b8e4471f97251afc491141f06ec867"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:28:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-19 01:24:48.000000000","tz":540},"subject":"Fix missing stestr","message":"Fix missing stestr\n\nThe stestr library has been used to run unit tests since [1], so it\nshould be part of the test requirements.\n\nAlso remove a few unused test requirements.\n * python-subunit is not directly used but is installed as a dependency\n   of stestr, so can be removed from the list\n * No test implementation dependent on tempest exists in this repo\n\n[1] 7f3506b2956c4f6edc24703ea8099ce783715897\n\nChange-Id: I1e5d216cabcdf2db9913d90aca492dd4702be7e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d86418da372df81c3cf1e3272f9ae6333207540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4d86418da372df81c3cf1e3272f9ae6333207540"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
