)]}'
{"id":"openstack%2Foctavia~952844","triplet_id":"openstack%2Foctavia~master~I3929e856fb97dec25fd6a9ddd5d1988bd091b409","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-25 21:51:38.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-06-25 14:07:56.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I3929e856fb97dec25fd6a9ddd5d1988bd091b409","subject":"Remove zake from optional runtime dependency","status":"MERGED","created":"2025-06-18 15:38:09.000000000","updated":"2025-06-25 21:53:27.000000000","submitted":"2025-06-25 21:51:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952844","meta_rev_id":"1106b027d0f3e9fdba139b7e904eb2d2bc63dae3","_number":952844,"virtual_id_number":952844,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-25 21:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-06-20 18:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-06-25 14:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-20 19:56:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-06-25 14:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-18 18:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-20 18:08:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-20 19:56:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-20 19:56:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"201a2c755078850c0a8e163dcb31552da068fcaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 15:38:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f723d830738541e34b9a9ffd5e137f0bd153174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-18 18:45:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f8f9ce72bf34761bb0745effc9698ef\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/031aa34aeb6c4ac2be5681bcc1738b72 : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbffc113b4fb43dfb15ecc8bc5451aa9 : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/813c8209c28743e58b0125441940f03b : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63f2bfa94b0246859c3fd0d30f8ac8ec : SUCCESS in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d40e10332264af093eeb888d865906d : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af9587003ee24029af201981667f3f4c : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5759cb5e59b419ba67786dc6022b067 : SUCCESS in 12m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/11caae004f1346fcbe8415410dc9bdeb : SUCCESS in 6m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/bbee727469eb4a278dbc79147e8814e5 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23a3e0f5af3043849ec5cf5fc0b4e314 : SUCCESS in 4m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2393d8f86c1a4caf9c2d7532eb978b4b : SUCCESS in 6m 26s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/843c756ecaff4275bb7a4aa7034679b3 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e69df5f1284438fa07d39ffe010e31f : SUCCESS in 1h 17m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/55748e44cfc947768681d72290015d16 : SUCCESS in 2h 11m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d1010dcf533b491ab06d3439ee06da65 : SUCCESS in 1h 11m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c25900f9f144d50b73a081b18f08ff6 : SUCCESS in 2h 04m 57s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe577ff2462644fc9bf6f21f5e08b7d6 : SUCCESS in 1h 20m 26s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9842f4c5bc6f4b5c9ca3b3bb60570bda : SUCCESS in 2h 34m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fb0b4a0f64724b11bc7d6d104457c64a : SUCCESS in 1h 14m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32a519021f9f4a6fa74c149a1f176236 : SUCCESS in 47m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f00689f865b41a2a7c47c5ff137be22 : SUCCESS in 52m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f5c98eae5904337b443da92ddaf5343 : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0a25212fe7d43a1a26a7ae5ff564682 : TIMED_OUT in 3h 04m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9adf5a5bf0424098888e0fbce55b190c : TIMED_OUT in 2h 34m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16cfd2950872c9b41129f91e2ad992849c9773a3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-20 18:08:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"775db87bddd1510cb79c05b05c0cc295afb0012a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-20 19:56:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"66edd5df2a3df8ad2e368d1a39abb071875593ad","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-06-25 14:07:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6672887df33a225c32d20a3922684ab26547ba0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 14:08:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a359f821cfa66ac185a0bde26d0d36b843d69fe9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:37:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d69556cd70e4c7c92099dfc70beebdb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f623e9fbbdb4453cb9a739adeacecfba : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b5a0845cce4d9685290442a52810b2 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32153ae7ab744b7aa9df1ddb10cdd551 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84ca2614877f470d9c2ec499e618a656 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c43ceba972b453fa125a38aa15c12ab : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d04507fe37e4c4d9ec254bd0cdc1f1b : SUCCESS in 3m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7494f6e0228c4ce289210cd4dabee913 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cfbcf54b33cb4b1581d08a70c1916e57 : CANCELED in 7m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ec4e5276acb540b1be5a090e686027ae : CANCELED in 9m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/51608ce6d0d24ce4846104f320caa791 : CANCELED in 8m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9a0887f8f9d431eb5e159d0133ad5b2 : CANCELED in 9m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ed0c531e5bfd4f349b3e84323551bf72 : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f34479a929919bc5fd093041bba52e8427b5813b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-25 16:39:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2451f942d97c5bd0a12d3dd2b70139a7547a5525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 19:39:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b8b580ede8496d84e3285639d988b3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3fa19f3e1ed2453eba156c5389c841d9 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a7646225c514c178f0aef41078649f5 : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12cb41762884adf909b761e8837ca69 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01cce57be7a54cb1afd31e68038eef34 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2078ea2671b249ac815584b6a8424efe : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d466c437c5f4333afb8e1096fee4e2c : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d406778899471fb41972d0bb509f51 : SUCCESS in 12m 00s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/56405efb7b104a64a77893924bef82ba : SUCCESS in 5m 33s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e940714baa5f449593ba33702f7c58b6 : SUCCESS in 7m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0a7a7b3c51db42b9985c00fa161e7f10 : SUCCESS in 4m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a9af276f2d2e483699c4adcb0007990e : SUCCESS in 4m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5ee581dbf02f4782afb69615a3b62f80 : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4eb2af46502a47f99918482abb0b2819 : SUCCESS in 41m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fb35a9999c014e1da61dbf4b0075d318 : SUCCESS in 1h 54m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bd206cffde7f41ca8e80eb7de8de923c : SUCCESS in 57m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8978770df6bf406da741184a48720dfe : SUCCESS in 1h 48m 15s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/819f4969d0a248a4b16388061d4d052f : SUCCESS in 55m 10s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7d4daad562ec40c7b011bd99d5372b20 : SUCCESS in 1h 56m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/710d3761043e401bacc7325563901f74 : SUCCESS in 1h 05m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01071a6d4bb54787840a8d98e497ea8a : SUCCESS in 37m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c9bf5def41641e59cee9eea9c382829 : SUCCESS in 37m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/531c50adb4654c25831272978ed88e12 : SUCCESS in 2h 24m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be51f21b97b848d2ac4d89163634f36d : SUCCESS in 2h 57m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a3afc533917f4abeaea09163b71261cb : SUCCESS in 2h 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9dd063f446bf160f2843011662388afa628d4211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 19:39:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"553cdd57c6b4c7ab493c93853af89c7895be032a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 21:51:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d1173ad4114d7da3c6d962cedb1979\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3a49a0fdce3848be833d3db2887d91b5 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885f539e60524a66b07a4c33ad6c4149 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/453a1a6de4a045d6bdc82c9fb3837617 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b8e1c0a0a09439db347beab0c0b66d3 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f40922f2664b94802147154f0801d8 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/55104c2592fc4a2c908b1a096e566e2e : SUCCESS in 6m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b6e58179a3434d99ac7a93136fc7e324 : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6494b86094644c99a668394f1337f53 : SUCCESS in 1h 17m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/227ff4f08a1c4510a0e2488c9c2aaddf : SUCCESS in 2h 05m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ddb5266a77614920973a18853d3bb3bc : SUCCESS in 1h 21m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abd381e187994d12b6611413c11383e6 : SUCCESS in 46m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0ce39f0aff549189a8e8242e2a22a1d : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3574fc0b99560e7f5c9f4149927ec4ca29594441","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 21:51:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1106b027d0f3e9fdba139b7e904eb2d2bc63dae3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 21:53:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c4c26a3a36540eea4aafc71ea8ff758\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d19a25a32db4337af05324b712a465a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e852a72969a452f30b738da63e8e788dfd6d9a81","revisions":{"e852a72969a452f30b738da63e8e788dfd6d9a81":{"kind":"REWORK","_number":1,"created":"2025-06-18 15:38:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/952844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/952844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/952844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/952844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/952844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/952844/1"}}},"commit":{"parents":[{"commit":"2fb6ba9343e39f60f058a781afd14b906dd84daa","subject":"Fix SSH disable during amphora image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fb6ba9343e39f60f058a781afd14b906dd84daa"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:36:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-18 15:36:50.000000000","tz":540},"subject":"Remove zake from optional runtime dependency","message":"Remove zake from optional runtime dependency\n\nzake is used only by tests, and is not used by actual functionality.\nThe optional runtime dependency was removed from taskflow by [1].\n\n[1] https://review.opendev.org/c/openstack/taskflow/+/939000\n\nChange-Id: I3929e856fb97dec25fd6a9ddd5d1988bd091b409\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e852a72969a452f30b738da63e8e788dfd6d9a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e852a72969a452f30b738da63e8e788dfd6d9a81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
