)]}'
{"id":"openstack%2Foctavia~957097","triplet_id":"openstack%2Foctavia~master~I50837b6f6a832c5bca37226fbc50d1160907a91b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-10 21:22:25.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-08-19 07:21:14.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I50837b6f6a832c5bca37226fbc50d1160907a91b","subject":"Use native String type","status":"MERGED","created":"2025-08-12 08:50:25.000000000","updated":"2025-09-10 21:23:26.000000000","submitted":"2025-09-10 21:22:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957097","meta_rev_id":"537f90701a87fd1bb77412c42beee30a0e94ff53","_number":957097,"virtual_id_number":957097,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-10 21:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2025-08-22 05:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2025-08-19 07:21:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-08-12 18:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-09-10 19:31:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-12 11:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-12 18:43:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-08-13 07:18:24.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-08-13 07:18:24.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-08-22 05:40:59.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"4468ab09541f2e480e7336a212980180a921703c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-12 08:50:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04f2361ecda6275174fe4b1c7e7378a033e4321d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 11:34:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d28cc8216147a3828b94bba65184eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f63c8067c9347bf824ba2c4e5c406ba : SUCCESS in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607f0d37dcb0463792b4205c0e28c78f : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd2464b9351e45a3baccaea9da14a8b4 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f92fac49f78465990364f91e0c4e0c0 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8041dffb34e6487fb7f3639ea393b147 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c578c7d570484199935973ed4ec8ce0c : SUCCESS in 14m 01s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/031b4470816a4fd893e41f1a2903c807 : SUCCESS in 6m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/59e080a272c149fe9446d314e92c9b54 : SUCCESS in 9m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/827fe7e189054374905cfbb6401e38cf : SUCCESS in 1m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd0ae275a25c4c289c15f9073917ac9d : SUCCESS in 5m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c4ac3aedecd14172b8871e8a1cfe0647 : SUCCESS in 4m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30f8219177b0414293cff475249a2a5e : SUCCESS in 1h 25m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ce7ffcd954554712a618d97656c94504 : SUCCESS in 1h 51m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/34bdefe1c93141869d5a4c23d596b6af : SUCCESS in 1h 07m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a495032772c41cd98194e36bf071520 : SUCCESS in 1h 45m 24s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a17db60a79a4f8cbddb41e8a9477330 : SUCCESS in 58m 38s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dec36325cd264b9ea822b9b618d7c0a7 : SUCCESS in 1h 56m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/134d153c7cd340ca9c828865703689d1 : SUCCESS in 1h 00m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be9cbddeae644ce3a8efead51b7ca29a : SUCCESS in 36m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b15a1861a9b4456bb16c5e274ba4d27 : SUCCESS in 41m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48163e17ae6c491f848cacd325da6637 : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1579438112b24fa5b5d206facab307ea : SUCCESS in 2h 31m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a415ba3adead4e08aa064bdbc5cfe5ff : SUCCESS in 2h 42m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f66450ec814649a198ea3aa7ff85a1d2 : SUCCESS in 2h 25m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3229502425bd5bd7cc4d41717e4021d6d03d79a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-12 11:38:28.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d803517d6dfb1ed7c2dd7dd18c1d71fde6cc61dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 14:43:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898583b81a1a4cef9a093102735b3e3f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ff4b83ac8b14c79aef90754379adbd9 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e4e15ded414621b2d0c10e88d75589 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6678041d1a6e4bdc9000c60b2436a29b : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/79492e0f3a3f480e9c86abf64a199353 : SUCCESS in 6m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10bedb8f29a240cb915e541a19cc1765 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93167ec8c6c42ecbf011336a64f0371 : SUCCESS in 13m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9a01fdeb50b34fa89258297b40bd2b05 : SUCCESS in 5m 16s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/16b2d27801c1426384b3e7cd22959bae : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/87154c88dd544347bfa7d4a43da5e43d : SUCCESS in 4m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41b036f9f66a43469cfcb05f40274c13 : SUCCESS in 7m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/da7419dc2ae248f49c93c0daf8bbe6b3 : SUCCESS in 9m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/052015a94599462aa898e0c5d298c5f9 : SUCCESS in 1h 32m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3e4e3400162e48d1a93b68de36ed97a4 : SUCCESS in 1h 55m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1dcdd1774a084ef3a846fcc8507d3d09 : SUCCESS in 1h 11m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/549a26ed29174f3281d228d86afd65b2 : SUCCESS in 1h 44m 16s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0b1c58f18364d9c8690d6743edf1461 : SUCCESS in 1h 05m 52s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f715ea3fe7094fddb86751643dc33450 : SUCCESS in 1h 59m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2f8d694c7ea34a5fbedb4dee9be32174 : SUCCESS in 1h 13m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d29ecd09bc344604a5ffc55f8c213280 : SUCCESS in 39m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a336dfd43fa44ac2baf3323bc8a7e06e : SUCCESS in 42m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0a64f5f4d9e4a22a4f22dbf4b7ca869 : SUCCESS in 33m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6857f0cf1e7c4cecbca78f84841662ca : SUCCESS in 2h 27m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f2f9276921146eb8190711890367dac : TIMED_OUT in 3h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cc78f5ef0b4742b6add9b4c3cf4f129a : TIMED_OUT in 2h 33m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3dc5fc12e15a86cc856de8153e704aecba62826","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-08-12 18:43:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8e06d409e2fce518281e04fbfc87ece3164d383a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-19 07:21:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a487caa59a0519125bda0c8a91a9497f04c38852","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2025-08-22 05:40:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a7f15d3387b42558d6925ef077c252aa115bc033","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-09-10 19:31:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e0eeff3bb67735df67e933c8ca9f68b379dcf71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 19:31:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2db0d0e0a3bb7ef98c4b0c3c75946ec6ab837f90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 21:22:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aa3075eb47c4b56bcbcedd9e305b511\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c42532d42445b482f5a67a551f5695 : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd54b058a0114c29874b5232e7731451 : SUCCESS in 6m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d81879873c14055b68b029d93341ea0 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f41847aa98345fca734854b08e2c895 : SUCCESS in 12m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c1b14845de64b15929e2eaffb12220b : SUCCESS in 5m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4e34814afec249bab81a960d64365f77 : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0ce0cb57c864873ad72faa42bf2afd4 : SUCCESS in 40m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/83a906fd1f5c461c8321fdf5a35c6173 : SUCCESS in 1h 50m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/22398a2025ac420db2668b2e366e65a7 : SUCCESS in 1h 08m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/426c672b7e5446c3979414b7944add32 : SUCCESS in 35m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c577899c0d44f7e8eb1749ef4052dda : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":2},{"id":"590ae565cc5585238e362874600e37cc26647cc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 21:22:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"537f90701a87fd1bb77412c42beee30a0e94ff53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 21:23:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce0ca526bde48fa8112249081d9e79b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ba279bfa774aafb4edd4999bb92c20 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5fc42b528e93f86b4198e373c7360e2f1c897465","revisions":{"3c0e1cfb148b9f198eb674bf805af617bc142b63":{"kind":"REWORK","_number":1,"created":"2025-08-12 08:50:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/957097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/957097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/957097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/957097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/957097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/957097/1"}}},"commit":{"parents":[{"commit":"f88e706c01b0c851faced2be255dddb568e74ace","subject":"Merge \"Remove deprecated datetime.utcnow()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f88e706c01b0c851faced2be255dddb568e74ace"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 08:49:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 08:50:05.000000000","tz":540},"subject":"Use native String type for","message":"Use native String type for\n\n... instead of own type class, to leverage common implementations\nbetter.\n\nChange-Id: I50837b6f6a832c5bca37226fbc50d1160907a91b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c0e1cfb148b9f198eb674bf805af617bc142b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c0e1cfb148b9f198eb674bf805af617bc142b63"}]},"branch":"refs/heads/master"},"5fc42b528e93f86b4198e373c7360e2f1c897465":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-08-12 11:38:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/957097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/957097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/957097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/957097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/957097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/957097/2"}}},"commit":{"parents":[{"commit":"f88e706c01b0c851faced2be255dddb568e74ace","subject":"Merge \"Remove deprecated datetime.utcnow()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f88e706c01b0c851faced2be255dddb568e74ace"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 08:49:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-12 11:38:28.000000000","tz":0},"subject":"Use native String type","message":"Use native String type\n\n... instead of own type class, to leverage common implementations\nbetter.\n\nChange-Id: I50837b6f6a832c5bca37226fbc50d1160907a91b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fc42b528e93f86b4198e373c7360e2f1c897465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5fc42b528e93f86b4198e373c7360e2f1c897465"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
