)]}'
{"id":"openstack%2Foctavia~958283","triplet_id":"openstack%2Foctavia~stable%2F2025.1~Id90786ce9c451f4d494da789506014f8cf44e850","project":"openstack/octavia","branch":"stable/2025.1","topic":"ci-dbg","hashtags":[],"change_id":"Id90786ce9c451f4d494da789506014f8cf44e850","subject":"DBG: print venv info during grenade verify","status":"NEW","created":"2025-08-22 15:08:27.000000000","updated":"2025-08-28 19:18:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a83d49b17460ca31dfcb820e823a455a51ecc174","_number":958283,"virtual_id_number":958283,"owner":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-22 21:17:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-22 18:00:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1f47a02437e9b381ee66ac2941d6311c287153a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"date":"2025-08-22 15:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db8b768c581e9f9fa7cbaea9f511c2d9082806e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 18:00:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f903c18abf0645b88608fc815922c006\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5828f88280834f1e8f445a30d178fccf : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ccaaecb02f4c0bab58066a1c0195a3 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baad7fff17fa43c6a3a9cdf7beb583c2 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8311892c2f0844e199ea97e0061407d4 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034c14b5db2b4abcb0a8399fd7f3af53 : SUCCESS in 12m 26s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e7d0f020077a4c4daf6db2f3af108540 : SUCCESS in 4m 18s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/79ad82782aa54829908187f37e88a5d4 : SUCCESS in 4m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/165284534675464aadb66d0a52f413d4 : SUCCESS in 4m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5e89c5c66fa941baaa522391da36cb76 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62908760ddb34c45b974167b3cdcb4dd : SUCCESS in 5m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/15a5d5176cc04ceebd5200194ff8dcee : SUCCESS in 8m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0ee85cca1eef4315bb27b79d9cf5b8a7 : SUCCESS in 1h 18m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cc3512c242c74c67ad5f07e47df59158 : SUCCESS in 1h 54m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dfb313d324d94845b480f7b9741cda91 : SUCCESS in 1h 08m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99e22c0778c24194b550ac987034a0d0 : SUCCESS in 1h 44m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/899099a3c6f542b2a142c4a1398ffac6 : SUCCESS in 1h 16m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32bdcab98c5145fa842a53569820737d : SUCCESS in 1h 57m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4e65e0b49f3748d5950b1b5e2cf96535 : SUCCESS in 1h 08m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2c17cf9544f460db16fc5a1a9e9c26a : SUCCESS in 41m 03s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c886ec7283154e63a06f924255bac2a8 : FAILURE in 36m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8753f331fd42488780db41f558ba976f : FAILURE in 42m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67a73659ac6a43a4888d82ba230d83fb : SUCCESS in 33m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2fe719f7b15a4b29a44a62ca5e413e78 : SUCCESS in 2h 25m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/782b3231c6fa4b2c9cf0d1f4699fb58b : SUCCESS in 2h 51m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/047b8f3011f34adc908b5b7461283970 : TIMED_OUT in 2h 33m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6377f903e0174c5da4ebe63bd9e6f8a9 : FAILURE in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2abb1fda3757820fd9a058ace646693233e13985","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"date":"2025-08-22 18:13:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3bd25256f41d306231bf1145aa3939a46b7b3f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 21:17:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db62f52fce7f4014851bdf1621790f0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/590b55a22e3c475f95f213b37e7998bf : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceac374be3de49b5a31215e5b670bc0f : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/732b553a689d43d1837489b24e632238 : SUCCESS in 7m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62c90837866740ccbb019b91cac8f97b : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c171912ae9f40b4a5360946ccd64e1a : SUCCESS in 13m 50s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5c5ad965a7254370aea4ab4b826900ea : SUCCESS in 7m 01s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/da134dd5736546c1bc568133839c9488 : SUCCESS in 9m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d4cac3170ab6472cbf18a8ae150b8e79 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a17d798e9c9c4612b6c3fdb3da4f0ab3 : SUCCESS in 4m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df89a0e6632546499d1320281ee8fea9 : SUCCESS in 8m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/95b672425fa84ca38761f67bb94dc951 : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4822807ab4b3497e83d67207a79ee240 : SUCCESS in 1h 41m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1dc26cb4e0404219993f423b81615d6c : SUCCESS in 1h 48m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/30eb946af9b14a9ebb4687d16dbc658f : SUCCESS in 1h 31m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/179ef737f5ad4fc08d29427878224245 : SUCCESS in 2h 29m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02bdd7689f464368a6be24e058d9b1ea : SUCCESS in 1h 36m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e62290fd044940c8be3e73baab7953ef : SUCCESS in 1h 53m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b2a822a6afa547a8b4fc9cc5378c7ef7 : SUCCESS in 1h 26m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c926d2add41440eaae647633b3b30337 : SUCCESS in 24m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35ba4ff51e3948b6ba08a82d191aaa05 : FAILURE in 40m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d73cbecda5e14c7581b64c96a88a219c : FAILURE in 44m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f906e1d96b064247bdda70764b559bcb : SUCCESS in 38m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c18446b8e0964166ac55434adfd7e6c0 : SUCCESS in 2h 28m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f8ef9e19eac4bc2a657710962f7bedf : TIMED_OUT in 3h 03m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6475488ad0f449cc80231b3faa89e7cd : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8f142d63b9947179a9659789f192877 : FAILURE in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a83d49b17460ca31dfcb820e823a455a51ecc174","author":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"date":"2025-08-28 19:18:30.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1e7dd8720f25a3f55a63c0733f84fdaf4dc56b9","revisions":{"c00b82682e894935d4b4496033b52820d0011114":{"kind":"REWORK","_number":1,"created":"2025-08-22 15:08:27.000000000","uploader":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"ref":"refs/changes/83/958283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/958283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/958283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/958283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/958283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/958283/1"}}},"commit":{"parents":[{"commit":"b8824ede82c42b81ca35248bd6ab6fb8bf51f3cd","subject":"Add an octavia-wsgi compatibility script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8824ede82c42b81ca35248bd6ab6fb8bf51f3cd"}]}],"author":{"name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","date":"2025-08-22 15:07:02.000000000","tz":-300},"committer":{"name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","date":"2025-08-22 15:08:19.000000000","tz":-300},"subject":"DBG: print venv info during grenade verify","message":"DBG: print venv info during grenade verify\n\nDO NOT MERGE\n\nChange-Id: Id90786ce9c451f4d494da789506014f8cf44e850\nSigned-off-by: Wesley Hershberger \u003cwesley.hershberger@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c00b82682e894935d4b4496033b52820d0011114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c00b82682e894935d4b4496033b52820d0011114"}]},"branch":"refs/heads/stable/2025.1"},"a1e7dd8720f25a3f55a63c0733f84fdaf4dc56b9":{"kind":"REWORK","_number":2,"created":"2025-08-22 18:13:49.000000000","uploader":{"_account_id":37881,"name":"Wesley Hershberger","display_name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","username":"whershberger","status":"Support Engineering @ Canonical"},"ref":"refs/changes/83/958283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/958283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/958283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/958283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/958283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/958283/2"}}},"commit":{"parents":[{"commit":"b8824ede82c42b81ca35248bd6ab6fb8bf51f3cd","subject":"Add an octavia-wsgi compatibility script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8824ede82c42b81ca35248bd6ab6fb8bf51f3cd"}]}],"author":{"name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","date":"2025-08-22 15:07:02.000000000","tz":-300},"committer":{"name":"Wesley Hershberger","email":"wesley.hershberger@canonical.com","date":"2025-08-22 18:10:51.000000000","tz":-300},"subject":"DBG: print venv info during grenade verify","message":"DBG: print venv info during grenade verify\n\nDO NOT MERGE\n\nChange-Id: Id90786ce9c451f4d494da789506014f8cf44e850\nSigned-off-by: Wesley Hershberger \u003cwesley.hershberger@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1e7dd8720f25a3f55a63c0733f84fdaf4dc56b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1e7dd8720f25a3f55a63c0733f84fdaf4dc56b9"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
