)]}'
{"id":"openstack%2Foctavia~959405","triplet_id":"openstack%2Foctavia~master~I0404b3d1af16fce77f1eb6b1b0af05e9bdfb2c2d","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-30 07:32:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0404b3d1af16fce77f1eb6b1b0af05e9bdfb2c2d","subject":"Stop explicitly adding None as a choice","status":"MERGED","created":"2025-09-04 01:26:16.000000000","updated":"2026-07-30 07:35:33.000000000","submitted":"2026-07-30 07:32:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959405","meta_rev_id":"468f1927faed11884b19e483f025132d680d77a7","_number":959405,"virtual_id_number":959405,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-30 07:32:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2025-09-18 13:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2026-07-30 05:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-04 04:21:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 13:59:14.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"c71d9831f70b6f0ce3c94a84ce8b8f7b799b9fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-04 01:26:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"191c8df22d571f30cbf9aeaed865756bfe543fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 04:21:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/759de9e436c7429eb45de1f90b9ccff7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b042dd0b4fb44848443328d17d1aa1b : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f723806185154a6284c2a133a48b38da : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5d481c9f83647de8f407a6e46efac86 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01e1c6a461f34e35bf366b5af8a966a7 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2f1e2c7241246979350c995efed77eb : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f15536fa9794096af74b5e33bd77471 : SUCCESS in 6m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/98f535b8be8c477cbeb452b2ee8fc13a : SUCCESS in 8m 00s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f9c5fe651e504bfb96926a8342701343 : SUCCESS in 4m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ff954c9f84cd4873bda255d18a34294d : SUCCESS in 3m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d7733ea46e2472db2fd092e400909a6 : SUCCESS in 5m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e6ec85ab54124cd7a5b9f55c5fd5ae83 : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ac00333f61a4d85bfcde1f5c55674f4 : SUCCESS in 41m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/06e5287f4aa540f9bacf5a5fc0b6023d : SUCCESS in 1h 56m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/01b6055367324ae88857ba1b614e7db3 : SUCCESS in 1h 11m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/98399b67a23845f691963b248f550251 : SUCCESS in 2h 00m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6a4623ec682243a5b11bdae8d000e11e : SUCCESS in 1h 11m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dbe6a8bd4ec1492db3d75419f5604137 : SUCCESS in 39m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d4f393edc99452ea03792d0c8a967f7 : SUCCESS in 46m 00s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36d7e9c3eac34c508626b8b9f69ffa9c : SUCCESS in 37m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a89392f49224750844ee807de606dfc : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d94f9e83eff405b9dff8348c69c78e8 : SUCCESS in 2h 54m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/61d79a2a7bf44f02b81533214e436d4e : SUCCESS in 2h 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26181bfa447c87ae387ce590afba0b3326748197","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 13:59:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"513994d4916608b73b51a0e4db57fc86666d41a0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-07-30 05:50:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c697e3b67ecaf96e59cd16ffa5644aafec1cc535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 05:51:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"869642f19d4745d6e60243de2229b9b586f66d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 07:32:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef14e803056446a9971df0adbc6a0beb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd7bdc500fe47e497ad34bfd8f3d6a7 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30e7d1ad554849ce9a134737a4fe2b11 : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/996aeb3d69fe427fa486a9b805b06e70 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93aa8654fc1149539b142e4aa1157ce6 : SUCCESS in 13m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d472c676f2314636951a390c02d895e9 : SUCCESS in 4m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/40eda0930f08490783c6a826d7d95bc8 : SUCCESS in 3m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/43ecb5ce6c5c459cae9bc6659869326a : SUCCESS in 5m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4dcd11efc137471ab186290462e6c6ed : SUCCESS in 59m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b4ee4033757e4f14b560fdfde4cb8e77 : SUCCESS in 1h 36m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/13643ba68afc447d9e262e257dab13a3 : SUCCESS in 1h 09m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b96d6e5f7d27478dada9fe9e9ac18bc1 : SUCCESS in 42m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e24c6a3dfca4f8e9fce13902e99a172 : SUCCESS in 52m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9ff36494c4124745adaaab0dd9a08789 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":1},{"id":"01b87871c7935ea2728a4aa8028ca04ee6b0e839","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 07:32:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"468f1927faed11884b19e483f025132d680d77a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 07:35:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed139e6746f459dbd419a59af7fbbde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6804a864835c4a7aa58fb2bf49ebe920 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac6e84302d6f673e71bf260d1a97d264bdb40354","revisions":{"ac6e84302d6f673e71bf260d1a97d264bdb40354":{"kind":"REWORK","_number":1,"created":"2025-09-04 01:26:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/959405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/959405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/959405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/959405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/959405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/959405/1"}}},"commit":{"parents":[{"commit":"00599cb91ce9b5ab80119e329845e54f6e1e0a10","subject":"Merge \"Reduce tune.ssl.cachesize for HTTPS terminating listeners\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00599cb91ce9b5ab80119e329845e54f6e1e0a10"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-04 01:24:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-04 01:25:38.000000000","tz":540},"subject":"Stop explicitly adding None as a choice","message":"Stop explicitly adding None as a choice\n\nCurrently oslo.config provides no mechanism to allow users to set\n\"None\" in their config. None is a special value which is used only when\nthe option is not defined in a config file.\n\nChange-Id: I0404b3d1af16fce77f1eb6b1b0af05e9bdfb2c2d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac6e84302d6f673e71bf260d1a97d264bdb40354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac6e84302d6f673e71bf260d1a97d264bdb40354"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
