)]}'
{"id":"openstack%2Foctavia~961053","triplet_id":"openstack%2Foctavia~stable%2F2025.1~I4a6e4830d0e916b1af1a11dd10097980a57d97ea","project":"openstack/octavia","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2025-10-17 06:44:23.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-10-17 03:22:42.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I4a6e4830d0e916b1af1a11dd10097980a57d97ea","subject":"Fix race condition in cascade delete","status":"MERGED","created":"2025-09-14 03:35:08.000000000","updated":"2025-10-17 06:45:40.000000000","submitted":"2025-10-17 06:44:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961053","cherry_pick_of_change":897466,"cherry_pick_of_patch_set":2,"meta_rev_id":"97d2ae03384a4bdb63bc5262ceba1c0588a20509","_number":961053,"virtual_id_number":961053,"owner":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 06:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-10-17 03:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-16 22:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-10-17 03:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-14 03:35:08.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-09-14 06:41:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 13:48:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-10-16 16:51:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6d8d69da9e71bd4e025a33560132130b306e1b92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2025-09-14 03:35:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fc542e6ee16534f7039ca394a575fb8ac1381857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 06:41:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08e5bbfdbd964f61b6f1c25831f51c7c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48a6d6e9722d48e1bae10a705b2a49e1 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae7d5a86dd043eca1c73f22a679e149 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ed1e5dc7f104e2ab263d7733f4e8a0c : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08e1f60ea01a407b93dd221dd3001e84 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d16722026d04b2190769fc563afc6aa : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26ddd5a8dc6749518b1d6788bb100878 : SUCCESS in 5m 01s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9c0f44e6f97e4661adfe6b04c1c527e8 : SUCCESS in 7m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/4afef8c2935946c2bb5c92b96afb8d4d : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/abc1d1e5138d4a4293f89713c766ed27 : SUCCESS in 4m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/548ca32d8c6647ac966cbd90c26e22a3 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c6f7899d0df642aea4bdcaf65ebc8e9c : SUCCESS in 7m 01s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7c0b8bfc0dd941b39d67922fdcffd646 : SUCCESS in 8m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9b124c7bda8041c19e41dbc00108c11d : SUCCESS in 1h 08m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9f1347e7053c4e86bc7c29b9607ec7ce : SUCCESS in 1h 47m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0476d063030545d6a998eb7011f3b5f2 : SUCCESS in 1h 02m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efb461423bbd422fa97012345282a8af : SUCCESS in 1h 46m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50c91bf73c0649958123ad0a2645aa2d : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d9dc1e98d9004f44ae4850328d0ab0cc : SUCCESS in 1h 50m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/040e616424944eb9b41d5299debc56d7 : SUCCESS in 1h 14m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49dbb474d92545d980d2cbda5cb31387 : SUCCESS in 38m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa879aeb415f458b8ea1b5c202991c4a : FAILURE in 1h 02m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/82dadd43c8a04db6b4cf9be4aaed4831 : FAILURE in 41m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8cee87bb36e420ebbe05ce5299e6f37 : SUCCESS in 59m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bfc55b256fd2437abf7cd0975dbac511 : TIMED_OUT in 2h 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a724867f7d834c83a3e7ba6d5bc84271 : TIMED_OUT in 3h 03m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/50f39d5ae8864530ae00d4173e5b83d3 : TIMED_OUT in 2h 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab4729ebab434df2aa2dcae4a61475ba : FAILURE in 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6da94635b83348d827f432eac6fae7fd10e91b14","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 13:48:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c17ae3ef18d7337fbc4c0f9d1478d82462a1b301","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-16 16:51:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"747cf9f5d0150ab001b02437b8ddfc1961c83713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 16:52:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e22c1214b66216a1abd79500a59fa2d6dfc71f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 17:04:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/670f4faf47064bc5bd1c51f1b09e5f30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9397eecc6e1c427baed3325b3aa9d629 : FAILURE in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e848cad3b5d84c2cba83b9bb56402d70 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90f14453ad184b07ba18fa26763be9f5 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c28122fbc44b7e9d14e785a944951b : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d152bd8c555e485293399933a40188bd : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/59bc9542955847f49e320585528777db : SUCCESS in 2m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/664bfb4fb9fc47c6b87fe8d2357f7b06 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ef161eee6e94a66b11b97ebc3180a25 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5fae1310401144c89587a741452ab454 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bcfec284b70e4b63929b9b5c312ea473 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d723ede3db047ef99aabe1bfcd87c37 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cddc7732ca90424ebcfd2d9cdd1d7a9c : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/utils.py\n  Comments left for invalid file octavia/tests/common/sample_data_models.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/functional/amphorae/backend/agent/api_server/test_server.py\n  Comments left for invalid file octavia/tests/functional/api/drivers/driver_agent/test_driver_agent.py\n  Comments left for invalid file octavia/tests/functional/api/v2/base.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_health_monitor.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7policy.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_l7rule.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_listener.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_member.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_pool.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/db/base.py\n  Comments left for invalid file octavia/tests/functional/db/test_models.py\n  Comments left for invalid file octavia/tests/functional/db/test_repositories.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_amphora_info.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_haproxy_compatibility.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalived.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_keepalivedlvs.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_loadbalancer.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_plug.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/api_server/test_util.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/agent/test_agent_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/health_daemon/test_health_daemon.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_haproxy_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_interface_file.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_ip_advertisement.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_keepalivedlvs_query.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_namespace.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_network_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/backends/utils/test_nftable_utils.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/haproxy/test_rest_api_driver_1_0.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/health/test_heartbeat_udp.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/jinja/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/amphorae/drivers/keepalived/test_vrrp_rest_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_get.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_listener.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/driver_agent/test_driver_updater.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_agent.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/api/drivers/test_driver_factory.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/auth/test_barbican_acl.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/common/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/generator/local_csr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_barbican_legacy.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_castellan_mgr.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_local.py\n  Comments left for invalid file octavia/tests/unit/certificates/manager/test_noop.py\n  Comments left for invalid file octavia/tests/unit/cmd/test_health_checker.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/logging/test_logging_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/lvs/test_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/jinja/test_user_data_jinja_cfg.py\n  Comments left for invalid file octavia/tests/unit/common/sample_configs/sample_configs_combined.py\n  Comments left for invalid file octavia/tests/unit/common/test_base_taskflow.py\n  Comments left for invalid file octavia/tests/unit/common/test_config.py\n  Comments left for invalid file octavia/tests/unit/common/test_constants.py\n  Comments left for invalid file octavia/tests/unit/common/test_data_models.py\n  Comments left for invalid file octavia/tests/unit/common/test_decorators.py\n  Comments left for invalid file octavia/tests/unit/common/test_exceptions.py\n  Comments left for invalid file octavia/tests/unit/common/test_keystone.py\n  Comments left for invalid file octavia/tests/unit/common/test_policy.py\n  Comments left for invalid file octavia/tests/unit/common/test_utils.py\n  Comments left for invalid file octavia/tests/unit/common/test_validate.py\n  Comments left for invalid file octavia/tests/unit/common/tls_utils/test_cert_parser.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/compute/drivers/test_nova_driver.py\n  Comments left for invalid file octavia/tests/unit/controller/healthmanager/test_health_manager.py\n  Comments left for invalid file octavia/tests/unit/controller/housekeeping/test_house_keeping.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_amphora_rate_limit.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/test_task_utils.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_amphora_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_health_monitor_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7policy_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_l7rule_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_listener_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_load_balancer_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_member_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/flows/test_pool_flows.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_amphora_driver_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_cert_task.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_compute_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_database_tasks_quota.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_lifecycle_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_network_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_notification_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_retry_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/tasks/test_shim_tasks.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_controller_worker.py\n  Comments left for invalid file octavia/tests/unit/controller/worker/v2/test_taskflow_jobboard_driver.py\n  Comments left for invalid file octavia/tests/unit/image/drivers/test_glance_driver.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_base.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_utils.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/statistics/drivers/test_update_db.py\n  Comments left for invalid file octavia/tests/unit/test_opts.py\n  Comments left for invalid file octavia/tests/unit/test_version.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/noop_driver/test_driver.py\n  Comments left for invalid file octavia/tests/unit/volume/drivers/test_cinder_driver.py","accounts_in_message":[],"_revision_number":1},{"id":"075d7314da9084d512901130df8930e9e60ed986","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-16 22:39:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_5112\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"}],"_revision_number":2},{"id":"31d73b3fb1def1cf0006e1997f8e574cda038f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 01:29:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850be2aca2f34a1087724fca8747cb38\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a49b50f8c2d4afdb9c6e49195eba766 : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee951c434b54cda9cb6b12dd852e006 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78edd4b0487c4579bc74d14e2494b7a5 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9231e800f3e84379b2b8262f269e2615 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7012bf3db6f4e0fb7cceb7ffdb8c96a : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/362383c75e7045a89a7f738af62ed281 : SUCCESS in 6m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/03c2c7d7805a4dbe8e6ef15ce83c822b : SUCCESS in 5m 16s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/22438d5c850449d3b7d9f954e8cc0bff : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0caf81dfacdd4400b17976795e35782f : SUCCESS in 3m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a578dce357b421a89d96c79e812784c : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5470c02033d44cecb3f0910aa7e3b633 : SUCCESS in 6m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/aa1f72dee0794cde8f3cdc6f02f81c1d : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7af400afc131474eac24278a6059a51a : SUCCESS in 46m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fa91a13c5c594ccf97de9602928ee035 : SUCCESS in 1h 44m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5a87fbca544b45648f7d81bf0ad24657 : SUCCESS in 1h 08m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/759c3c1fe7374ec38526e804137f3769 : SUCCESS in 1h 48m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3135c28694bc41d0aa6edc10df671100 : SUCCESS in 1h 09m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f81441aa59f94edc8141c736b2154d8f : SUCCESS in 1h 52m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5984884443d54c8aa09e458fe256eb2b : SUCCESS in 1h 15m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18922ccf22f849d783ff60a66dba31c2 : SUCCESS in 35m 45s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7da7eab69e3547b18f6ee9b75db58064 : POST_FAILURE in 35m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/14d209ef83104dbf8116ea550340a330 : FAILURE in 42m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e6b013aecc747b79eb8fc8e148e902d : SUCCESS in 40m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/95adafd39b764bd38dde92fb20f19afe : TIMED_OUT in 2h 33m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9977fb06cc2b4f7cb1b74fbc9a90752d : SUCCESS in 2h 48m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b6d1f51c051d45d1a015bf70b10b794a : SUCCESS in 2h 13m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7ab41097dac4130b046b56c86fbeef4 : FAILURE in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a63506d55144fbeac89db760ad50cdf3336ba92","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-17 03:22:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4666d152d696292265f5987deb94214f49232b1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 03:23:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ea419d98f7900ae82401a8867ab6262e54fe9775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 06:44:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c662bdc6bd1f42988b2894642f11620b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4a8fedd8e74d309b941f24b5a6b94d : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eadd98a42a248de864eb6550ba8d4d2 : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec648bd197c24efb9f4607dce392ccdf : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a5c84b0fd349fe97bd7a60f5c1190e : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e5c593aa6614a198b560b410204e0e5 : SUCCESS in 4m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c9f7d4ec0709400aa51f24b2f170021e : SUCCESS in 3m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/38c3634af11c4d668cca96c0ab519c09 : SUCCESS in 4m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b2646cad707a4baeaec6497f18d9282d : SUCCESS in 34m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/778a9ea7fb164373806204ea67e35709 : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a507ea70c9204f68899380de8629924e : SUCCESS in 1h 04m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d87a368ccb8483188306d865f28c971 : SUCCESS in 1h 46m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a334fc39ef774cf29c5e53d2b77a2298 : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":2},{"id":"40e67a4391a0e10ccb7ff8d0cd16f93760b94501","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 06:44:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"97d2ae03384a4bdb63bc5262ceba1c0588a20509","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 06:45:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba33fd809637492788972eb5e40859bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f16ff9946947718c5c715273a2b654 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7686485cca3c451a855cbde455ba1297 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6cb77654934f3e84641d7f7e0554492217ca6483","revisions":{"37a6072109771d26be2cc3b317945570c8f99a9f":{"kind":"REWORK","_number":1,"created":"2025-09-14 03:35:08.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/53/961053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/961053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/961053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/961053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/961053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/961053/1"}}},"commit":{"parents":[{"commit":"3b4c2efd71fc40d355ab02c5e066815fb4f8a7b9","subject":"zuul: octavia-grenade voting false","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4c2efd71fc40d355ab02c5e066815fb4f8a7b9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:13:57.000000000","tz":-240},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2025-09-14 03:35:08.000000000","tz":0},"subject":"Fix race condition in cascade delete","message":"Fix race condition in cascade delete\n\nupdate_vip was called multiple times concurrently when cascade deleting\na load balancer with many listeners, it may trigger a race condition\nwhen fetching, computing and updating the SGs.\nCalling update_vip for each listener is not necessary, it\u0027s now called\nonly once, that fixes the race condition and optimize the delete flow.\n\nCloses-Bug: #2038529\nChange-Id: I4a6e4830d0e916b1af1a11dd10097980a57d97ea\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 5c802aad949e4abd878f326c53123ddf8a299be0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37a6072109771d26be2cc3b317945570c8f99a9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/37a6072109771d26be2cc3b317945570c8f99a9f"}]},"branch":"refs/heads/stable/2025.1"},"6cb77654934f3e84641d7f7e0554492217ca6483":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-16 22:39:28.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/53/961053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/53/961053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/961053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/961053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/53/961053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/53/961053/2"}}},"commit":{"parents":[{"commit":"5a8d7fe13f074b30af88f26ff71d741a76660b77","subject":"Skip 4.0.{0,1} pylint releases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a8d7fe13f074b30af88f26ff71d741a76660b77"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:13:57.000000000","tz":-240},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2025-10-16 22:39:28.000000000","tz":0},"subject":"Fix race condition in cascade delete","message":"Fix race condition in cascade delete\n\nupdate_vip was called multiple times concurrently when cascade deleting\na load balancer with many listeners, it may trigger a race condition\nwhen fetching, computing and updating the SGs.\nCalling update_vip for each listener is not necessary, it\u0027s now called\nonly once, that fixes the race condition and optimize the delete flow.\n\nCloses-Bug: #2038529\nChange-Id: I4a6e4830d0e916b1af1a11dd10097980a57d97ea\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 5c802aad949e4abd878f326c53123ddf8a299be0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cb77654934f3e84641d7f7e0554492217ca6483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6cb77654934f3e84641d7f7e0554492217ca6483"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
