)]}'
{"id":"openstack%2Foctavia~961054","triplet_id":"openstack%2Foctavia~stable%2F2024.2~I4a6e4830d0e916b1af1a11dd10097980a57d97ea","project":"openstack/octavia","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2025-11-12 23:03:32.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-11-12 21:13:09.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I4a6e4830d0e916b1af1a11dd10097980a57d97ea","subject":"Fix race condition in cascade delete","status":"MERGED","created":"2025-09-14 03:36:13.000000000","updated":"2025-11-12 23:04:35.000000000","submitted":"2025-11-12 23:03:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961054","cherry_pick_of_change":961053,"cherry_pick_of_patch_set":1,"meta_rev_id":"6a2c0a9a2a1914035d2c1ea0e12df0b86a58caea","_number":961054,"virtual_id_number":961054,"owner":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-12 23:03:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2025-09-18 13:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2025-11-12 21:13:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2025-11-12 21:13:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2025-11-12 21:13:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-14 03:36:13.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-09-14 06:16:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 13:49:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-11-12 05:14:36.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5ebc4f33c88dce8fef64831a68dbd4a3083d94f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2025-09-14 03:36:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"376d33afac51edb9ca5e9e39d6f35a3d96f16fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 06:16:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4807493832f346b088dd919f85e615c9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecdac1e40fbb4d64918fd786fe1d338b : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71f73b32c75437a8f6c3c3f59be32b4 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f41686b22e947068920f8d23e183b3b : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/277c85a1870a46d6b7307f05b3655c67 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1293989bb9e435198f30ba235d0204d : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4768b88343eb47d481ed35f789347407 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/911de9b06299444aaf90ef51f50d3986 : SUCCESS in 6m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a42e2196dd354d65b549b48bab77136e : SUCCESS in 6m 04s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/8ae36c8fb6794e18b247620b45bae042 : SUCCESS in 6m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/56d2c3f11aa9494b8edd322c8d5b3617 : SUCCESS in 4m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd24ac471540471e8f14d3b7698766ce : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/357e018cddd84d1eb9c99ded9302f928 : SUCCESS in 7m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e9a287a633df4c1ea2fb6234a6d77c7f : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/fbcc8120c8e84bd483155a1f96bc8ed0 : SUCCESS in 1h 12m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/19631af1594748e9abfe425a6cf90f7b : SUCCESS in 1h 43m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/83f6e35f8223426c90b7e406d26accae : SUCCESS in 1h 38m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c107a09bcdb44872890a0cd0ebb089ae : TIMED_OUT in 2h 34m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/06c4a75f57654f6caf9880fdfc129d80 : SUCCESS in 1h 07m 42s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/820837dae82849c5bc4fa7f0c37ee3cd : SUCCESS in 40m 48s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/f9d242fb754e42eea5c01713ba2711a5 : SUCCESS in 50m 42s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/8ce25efc635f4882846766cc0a623fe0 : FAILURE in 58m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a3e4703ba74c43b4af3258bf2c406cda : SUCCESS in 53m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/b25f1479905d4288a3bc4628886dd0f9 : TIMED_OUT in 2h 34m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/597781f3a9d9417e9b46836da391fc40 : FAILURE in 56m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/33b5f2ebcd204d22bdede70128ae3cda : TIMED_OUT in 2h 32m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/21e62b2f52f640bbb804007acbc7ef07 : FAILURE in 46m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dd1319c24f5daff8c5f8820aa0a989e71f81d57","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 13:49:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd923326bf3e3a0489545f9651872585474ca5cd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 13:49:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"135e171f49c162edec1c63f71822b66e496e635b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 16:20:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dd32dcce10e4a18930604ef8810b82e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74439b6f902c4f8abbccbc97528f21d3 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c72935be164e70a946eac6a548e02c : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdfd497e99fe4d318ceffa84e360df92 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e35c6c5d21dc4a5088357b214c82b2b5 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b40d1db500de4baba908580f889cdf8d : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fba1fe6e7dd43d59d24839e47503c5b : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2ee07552a6f4350a0165866cb1d86df : SUCCESS in 5m 24s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/23a9842057594d1892631e02cc0aa7eb : SUCCESS in 7m 17s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3db8a03f218c461b8da54fd44aeafd2e : SUCCESS in 7m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4a12715b8a8d4e6ba1ff8bffdbdd45cf : SUCCESS in 3m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5522a1e4330a44c7824162bedbfa07c3 : SUCCESS in 9m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d631cd87bba481cbdcd3e4380d85bed : SUCCESS in 7m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/781d193041b346739977ae6383fada8c : SUCCESS in 5m 02s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/c0927eaef0a744d3acdc666b91b50f2c : SUCCESS in 1h 11m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/56e8614c27944a3ebc946cbf606b4dcd : SUCCESS in 1h 47m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/292ba3b118d14a1ea08d83b6b0d9cc19 : SUCCESS in 1h 04m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/258787a0ab9d4e93ae9029ec3fd8e19f : SUCCESS in 1h 47m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-2 https://zuul.opendev.org/t/openstack/build/58d2728b485b4af6b18cbea73172c436 : SUCCESS in 1h 04m 50s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/a299417c73b3433796dd8d8e6dd4e799 : SUCCESS in 32m 48s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/50ab257652c74dd7acc6679a34e7c93c : SUCCESS in 39m 29s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6a4016dad5024c10b6ec9741a687f101 : FAILURE in 40m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-2 https://zuul.opendev.org/t/openstack/build/6337e660904f48c4b56da2e08fa03e75 : SUCCESS in 31m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5dae51268b0c44a29f98b6d9ae17f473 : SUCCESS in 2h 28m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2f0d5c13be36429ba4e3d4e47425da2d : FAILURE in 39m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-2 https://zuul.opendev.org/t/openstack/build/07d30382fa4f45b88311f6825d057a19 : FAILURE in 2h 14m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-2 https://zuul.opendev.org/t/openstack/build/d5fe923245b4459f9808f24d540657a2 : FAILURE in 30m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07ca03ff3fe8f1811ad4934f8f2667d82c9888f4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-11-12 21:13:09.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b3ca4e7cfbf7c6b969cbbd4365a0962ea46e0cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 21:13:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3463c86a05c6a8aaa4d4acd06bb01d6f761cbd4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 23:03:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797a04e659e544f7b59fb674da2e1370\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9efa158933f94402ab22537e04dc4b46 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd0ec1251c164d7bac34e9dcf7cc76f4 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4c929d9e4fa4cf79aba2705ce474fab : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4bee3f45324a26b45a68314809193b : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/952dd803f9024a11b2012e9c8ee313e6 : SUCCESS in 4m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fbff4f1d1b4f4774ae2e50eb14d6e0e0 : SUCCESS in 6m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/19db955273664667b5c30dbd85d62ce7 : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api-stable-2024-2 https://zuul.opendev.org/t/openstack/build/15402cc656d642719185060a24716004 : SUCCESS in 1h 00m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/2afa6f4b9d4a425eab3f75150a623ce3 : SUCCESS in 1h 49m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-2 https://zuul.opendev.org/t/openstack/build/5c85abb57a644e948c398b35d4751c80 : SUCCESS in 1h 01m 08s\n- octavia-v2-dsvm-tls-barbican-stable-2024-2 https://zuul.opendev.org/t/openstack/build/e3ded9c9c98c4337b13be917bb7165ee : SUCCESS in 31m 34s\n- octavia-grenade-stable-2024-2 https://zuul.opendev.org/t/openstack/build/498a9983fe534bfb98bf83e451a2e3e4 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":1},{"id":"841de16bf22add3fb37eed12c3d0843c39f8005d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 23:03:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a2c0a9a2a1914035d2c1ea0e12df0b86a58caea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 23:04:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dc21ce184514e72bdb05d7d468d7973\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4605436e4a8d44f897494328f4b6ebde : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ea01a446d1442fca661068dbc44fb1d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98d96037bdae3a46b94c4a53f63f348c2cba078a","revisions":{"98d96037bdae3a46b94c4a53f63f348c2cba078a":{"kind":"REWORK","_number":1,"created":"2025-09-14 03:36:13.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/54/961054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/54/961054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/961054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/961054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/54/961054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/54/961054/1"}}},"commit":{"parents":[{"commit":"8249ae9b837acb14a60e8d235cd0e8e1b4dde1c1","subject":"Add an octavia-wsgi compatibility script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8249ae9b837acb14a60e8d235cd0e8e1b4dde1c1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:13:57.000000000","tz":-240},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2025-09-14 03:36:13.000000000","tz":0},"subject":"Fix race condition in cascade delete","message":"Fix race condition in cascade delete\n\nupdate_vip was called multiple times concurrently when cascade deleting\na load balancer with many listeners, it may trigger a race condition\nwhen fetching, computing and updating the SGs.\nCalling update_vip for each listener is not necessary, it\u0027s now called\nonly once, that fixes the race condition and optimize the delete flow.\n\nCloses-Bug: #2038529\nChange-Id: I4a6e4830d0e916b1af1a11dd10097980a57d97ea\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 5c802aad949e4abd878f326c53123ddf8a299be0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98d96037bdae3a46b94c4a53f63f348c2cba078a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98d96037bdae3a46b94c4a53f63f348c2cba078a"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
