)]}'
{"id":"openstack%2Foctavia~961055","triplet_id":"openstack%2Foctavia~stable%2F2024.1~I4a6e4830d0e916b1af1a11dd10097980a57d97ea","project":"openstack/octavia","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2025-10-14 02:22:49.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-10-14 00:44:27.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I4a6e4830d0e916b1af1a11dd10097980a57d97ea","subject":"Fix race condition in cascade delete","status":"MERGED","created":"2025-09-14 03:36:39.000000000","updated":"2025-10-14 02:24:46.000000000","submitted":"2025-10-14 02:22:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961055","cherry_pick_of_change":961054,"cherry_pick_of_patch_set":1,"meta_rev_id":"c16f31017deedeb786077363e6c04c27bf78efe0","_number":961055,"virtual_id_number":961055,"owner":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 02:22:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-10-14 00:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-13 16:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-10-14 00:44:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-14 03:36:39.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2025-09-14 06:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 13:49:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-10-09 17:48:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f1d2352b6aa86f96478149d97476281d3922066b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2025-09-14 03:36:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"bc5345067435ba37f3f39b2cff2c15d25daed96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 06:12:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be266f238c104334ae7c6c4db5ffdb7a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f49e80a37ba4df5bb0a9affcb774507 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ecb51614c2499cba3348caf12e4ba2 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b3b5e5673864b55bf32940eade6a66a : SUCCESS in 7m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae29ec0eab194b22a715e8726597ac9c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075e5830161b46f9a081153b2a016282 : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8155eeb3a3e47fbb9d62dcc211b64d7 : SUCCESS in 5m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/771c5aae00f04f7eaaf208da04708262 : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b6879b7024e4d82a2f271299a4c2e85 : SUCCESS in 7m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4879b7c0fd1646bb9c69898308feade7 : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70b3b9a7552947d1a38e4b201ec1ccc1 : SUCCESS in 6m 14s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b1aa5d17c1af4e11a3137032c0e783fe : SUCCESS in 1h 21m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/eda2e7d6380040cd94e476979c9c3d85 : SUCCESS in 1h 51m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c63c4324fed4400b8e16bc4210ccd986 : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8c1d18e650794a8eabd43f44ea63be99 : SUCCESS in 2h 01m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/236d46332f0c4966ac8e6d9c52b94ebb : SUCCESS in 1h 28m 56s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/b97c0c61e6d74edca9e5cdffefcead5f : SUCCESS in 57m 39s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/6ac683d8c527442b89b1b943285fc0d4 : SUCCESS in 1h 03m 54s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/99eb86bee5134ece842f2d1b23530a32 : FAILURE in 44m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/583596291cdf4bbd879cb54f535daada : FAILURE in 52m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/19e50e3a6ed14569b03f4236112f6c01 : SUCCESS in 40m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/72c8e4ff46bc499796692a6673fbe5fe : SUCCESS in 2h 31m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f880e06db8a04291b737df9382674dee : RETRY_LIMIT in 3m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d0574cbb89c245e1b78987ea3973d859 : FAILURE in 27m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/8848d38f00a24311be1c2b7b31270f9c : FAILURE in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f81b3f7d35fee961250100f59a6d469956ea36a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 13:49:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"feff580d57c2a0406acb3afb0274d23ec676e5e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-09 17:48:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efeb0fe0c9cf6e6d3d315af8af36c82e37dd0b1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 17:48:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef2b028940b32301f686b66e877782e9dc887b0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 19:33:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b528d26fc464700b54ac509dee0b90e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46add7b6a6945e38036bc03c8560df8 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7932269917843eca79a3fa3af93f20a : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/877a425dcec54664b710a2567a1758ef : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210aefe32fbc44779752130df848b095 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f04375051a8340898a5c82cbcf5984f2 : SUCCESS in 6m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11ba65cf8b2547c69c02d7e98c4e5573 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4be29f38a1d54000a985d05d219f1700 : SUCCESS in 6m 04s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/d67b380d878a4520a6505ec3465603d6 : SUCCESS in 1h 38m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/5a30c97c5e424998a07dc592e658db52 : FAILURE in 1h 43m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/3f2ddeb1600f4425833dc0bc5a2d99eb : SUCCESS in 1h 17m 39s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/025c1e9da245450c8a8d88b06fc215e5 : SUCCESS in 51m 42s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c9c6a63243ba495f8cb937d9f90c8738 : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":1},{"id":"11bb531be4e89f6f3abc8459d1678206268d7c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-13 16:58:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_5112\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"}],"_revision_number":2},{"id":"c61cfe54281aa4f06120057966231592277b809c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 19:20:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc8635fb60124c3aa3321d234552a4f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d1f39cdedb4a28a69c74941bf36d40 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f94a6682378146fea444202a52c1ce6f : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bee6d71f73849fc8027fdb724b62ab8 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15b6c823ae244d16a748663e227d4b83 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4485311e6e3d4b498be73325e4b52527 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40cef8cae12d476491d48a5bd791c790 : SUCCESS in 5m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c7805f629cf94a8285de18d81e499fdc : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/21529c6ce0be4a16802049226818af2c : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da209a63f71b4374b3c1f8b6a749bd63 : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d7b79457bec41fdab3b31b0aa454e67 : SUCCESS in 6m 44s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/7dd1e0de013f49f583a844288f603cd0 : SUCCESS in 1h 02m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f397627c28c9491896b1fdb6a6462eac : SUCCESS in 1h 39m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/f08248e5337b46bfa81915f2072d9899 : SUCCESS in 1h 01m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4f809f050f4842928a7a9fb4ae332d95 : SUCCESS in 1h 46m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2024-1 https://zuul.opendev.org/t/openstack/build/996c1b9f407047bc8a2b89268480dc95 : SUCCESS in 58m 32s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/68ef311802dd4103bb7ab365bc2ea805 : SUCCESS in 37m 38s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c6202015bb8c4bc4afb27d6fead092d2 : SUCCESS in 35m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ab643408fbe5412eaf3e4fbc4ac25350 : FAILURE in 42m 24s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips-stable-2024-1 https://zuul.opendev.org/t/openstack/build/228adbaa6a8a4688b38ad0062dc85ba2 : FAILURE in 48m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2024-1 https://zuul.opendev.org/t/openstack/build/a3cd3c1710bc4567aa90f8ac06a7640d : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2024-1 https://zuul.opendev.org/t/openstack/build/42f935a4215244e9b4c9e47d4d484ba6 : SUCCESS in 2h 19m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2024-1 https://zuul.opendev.org/t/openstack/build/4f972666596140d68a2714c201590c50 : RETRY_LIMIT in 4m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2024-1 https://zuul.opendev.org/t/openstack/build/472717b7c0944c87afba1f666c9727a7 : FAILURE in 23m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2024-1 https://zuul.opendev.org/t/openstack/build/7ab1b29cae384882a4f6946e74738adc : FAILURE in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9646b73580fd48f63b3d3a46844137083b98304c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-10-14 00:44:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"caf6236bd43a2b105ad0a65ad97befabb6e1793a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 00:44:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ca8c5cbc67aca9be371489d1cc874376ced02b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:22:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89021a2b445744fb8afaa9aefd068ba1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59b215ceaec49b793e52c1f56b935e7 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e138bccf042e41a3a142e9ac0eb2fe72 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b16ed5402c554e1cb48239fd62e25e1a : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a78b871896c4cce89812a75f1f7b14d : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6c6066cf0ab4fcaa0e1d798684efae0 : SUCCESS in 5m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4edc5be8c20495eac4a311def77ebd3 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6060947ecdd647c98705d7a35e707665 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api-stable-2024-1 https://zuul.opendev.org/t/openstack/build/c8f6b472a3a34c8cbd56fdff50ff08ee : SUCCESS in 1h 00m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/cce96b762ff14f1ba5d5eee19569f0b0 : SUCCESS in 1h 37m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2024-1 https://zuul.opendev.org/t/openstack/build/ed6fc4945dd6481e9eadc1e2bbefd1cd : SUCCESS in 1h 06m 10s\n- octavia-v2-dsvm-tls-barbican-stable-2024-1 https://zuul.opendev.org/t/openstack/build/29429060a0f94b8ca4eb622a6aad8c01 : SUCCESS in 32m 57s\n- octavia-grenade-stable-2024-1 https://zuul.opendev.org/t/openstack/build/64299be50b954aeda01369a81a14df1b : SUCCESS in 36m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d8ace7615653dd7313eb67cdf2529c33580f06c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:22:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c16f31017deedeb786077363e6c04c27bf78efe0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 02:24:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82321cf7abf743128a1bb94e805788f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071df34357604bd395ca6eecbe7232ab : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f2df1a62cc64bf4a7401a6f64c72f98 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e66d99f3486c0d6b9e5dd05baaa98a75145bd6eb","revisions":{"daf4bf978e1b014a0ba7a37770270fefd86b0338":{"kind":"REWORK","_number":1,"created":"2025-09-14 03:36:39.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/55/961055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/961055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/961055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/961055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/961055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/961055/1"}}},"commit":{"parents":[{"commit":"041b0795972afe103ea802b47105f38ad45f65cb","subject":"Add an octavia-wsgi compatibility script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/041b0795972afe103ea802b47105f38ad45f65cb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:13:57.000000000","tz":-240},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2025-09-14 03:36:39.000000000","tz":0},"subject":"Fix race condition in cascade delete","message":"Fix race condition in cascade delete\n\nupdate_vip was called multiple times concurrently when cascade deleting\na load balancer with many listeners, it may trigger a race condition\nwhen fetching, computing and updating the SGs.\nCalling update_vip for each listener is not necessary, it\u0027s now called\nonly once, that fixes the race condition and optimize the delete flow.\n\nCloses-Bug: #2038529\nChange-Id: I4a6e4830d0e916b1af1a11dd10097980a57d97ea\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 5c802aad949e4abd878f326c53123ddf8a299be0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/daf4bf978e1b014a0ba7a37770270fefd86b0338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/daf4bf978e1b014a0ba7a37770270fefd86b0338"}]},"branch":"refs/heads/stable/2024.1"},"e66d99f3486c0d6b9e5dd05baaa98a75145bd6eb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-13 16:58:58.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/55/961055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/961055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/961055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/961055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/961055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/961055/2"}}},"commit":{"parents":[{"commit":"cc3dd1abe3e23071662494786a5e5bddb3538e89","subject":"Reduce tune.ssl.cachesize for HTTPS terminating listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cc3dd1abe3e23071662494786a5e5bddb3538e89"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:13:57.000000000","tz":-240},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2025-10-13 16:58:58.000000000","tz":0},"subject":"Fix race condition in cascade delete","message":"Fix race condition in cascade delete\n\nupdate_vip was called multiple times concurrently when cascade deleting\na load balancer with many listeners, it may trigger a race condition\nwhen fetching, computing and updating the SGs.\nCalling update_vip for each listener is not necessary, it\u0027s now called\nonly once, that fixes the race condition and optimize the delete flow.\n\nCloses-Bug: #2038529\nChange-Id: I4a6e4830d0e916b1af1a11dd10097980a57d97ea\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n(cherry picked from commit 5c802aad949e4abd878f326c53123ddf8a299be0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e66d99f3486c0d6b9e5dd05baaa98a75145bd6eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e66d99f3486c0d6b9e5dd05baaa98a75145bd6eb"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
