)]}'
{"id":"openstack%2Foctavia~961406","triplet_id":"openstack%2Foctavia~master~I2990497c13bc8f76a2a1e93fb5e95d649f7dd8a1","project":"openstack/octavia","branch":"master","topic":"Update-amphora-netns.systemd","attention_set":{},"removed_from_attention_set":{"37920":{"account":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"last_update":"2025-09-18 15:55:15.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I2990497c13bc8f76a2a1e93fb5e95d649f7dd8a1","subject":"Update amphora-netns.systemd","status":"NEW","created":"2025-09-16 19:12:59.000000000","updated":"2025-09-18 18:44:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"085106b6907b4189b9b844893c4630cc57114070","_number":961406,"virtual_id_number":961406,"owner":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-18 18:44:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-09-16 22:17:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-18 06:29:46.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"747e214972f2e69ef5641c5d5f409cfb6f2395d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"date":"2025-09-16 19:12:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bd0653786a3a0f04584921f53402f35979300b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-16 22:17:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a391cd39195c439a904d55a945cb84ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d89a2dc9d2d48379dc37911a5319cec : FAILURE in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf7aafd678f04cc793fca94f5251c2e3 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b919e9471d1d402dbbd3192e17ec54e4 : FAILURE in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e8b50daf23784410b9c48952e5f431cb : FAILURE in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e326ce91851d432a9ec7d3389e56703f : FAILURE in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a6716f263b467da042af17cb65c3a3 : SUCCESS in 9m 31s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0e3fc4d0d25342da8da0ed33abd605ee : FAILURE in 6m 41s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/9a59aca85957462da8b0ece5ec5f2497 : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c712ff2a56a9489eb78e044a40903b4a : SUCCESS in 5m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a0d3347f87a45a0b50b06d5ba19ae24 : SUCCESS in 8m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c85ed05cab8745f1967c4aead4213ab6 : SUCCESS in 9m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2dc176bd55d545259a327ac352374d4f : SUCCESS in 1h 06m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e349484167b7406cbd89dac142232d66 : SUCCESS in 1h 52m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4186fc74e73c44839715351d271db14e : SUCCESS in 1h 08m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c62016827f344569848221bf38131873 : SUCCESS in 1h 55m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/74246f4b377944eaa67f85954d898563 : SUCCESS in 1h 10m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5f2528eb8db744b88ba452f54f08db35 : SUCCESS in 37m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d2e2fb891f94aa988f1943271cc6143 : SUCCESS in 40m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c5327ef63d04fd9a4d4e9140ed57ef0 : SUCCESS in 33m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/90b94b07f4a34c7e8dda5bb582c7a43c : SUCCESS in 2h 16m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/125304b9d8ab4291ab3f16cd161cb847 : TIMED_OUT in 3h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2b8907d3f9064a5f8c9872ce06d6e072 : SUCCESS in 2h 27m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9b6e5ec1f2d83e0f901261eee9750e1b1e9ae29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"date":"2025-09-17 18:13:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b6bf851970be7ffc8618e4595f71fb2b085db4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"date":"2025-09-17 18:45:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7be52c6d3ee8f7397e5be6f1ee9aebc4d1369abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 22:05:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe8d2e01c4e44744bde061de702ed17d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9e4aa24288a41a7b766b78825b8c2d8 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1684e227402c42b99960c0dba686e739 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73a9eb5192224245b9e44910d50599a7 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16725afbffe54ae6875d5f827c1efeb5 : SUCCESS in 6m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/117a413ed62e4ae2b8051e9984f54af0 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819dd1fd96b34db18b1797f030bc8723 : SUCCESS in 12m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/feaafbce17b44666ab173e9230063cc4 : SUCCESS in 5m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/d971dd53deab4000b8fc1eaeb3c8a98b : SUCCESS in 7m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0bfcd9cacc014fa99ca769aa7b219080 : SUCCESS in 5m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d6d4bb741f2b4e01964563196556946a : SUCCESS in 6m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/869fff63c56d4bf387202ec41357fb2c : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5665a70c97ff4e73aafbf2fc0783d6c6 : SUCCESS in 1h 34m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9329a951ccea4280b7bc4248d58f428a : SUCCESS in 1h 53m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/51bd363d533f42da99b7431c106c6824 : SUCCESS in 1h 01m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/479d12fe10834a2c8727faa3d4852c61 : SUCCESS in 1h 55m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/10defc3f34c54458a8e23e84f5b0a219 : SUCCESS in 1h 07m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7034123b65d4166b4e31043c569ca6a : SUCCESS in 38m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/79dabcd6ff6441ee8d9465759119adc0 : SUCCESS in 46m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/672fce0344834f18b86042d66baea4d0 : SUCCESS in 32m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71ba0da8d58b40059aa3d158d06a27bd : SUCCESS in 2h 25m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e8cc558f2c70440e90fde037c5baf45b : TIMED_OUT in 3h 02m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aaea4442cf1e46ffbb1d6d39dca79d74 : TIMED_OUT in 2h 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"01400590e19a3a8a727784d82478c77f02b77af7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-09-18 06:29:46.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"38c80a57771653ec9bde73baf010e697efac6ef4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"date":"2025-09-18 15:52:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4098f593a5d013b722dccfc5ab0fe16fcfc18a2e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"date":"2025-09-18 15:55:15.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"085106b6907b4189b9b844893c4630cc57114070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 18:44:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7fda5be63a4bac8ac5d7f0e57e1684\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/289ffd9ff49948039e92d05dc87ef0b0 : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75efb05ea5d407eae842826f36f4546 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9e18d9131874e679a07a890a5ad94ba : SUCCESS in 6m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a20cacb0d054031a0f7172b0153b70d : SUCCESS in 7m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ecafd2588cb4ed2a087c38265a614af : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57197702722b4cc68801206a3962fcd5 : SUCCESS in 13m 06s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cad49799b8a04b018aed2f3f99afdb00 : SUCCESS in 4m 57s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c89f77a7b496419d9e0b7c592aaf6937 : SUCCESS in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fdb923c4c76c4a87b5721122a07acf45 : SUCCESS in 3m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5aa6819c0d34c9885c009e6248b3a11 : SUCCESS in 4m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/984e414c675e47a7986e9da229025926 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ef7a01dee2b41c78c2fadc5e20bdbfb : SUCCESS in 1h 27m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9d9829874d764caba6c314f76a434238 : SUCCESS in 1h 55m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/172149cfeb494de2a119d30bec221176 : SUCCESS in 1h 08m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fa4ee15a965844739e63d5ddfcd9736b : SUCCESS in 1h 57m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c6c29c799fe340159709afa74ee0163f : SUCCESS in 1h 07m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ff4ce5aae2048b385908a1e8f47f93f : SUCCESS in 37m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/06afb411493a498289e81743bdb4619f : SUCCESS in 47m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31c1985799404d169f1b5d4e1ed47157 : SUCCESS in 35m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f9e44bee569743d9b023fe871535641c : SUCCESS in 2h 32m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/240ecc0d68ea4e3cbadb399d7047fabd : SUCCESS in 2h 43m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/43a2d2556ee449db833a0a1347144679 : TIMED_OUT in 2h 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0a3703159ea3d907d4a9981bee8c45f7ac4a0d2a","revisions":{"343495a53422f72f305c2120be9ee65448b4b41e":{"kind":"REWORK","_number":1,"created":"2025-09-16 19:12:59.000000000","uploader":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"ref":"refs/changes/06/961406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/961406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/961406/1"}}},"commit":{"parents":[{"commit":"26f84f4a5aaaaf1a587263b9a31da37a9f23b97e","subject":"Merge \"Fix Amphora controller IP list update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26f84f4a5aaaaf1a587263b9a31da37a9f23b97e"}]}],"author":{"name":"sizovv","email":"pakemon32@gmail.com","date":"2025-09-16 18:53:15.000000000","tz":180},"committer":{"name":"sizovv","email":"pakemon32@gmail.com","date":"2025-09-16 19:12:41.000000000","tz":180},"subject":"Update amphora-netns.systemd","message":"Update amphora-netns.systemd\n\nThe previous version of the shell pipeline in amphora-netns.systemd could fail after an amphora reboot (e.g., due to human error), which in most cases triggered its recreation and increased downtime. Also added a variable to the Jinja render() function to eliminate hadrcoding.\nThis update also improves code readability.\n\nChange-Id: I2990497c13bc8f76a2a1e93fb5e95d649f7dd8a1\nSigned-off-by: sizovv \u003cpakemon32@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/343495a53422f72f305c2120be9ee65448b4b41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/343495a53422f72f305c2120be9ee65448b4b41e"}]},"branch":"refs/heads/master"},"dd4d638f8be6e728e46d0d6f77ac6727f5861c71":{"kind":"REWORK","_number":2,"created":"2025-09-17 18:13:18.000000000","uploader":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"ref":"refs/changes/06/961406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/961406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/961406/2"}}},"commit":{"parents":[{"commit":"26f84f4a5aaaaf1a587263b9a31da37a9f23b97e","subject":"Merge \"Fix Amphora controller IP list update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26f84f4a5aaaaf1a587263b9a31da37a9f23b97e"}]}],"author":{"name":"sizovv","email":"pakemon32@gmail.com","date":"2025-09-16 18:53:15.000000000","tz":180},"committer":{"name":"sizovv","email":"pakemon32@gmail.com","date":"2025-09-17 18:11:51.000000000","tz":180},"subject":"Update amphora-netns.systemd","message":"Update amphora-netns.systemd\n\nThe previous version of the shell pipeline in amphora-netns.systemd could fail after an amphora reboot (e.g., due to human error), which in most cases triggered its recreation and increased downtime. Also added a variable to the Jinja render() function to eliminate hadrcoding.\nThis update also improves code readability.\n\nChange-Id: I2990497c13bc8f76a2a1e93fb5e95d649f7dd8a1\nSigned-off-by: sizovv \u003cpakemon32@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd4d638f8be6e728e46d0d6f77ac6727f5861c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd4d638f8be6e728e46d0d6f77ac6727f5861c71"}]},"branch":"refs/heads/master"},"dba56982ee6b995aa21785b0acc3bd61445586e3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-09-17 18:45:38.000000000","uploader":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"ref":"refs/changes/06/961406/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/961406/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/961406/3"}}},"commit":{"parents":[{"commit":"26f84f4a5aaaaf1a587263b9a31da37a9f23b97e","subject":"Merge \"Fix Amphora controller IP list update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26f84f4a5aaaaf1a587263b9a31da37a9f23b97e"}]}],"author":{"name":"sizovv","email":"pakemon32@gmail.com","date":"2025-09-16 18:53:15.000000000","tz":180},"committer":{"name":"Vadim Sizov","email":"pakemon32@gmail.com","date":"2025-09-17 18:45:38.000000000","tz":0},"subject":"Update amphora-netns.systemd","message":"Update amphora-netns.systemd\n\nThe previous version of the shell pipeline in amphora-netns.systemd could fail after an amphora reboot (e.g., due to human error), which in most cases triggered its recreation and increased downtime. Also added a variable to the Jinja render() function to eliminate hadrcoding.\nThis update also improves code readability.\n\nRelated-Bug: #2027995\nChange-Id: I2990497c13bc8f76a2a1e93fb5e95d649f7dd8a1\nSigned-off-by: sizovv \u003cpakemon32@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dba56982ee6b995aa21785b0acc3bd61445586e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dba56982ee6b995aa21785b0acc3bd61445586e3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0a3703159ea3d907d4a9981bee8c45f7ac4a0d2a":{"kind":"REWORK","_number":4,"created":"2025-09-18 15:52:32.000000000","uploader":{"_account_id":37920,"name":"Vadim Sizov","display_name":"Vadim","email":"pakemon32@gmail.com","username":"sizovv","status":"System administrator in cloud ☁️.\nOpenStack enjoyer."},"ref":"refs/changes/06/961406/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/961406/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/961406/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/961406/4"}}},"commit":{"parents":[{"commit":"26f84f4a5aaaaf1a587263b9a31da37a9f23b97e","subject":"Merge \"Fix Amphora controller IP list update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/26f84f4a5aaaaf1a587263b9a31da37a9f23b97e"}]}],"author":{"name":"sizovv","email":"pakemon32@gmail.com","date":"2025-09-16 18:53:15.000000000","tz":180},"committer":{"name":"sizovv","email":"pakemon32@gmail.com","date":"2025-09-18 15:51:18.000000000","tz":180},"subject":"Update amphora-netns.systemd","message":"Update amphora-netns.systemd\n\nThe previous version of the shell pipeline in amphora-netns.systemd could fail after an amphora reboot (e.g., due to human error), which in most cases triggered its recreation and increased downtime. Also added a variable to the Jinja render() function to eliminate hadrcoding.\n\nReplaced an awk-based pipeline with a more straightforward grep and cut combination for better readability, reformatted the command into a multi-line structure.\n\nRelated-Bug: #2027995\nChange-Id: I2990497c13bc8f76a2a1e93fb5e95d649f7dd8a1\nSigned-off-by: sizovv \u003cpakemon32@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a3703159ea3d907d4a9981bee8c45f7ac4a0d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a3703159ea3d907d4a9981bee8c45f7ac4a0d2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
